)]}'
{"id":"openstack%2Foslo.messaging~765248","triplet_id":"openstack%2Foslo.messaging~master~I76c1df18a41f8b1449309ce90c55d085de27497f","project":"openstack/oslo.messaging","branch":"master","hashtags":[],"change_id":"I76c1df18a41f8b1449309ce90c55d085de27497f","subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","status":"ABANDONED","created":"2020-12-03 07:19:13.000000000","updated":"2020-12-11 14:15:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"82258c1a9c347b73fd3a2ddf61bb6db72c68a970","_number":765248,"virtual_id_number":765248,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-03 09:37:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2020-12-03 08:23:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 08:23:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-12-03 09:37:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"131ba11c1821c69b9a06fb413db7f5a65fcdaa82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-12-03 07:19:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7314205e60bbc61975f5208d51cd51b8a41b6aed","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-03 08:23:31.000000000","message":"Patch Set 1: Code-Review-1\n\nDuplicated with https://review.opendev.org/c/openstack/oslo.messaging/+/761346","accounts_in_message":[],"_revision_number":1},{"id":"4ed9e88fec3ae769970a7af6b672dc6f3da0cc99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 09:37:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3000c07c15224e639574bd7e6a2db347 : SUCCESS in 1h 19m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c4c5cb216dc4239afebbf2957243a42 : SUCCESS in 8m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4558b327146d436b8fba9e77752120c1 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c41179014564658af46b89d22557df2 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80e8a0f23c7347f89056a2f8c11cc4d7 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bd943757f9244e99f0ec7766eaa647f : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d16af6ea5df4476b8d35cb437c74a5c9 : FAILURE in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbcd48588da44c1ba0e78427c8b270d2 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d59a20ced614e05b18c9714902d0f10 : SUCCESS in 6m 53s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/1066b15958284277b61e5ef2f0c0016b : SUCCESS in 6m 38s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/c4420bc039ed4dd997feaa48d747e127 : FAILURE in 7m 43s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/b67c9fd344c64b9a906770981ec819fe : SUCCESS in 7m 56s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/3639e449326c4804baa2c5722fbef96f : FAILURE in 7m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/543bbeb1f943409e9950e3003e924589 : SUCCESS in 1h 24m 19s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/89a0c2e226e04c168af933d4b48165f0 : SUCCESS in 1h 32m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2a2c82666ee740e6b06094c5070ca8e2 : SUCCESS in 1h 27m 26s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8cf74d5fc4904b06962d3042a151223f : SUCCESS in 1h 04m 55s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/649319c832ed41979c98ba46f0ebde41 : SUCCESS in 1h 08m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82258c1a9c347b73fd3a2ddf61bb6db72c68a970","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-12-11 14:15:48.000000000","message":"Abandoned\n\nDuplicate.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42fe05337a3b92cf832bbf008f89c179b8b3d21f","revisions":{"42fe05337a3b92cf832bbf008f89c179b8b3d21f":{"kind":"REWORK","_number":1,"created":"2020-12-03 07:19:13.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/48/765248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/48/765248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/765248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/765248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/48/765248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/48/765248/1"}}},"commit":{"parents":[{"commit":"0c4203bbdc463010d756a71bbd06720867122bcb","subject":"Python 3.9: isAlive is removed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c4203bbdc463010d756a71bbd06720867122bcb"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-03 07:18:05.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-03 07:19:04.000000000","tz":480},"subject":"Replace deprecated UPPER_CONSTRAINTS_FILE variable","message":"Replace deprecated UPPER_CONSTRAINTS_FILE variable\n\nChange-Id: I76c1df18a41f8b1449309ce90c55d085de27497f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/42fe05337a3b92cf832bbf008f89c179b8b3d21f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/42fe05337a3b92cf832bbf008f89c179b8b3d21f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
