)]}'
{"id":"openstack%2Foslo.messaging~76686","triplet_id":"openstack%2Foslo.messaging~master~Ic2ede3046709b831adf8204e4c909c589c1786c4","project":"openstack/oslo.messaging","branch":"master","topic":"bug/856764","hashtags":[],"change_id":"Ic2ede3046709b831adf8204e4c909c589c1786c4","subject":"Slow down Kombu reconnect attempts","status":"MERGED","created":"2014-02-26 23:23:08.000000000","updated":"2014-03-04 04:33:25.000000000","submitted":"2014-03-04 04:33:25.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fb7d233b5f8b4d64997349c780294ebfc932f459","_number":76686,"virtual_id_number":76686,"owner":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":2,"date":"2014-03-04 04:33:25.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":3189,"name":"Chet Burgess","email":"cfb@metacloud.com","username":"cfb-n"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"recommended":{"_account_id":3189,"name":"Chet Burgess","email":"cfb@metacloud.com","username":"cfb-n"},"all":[{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-03-01 00:08:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3189,"name":"Chet Burgess","email":"cfb@metacloud.com","username":"cfb-n"},{"value":2,"date":"2014-03-04 02:28:51.000000000","_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":2,"date":"2014-03-01 13:35:47.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"all":[{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":3189,"name":"Chet Burgess","email":"cfb@metacloud.com","username":"cfb-n"},{"value":1,"date":"2014-03-04 02:28:51.000000000","_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"date":"2014-03-01 13:35:47.000000000","_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":3189,"name":"Chet Burgess","email":"cfb@metacloud.com","username":"cfb-n"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":3189,"name":"Chet Burgess","email":"cfb@metacloud.com","username":"cfb-n"},{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-03-01 00:08:37.000000000","updated_by":{"_account_id":3189,"name":"Chet Burgess","email":"cfb@metacloud.com","username":"cfb-n"},"reviewer":{"_account_id":3189,"name":"Chet Burgess","email":"cfb@metacloud.com","username":"cfb-n"},"state":"REVIEWER"},{"updated":"2014-03-01 13:35:47.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2014-03-04 02:28:51.000000000","updated_by":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"reviewer":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"state":"REVIEWER"},{"updated":"2014-03-04 04:33:25.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f6ae23e0ac393f2b4eac9ec176ddd4aa24b73836","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-26 23:23:19.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"1cb25b1fe2a9c36c293dee9f8fbdc7f1b58c7c85","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-27 00:25:16.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/86/76686/1/check/gate-oslo.messaging-pep8/b61fe84 : SUCCESS in 1m 04s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/86/76686/1/check/gate-oslo.messaging-docs/ff0418a/doc/build/html/ : SUCCESS in 1m 27s\n- gate-oslo.messaging-python26 http://logs.openstack.org/86/76686/1/check/gate-oslo.messaging-python26/adafcf9 : SUCCESS in 52s\n- gate-oslo.messaging-python27 http://logs.openstack.org/86/76686/1/check/gate-oslo.messaging-python27/24a467d : SUCCESS in 1m 23s\n- check-devstack-dsvm-cells http://logs.openstack.org/86/76686/1/check/check-devstack-dsvm-cells/6c5ef05 : SUCCESS in 10m 44s\n- check-tempest-dsvm-full http://logs.openstack.org/86/76686/1/check/check-tempest-dsvm-full/671b235 : SUCCESS in 50m 37s\n- check-tempest-dsvm-neutron http://logs.openstack.org/86/76686/1/check/check-tempest-dsvm-neutron/3f9978e : SUCCESS in 36m 49s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/86/76686/1/check/check-tempest-dsvm-neutron-full/faa9063 : SUCCESS in 45m 59s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/86/76686/1/check/gate-tempest-dsvm-large-ops/1dcca8c : SUCCESS in 19m 26s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/86/76686/1/check/gate-tempest-dsvm-neutron-large-ops/b53362f : SUCCESS in 14m 38s\n- check-grenade-dsvm http://logs.openstack.org/86/76686/1/check/check-grenade-dsvm/0348ff1 : SUCCESS in 36m 48s\n","accounts_in_message":[],"_revision_number":1},{"id":"1a636b6fbecb7c944e3dcd6e4602b5da3bf218ee","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2014-02-28 21:56:13.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\nThis is dependent on another patch set that is still being prepared.  Will rebase once that is done.","accounts_in_message":[],"_revision_number":1},{"id":"31e032dba52aa7de1a9b42abd709a5a56752b857","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2014-02-28 22:24:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0428e8b97170aab075de0082d8b216547f000724","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-28 22:24:19.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"dbec29f20b8920c6ad90dc09f5341cad136f1b5f","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2014-02-28 22:24:28.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":2},{"id":"4f4801cc020d455193354a91330f8d4c5ef0b0aa","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2014-02-28 22:34:44.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(3 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"65403b15d699edb2d47854997caa293afa3a0e07","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2014-02-28 22:38:29.000000000","message":"Patch Set 2: No score\n\n(3 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"549f0a338652318b9b195ffc545d083c04ab55b4","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2014-02-28 22:38:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d929c2bacd057b96980dcc5750c08b4774cc8e6c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-28 22:38:43.000000000","message":"Patch Set 3:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"a353fc2c51c6fc24961f5fc24e551b24ea64b917","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-02-28 23:34:10.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/86/76686/3/check/gate-oslo.messaging-pep8/98d637a : SUCCESS in 58s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/86/76686/3/check/gate-oslo.messaging-docs/658919f/doc/build/html/ : SUCCESS in 1m 18s\n- gate-oslo.messaging-python26 http://logs.openstack.org/86/76686/3/check/gate-oslo.messaging-python26/028d553 : SUCCESS in 49s\n- gate-oslo.messaging-python27 http://logs.openstack.org/86/76686/3/check/gate-oslo.messaging-python27/995e5ae : SUCCESS in 1m 24s\n- check-devstack-dsvm-cells http://logs.openstack.org/86/76686/3/check/check-devstack-dsvm-cells/d00c3dc : SUCCESS in 12m 50s\n- check-tempest-dsvm-full http://logs.openstack.org/86/76686/3/check/check-tempest-dsvm-full/38961eb : SUCCESS in 52m 00s\n- check-tempest-dsvm-neutron http://logs.openstack.org/86/76686/3/check/check-tempest-dsvm-neutron/cc36c88 : SUCCESS in 33m 28s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/86/76686/3/check/check-tempest-dsvm-neutron-full/cc3b33c : SUCCESS in 44m 51s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/86/76686/3/check/gate-tempest-dsvm-large-ops/fa836d9 : SUCCESS in 19m 26s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/86/76686/3/check/gate-tempest-dsvm-neutron-large-ops/faae23e : SUCCESS in 17m 23s\n- check-grenade-dsvm http://logs.openstack.org/86/76686/3/check/check-grenade-dsvm/41fd46d : SUCCESS in 32m 47s\n","accounts_in_message":[],"_revision_number":3},{"id":"1b7b8a80a15c938022da416cb2e6ddc2bf6c9feb","author":{"_account_id":3189,"name":"Chet Burgess","email":"cfb@metacloud.com","username":"cfb-n"},"date":"2014-03-01 00:08:37.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"cb19d8fb2055895ffcc6734964e7cdb216fccbac","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2014-03-01 13:35:47.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"b02028f0f941f800f9e7149d2f4ba38e5ab16a38","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 02:28:51.000000000","message":"Patch Set 3: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"a9a8bbd4c893993398c3ae60ce2daa14f6b57199","author":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"date":"2014-03-04 02:28:51.000000000","message":"Patch Set 3: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":3},{"id":"85e99a022c26482c0e2c4f5fb5fb6719922d097f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 03:37:40.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- gate-oslo.messaging-pep8 http://logs.openstack.org/86/76686/3/check/gate-oslo.messaging-pep8/26814ec : SUCCESS in 1m 09s\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/86/76686/3/check/gate-oslo.messaging-docs/72aca1b/doc/build/html/ : SUCCESS in 1m 23s\n- gate-oslo.messaging-python26 http://logs.openstack.org/86/76686/3/check/gate-oslo.messaging-python26/fc85805 : SUCCESS in 1m 01s\n- gate-oslo.messaging-python27 http://logs.openstack.org/86/76686/3/check/gate-oslo.messaging-python27/4adddd7 : SUCCESS in 1m 50s\n- check-devstack-dsvm-cells http://logs.openstack.org/86/76686/3/check/check-devstack-dsvm-cells/150441c : SUCCESS in 11m 14s\n- check-tempest-dsvm-full http://logs.openstack.org/86/76686/3/check/check-tempest-dsvm-full/5d2ce74 : SUCCESS in 1h 00m 54s\n- check-tempest-dsvm-neutron http://logs.openstack.org/86/76686/3/check/check-tempest-dsvm-neutron/e6b9dc0 : SUCCESS in 35m 56s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/86/76686/3/check/check-tempest-dsvm-neutron-full/646e580 : SUCCESS in 42m 08s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/86/76686/3/check/gate-tempest-dsvm-large-ops/4a44f2e : SUCCESS in 17m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/86/76686/3/check/gate-tempest-dsvm-neutron-large-ops/cb96f7b : SUCCESS in 16m 25s\n- check-grenade-dsvm http://logs.openstack.org/86/76686/3/check/check-grenade-dsvm/255f9ec : SUCCESS in 46m 30s\n","accounts_in_message":[],"_revision_number":3},{"id":"0acaafafe2fc70c2a508a05254a6012915b3575d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 03:37:41.000000000","message":"Patch Set 3: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"fa1086f32c22389a8c14394bb97b52200e36c225","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 04:33:25.000000000","message":"Patch Set 3: Verified\n\nBuild succeeded.\n\n- gate-oslo.messaging-docs http://docs-draft.openstack.org/86/76686/3/gate/gate-oslo.messaging-docs/0569970/doc/build/html/ : SUCCESS in 1m 07s\n- gate-oslo.messaging-pep8 http://logs.openstack.org/86/76686/3/gate/gate-oslo.messaging-pep8/ee09cf9 : SUCCESS in 52s\n- gate-oslo.messaging-python26 http://logs.openstack.org/86/76686/3/gate/gate-oslo.messaging-python26/e83450b : SUCCESS in 1m 23s\n- gate-oslo.messaging-python27 http://logs.openstack.org/86/76686/3/gate/gate-oslo.messaging-python27/c2d60d5 : SUCCESS in 1m 17s\n- gate-devstack-dsvm-cells http://logs.openstack.org/86/76686/3/gate/gate-devstack-dsvm-cells/ececaf1 : SUCCESS in 10m 36s\n- gate-tempest-dsvm-full http://logs.openstack.org/86/76686/3/gate/gate-tempest-dsvm-full/cff5e09 : SUCCESS in 54m 09s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/86/76686/3/gate/gate-tempest-dsvm-neutron/e9c092a : SUCCESS in 32m 16s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/86/76686/3/gate/gate-tempest-dsvm-large-ops/f4b7054 : SUCCESS in 13m 49s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/86/76686/3/gate/gate-tempest-dsvm-neutron-large-ops/ff58436 : SUCCESS in 28m 32s\n- gate-grenade-dsvm http://logs.openstack.org/86/76686/3/gate/gate-grenade-dsvm/b030799 : SUCCESS in 31m 03s\n","accounts_in_message":[],"_revision_number":3},{"id":"52f6926cc2314ea19f0f507cdb20c3cfbef2b6f0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-04 04:33:25.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fcd51a67d18a9e947ae5f57eafa43ac756d1a5a8","revisions":{"a51df4c336b340b8748b20428120e1c19c434abd":{"kind":"REWORK","_number":1,"created":"2014-02-26 23:23:08.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/86/76686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/86/76686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/1"}}},"commit":{"parents":[{"commit":"7564e2cf4796167550df1eb8c09fad9b5d687dce","subject":"Merge \"User a more accurate max_delay for reconnects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7564e2cf4796167550df1eb8c09fad9b5d687dce"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2014-02-26 23:21:01.000000000","tz":-480},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2014-02-26 23:21:01.000000000","tz":-480},"subject":"Slow down Kombu reconnect attempts","message":"Slow down Kombu reconnect attempts\n\nFor a rationale for this patch, see the discussion surrounding Bug\n\nWhen reconnecting to a RabbitMQ cluster with mirrored queues in\nuse, the attempt to release the connection can hang \"indefinitely\"\nsomewhere deep down in Kombu.  Blocking the thread for a bit\nprior to release seems to kludge around the problem where it is\notherwise reproduceable.\n\nChange-Id: Ic2ede3046709b831adf8204e4c909c589c1786c4\nPartial-Bug: 856764\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a51df4c336b340b8748b20428120e1c19c434abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a51df4c336b340b8748b20428120e1c19c434abd"}]},"branch":"refs/heads/master"},"d455e42391be649d3b1c7852e68dcf5f9dd68874":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2014-02-28 22:24:18.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/86/76686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/86/76686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/2"}}},"commit":{"parents":[{"commit":"0400cbf4f83cf8d58076c7e65e08a156ec3508a8","subject":"Gracefully handle consumer cancel notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0400cbf4f83cf8d58076c7e65e08a156ec3508a8"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2014-02-26 23:21:01.000000000","tz":-480},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2014-02-28 22:24:07.000000000","tz":-480},"subject":"Slow down Kombu reconnect attempts","message":"Slow down Kombu reconnect attempts\n\nFor a rationale for this patch, see the discussion surrounding Bug\n\nWhen reconnecting to a RabbitMQ cluster with mirrored queues in\nuse, the attempt to release the connection can hang \"indefinitely\"\nsomewhere deep down in Kombu.  Blocking the thread for a bit\nprior to release seems to kludge around the problem where it is\notherwise reproduceable.\n\nChange-Id: Ic2ede3046709b831adf8204e4c909c589c1786c4\nPartial-Bug: 856764\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d455e42391be649d3b1c7852e68dcf5f9dd68874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d455e42391be649d3b1c7852e68dcf5f9dd68874"}]},"branch":"refs/heads/master"},"fcd51a67d18a9e947ae5f57eafa43ac756d1a5a8":{"kind":"REWORK","_number":3,"created":"2014-02-28 22:38:42.000000000","uploader":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"ref":"refs/changes/86/76686/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/86/76686/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/86/76686/3"}}},"commit":{"parents":[{"commit":"0400cbf4f83cf8d58076c7e65e08a156ec3508a8","subject":"Gracefully handle consumer cancel notifications","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0400cbf4f83cf8d58076c7e65e08a156ec3508a8"}]}],"author":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2014-02-26 23:21:01.000000000","tz":-480},"committer":{"name":"Nicolas Simonds","email":"nic@metacloud.com","date":"2014-02-28 22:37:51.000000000","tz":-480},"subject":"Slow down Kombu reconnect attempts","message":"Slow down Kombu reconnect attempts\n\nFor a rationale for this patch, see the discussion surrounding Bug\n\nWhen reconnecting to a RabbitMQ cluster with mirrored queues in\nuse, the attempt to release the connection can hang \"indefinitely\"\nsomewhere deep down in Kombu.  Blocking the thread for a bit\nprior to release seems to kludge around the problem where it is\notherwise reproduceable.\n\nDocImpact\n\nChange-Id: Ic2ede3046709b831adf8204e4c909c589c1786c4\nPartial-Bug: 856764\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fcd51a67d18a9e947ae5f57eafa43ac756d1a5a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fcd51a67d18a9e947ae5f57eafa43ac756d1a5a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
