)]}'
{"id":"openstack%2Foslo.messaging~766946","triplet_id":"openstack%2Foslo.messaging~master~I00872541844323a95ffb5ef60ca41cf36a75560c","project":"openstack/oslo.messaging","branch":"master","topic":"fix-lc","hashtags":[],"change_id":"I00872541844323a95ffb5ef60ca41cf36a75560c","subject":"Update requirements and lower-constraints","status":"ABANDONED","created":"2020-12-14 12:58:47.000000000","updated":"2021-08-05 14:18:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0164b4a5163576bc6416874c735d424883011380","_number":766946,"virtual_id_number":766946,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-14 16:23:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-14 16:23:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c75c13d803d738dd2f59ac55b9fedbc086f8c879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-14 12:58:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff541a98617b8cbefae9f4b4a617e3a020ea81a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-14 14:22:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b22a7f877dba41ec7e86486aa4d0aa016c01f5f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 16:23:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6cefd800354948f0b1c5416f04fc5577 : FAILURE in 3m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e8bf591f8604114a76db7241d12c9c0 : SUCCESS in 1h 23m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/296f6f37b56d4bd2ab8bfaf468ca0585 : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c746c4d8c5534244a1c6ef3f1d77721f : FAILURE in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721a3c9d78634fd583fe9d465343476c : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0ecf384644e45b191a0773f9b5fa7dc : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8236bada390442cdb3a81b3febaa550b : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9fb25e96946441b99a0262b745ae92a : FAILURE in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55a2c6049304b09aebac7a05eda94fd : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8546fea37fd43e4a710dba0f58cdc97 : SUCCESS in 4m 00s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/d6be80c862bd4e24b50292c45441710a : SUCCESS in 5m 21s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/9fa711b8cf3e47dc82c164f57236e1d1 : FAILURE in 6m 40s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/e6669e80db744087a5779cb15ac38ac6 : SUCCESS in 5m 57s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/ef6f2a84c81b49d4915f7ac43aa4c78c : FAILURE in 6m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f63670e5ff5b452f830bcdbd40c3e4ee : SUCCESS in 1h 40m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2eb93894ecf94a2b9aa379b8c02f4a92 : SUCCESS in 1h 51m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ff1901ac7e3f479f9a372f84ca65c809 : SUCCESS in 1h 38m 28s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/afd7f75699fa4050bcc9c9926f43a7cc : SUCCESS in 57m 12s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/01cbeddcc9cd4ee2b738690849607e67 : SUCCESS in 1h 04m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0164b4a5163576bc6416874c735d424883011380","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-05 14:18:42.000000000","message":"Abandoned\n\nNot needed anymore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e734bcbe0d4559267f241941d085a7a1e28c9a34","revisions":{"c68df3c5157ffd6023ecb59f5d674a59f4fb3290":{"kind":"REWORK","_number":1,"created":"2020-12-14 12:58:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/766946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/46/766946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/766946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/766946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/766946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/46/766946/1"}}},"commit":{"parents":[{"commit":"54f8724be281b1acfae175d7ec1c14e439269117","subject":"Fix doc title rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/54f8724be281b1acfae175d7ec1c14e439269117"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-14 12:58:00.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-14 12:58:00.000000000","tz":60},"subject":"Update requirements and lower-constraints","message":"Update requirements and lower-constraints\n\nBring the versions specified in requirements.txt and test-req.txt\ncloser to what\u0027s actually being used in the py36 and py38 testenvs,\nand update lower-constraints to reflect the updated requirements.\n\nWhat has made this necessary right now is that the latest pip has\nintroduced a much stricter resolver, and the oslo lower-constraints\njob is failing because oslo is imposing some constraints which are\ntoo low for some dependencies, and we can\u0027t get away with that under\nthe stricter resolver.\n\nChange-Id: I00872541844323a95ffb5ef60ca41cf36a75560c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c68df3c5157ffd6023ecb59f5d674a59f4fb3290"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c68df3c5157ffd6023ecb59f5d674a59f4fb3290"}]},"branch":"refs/heads/master"},"e734bcbe0d4559267f241941d085a7a1e28c9a34":{"kind":"REWORK","_number":2,"created":"2020-12-14 14:22:01.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/46/766946/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/46/766946/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/766946/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/766946/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/46/766946/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/46/766946/2"}}},"commit":{"parents":[{"commit":"54f8724be281b1acfae175d7ec1c14e439269117","subject":"Fix doc title rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/54f8724be281b1acfae175d7ec1c14e439269117"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-14 14:21:02.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-14 14:21:02.000000000","tz":60},"subject":"Update requirements and lower-constraints","message":"Update requirements and lower-constraints\n\nBring the versions specified in requirements.txt and test-req.txt\ncloser to what\u0027s actually being used in the py36 and py38 testenvs,\nand update lower-constraints to reflect the updated requirements.\n\nWhat has made this necessary right now is that the latest pip has\nintroduced a much stricter resolver, and the oslo lower-constraints\njob is failing because oslo is imposing some constraints which are\ntoo low for some dependencies, and we can\u0027t get away with that under\nthe stricter resolver.\n\nChange-Id: I00872541844323a95ffb5ef60ca41cf36a75560c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e734bcbe0d4559267f241941d085a7a1e28c9a34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e734bcbe0d4559267f241941d085a7a1e28c9a34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
