)]}'
{"id":"openstack%2Foslo.messaging~768252","triplet_id":"openstack%2Foslo.messaging~master~I01a38754a31c891f2b3b9c7f8135690693df5d13","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1909036","hashtags":[],"change_id":"I01a38754a31c891f2b3b9c7f8135690693df5d13","subject":"Fix type of direct_mandatory_flag opt","status":"MERGED","created":"2020-12-22 19:36:34.000000000","updated":"2021-01-06 18:31:29.000000000","submitted":"2021-01-06 18:30:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768252-bug/1909036","meta_rev_id":"b9d8b17ebd569afe57712794d3d369d65b84a1c1","_number":768252,"virtual_id_number":768252,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-06 18:30:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2021-01-06 16:19:19.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2021-01-05 19:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2021-01-06 16:19:19.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2021-01-06 16:18:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 21:24:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-05 19:04:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-06 16:19:19.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"}],"messages":[{"id":"1785e8ad117c9557c6c50dd4103f09333fcb65c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-12-22 19:36:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de8806052d6cd7bf93c7beaa2f47e3adc23fd10f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 21:24:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcf31513796f4e0ca95d8aed86143837 : SUCCESS in 1h 23m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/903b067118b1430bb2523345afbcdaca : SUCCESS in 6m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac58139afc5d4b33b4acc022c6f21e17 : FAILURE in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d8a30a238914cb8a30b9b2234ee29af : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdb7a155e884427c8388f758f1e9d71a : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0679ae0525434199bc29a82878322ab2 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eb682c0d588448186d621f537b9d6ed : FAILURE in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3be57220ee34daca28ddbeb2c572963 : SUCCESS in 4m 08s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/96ca733e599e4ad3b717e65be139a22c : SUCCESS in 4m 36s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/cac120aa90d5433e961762bbce6fe0a9 : FAILURE in 7m 26s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/da0284eb2f68413485d6caa3185482cd : SUCCESS in 5m 23s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/63f055f9d29b4f6b84289c4f4825df6f : FAILURE in 7m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/05d6971096b744108e2d458d4c81daec : SUCCESS in 1h 26m 01s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d18e0e85263e40bb8b1e55460bff3c2d : SUCCESS in 1h 35m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/dc4ec9fe73d04c999040831c256f18ef : SUCCESS in 1h 41m 18s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/12e72cfa7f9a4cb8848877503ddfd475 : SUCCESS in 1h 09m 38s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1cecff51e18b41eeb5ad8255d0a383e9 : SUCCESS in 1h 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbd0e99559ec4166a44978856ee061e4d3b1c4fc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-05 19:04:39.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks Ben!\n\nLGTM, LC job must be dropped first to allow us to continue however I was thinking it was only triggered when *reqs.tx or LC files were modified.","accounts_in_message":[],"_revision_number":1},{"id":"eabfa198dcc5a5faf0d018133830b03d0e2cd763","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-06 13:14:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8aafb8ea7453319ab707f0f471079d97a099406d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 15:06:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1893a6ac6044e7080e2642191828b5c : SUCCESS in 1h 36m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b42318ade1ba4973a0fb0b3fa1c4bf63 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09a78d3a39b04d578ab61748ba42e748 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be9d21d42776425fb18b1f96db30f1eb : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54cea786d2ec4954be458d830879c0a7 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ac28a4456dc4dd6ad70a85b2415473f : FAILURE in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83adfc3e1184d85962bad758677cf6c : SUCCESS in 4m 18s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/2be27e51f262442e85a8b875f00802d7 : SUCCESS in 5m 15s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/0b5d52405e7e459fbfcf4d3258313933 : FAILURE in 5m 51s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/d7dc032fd98b48fb8bfc529ee342887b : SUCCESS in 5m 15s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/e29f34bf3067451fbc464799e3776907 : FAILURE in 7m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3870969692084e429459312bff9e20ff : SUCCESS in 1h 34m 24s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d42fb4ce7eaa482cb203d0616295f30b : SUCCESS in 1h 40m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a40e08fae83f43ec92320e3883cf789c : SUCCESS in 1h 33m 37s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4a5aa0ac16f24c5bba70c6d240b4259a : SUCCESS in 57m 21s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2dcc8956558742268f429163f3bbdfc8 : SUCCESS in 1h 06m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a74496f399e1a03103732c25eb042783959239ed","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-06 16:18:46.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"31bbd773aef22dc2e8b9dfa39828ddd6fff7f815","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-01-06 16:19:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9aa98d7d25232c0f5e8c07f2a77ef44c90b5b29b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 16:19:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ff4fa66a89cde9e8e96d0616fc9f65b2f86dfce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 18:30:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c629cfc08b8e41938df5c4d56163a843 : SUCCESS in 1h 33m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/037bee0dec2d4b7cb6b02c7d222f991d : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02bbbd24f7284f4f829bfb179c31fcef : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be367a9f8cbd43b58778a6b0f805caad : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44323081035e4eff9243828684ce7706 : SUCCESS in 4m 17s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/2f04dc2f73cc4838bfb560d9903cb81e : SUCCESS in 5m 34s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ab89b302d1f249719afa25ede2b227c8 : SUCCESS in 2h 07m 23s","accounts_in_message":[],"_revision_number":1},{"id":"900545cf12a53105659b112f817fe4ecfd6b50f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 18:30:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b9d8b17ebd569afe57712794d3d369d65b84a1c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 18:31:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/032ecb48c0914748a2e42eda91498850 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8f8b17030e93f6fe64e986ebae1de4011f007d5","revisions":{"b8f8b17030e93f6fe64e986ebae1de4011f007d5":{"kind":"REWORK","_number":1,"created":"2020-12-22 19:36:34.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/52/768252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/52/768252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/52/768252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/52/768252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/52/768252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/52/768252/1"}}},"commit":{"parents":[{"commit":"54f8724be281b1acfae175d7ec1c14e439269117","subject":"Fix doc title rendering","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/54f8724be281b1acfae175d7ec1c14e439269117"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-12-22 19:34:16.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-12-22 19:36:14.000000000","tz":0},"subject":"Fix type of direct_mandatory_flag opt","message":"Fix type of direct_mandatory_flag opt\n\nAn IntOpt with a default of True is invalid. I\u0027m a little surprised\nthis doesn\u0027t fail a defaults check somewhere, but it needs to be\nfixed regardless.\n\nLooking at where it is used, it appears the boolean type is correct.\nThis just changes the opt type to BoolOpt to match.\n\nChange-Id: I01a38754a31c891f2b3b9c7f8135690693df5d13\nCloses-Bug: 1909036\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b8f8b17030e93f6fe64e986ebae1de4011f007d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b8f8b17030e93f6fe64e986ebae1de4011f007d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
