)]}'
{"id":"openstack%2Foslo.messaging~769008","triplet_id":"openstack%2Foslo.messaging~master~Ib2b816728307166450a4cea2ccdb3c4b550a0713","project":"openstack/oslo.messaging","branch":"master","hashtags":[],"change_id":"Ib2b816728307166450a4cea2ccdb3c4b550a0713","subject":"remove unicode from code","status":"MERGED","created":"2021-01-03 08:11:54.000000000","updated":"2021-02-03 15:08:25.000000000","submitted":"2021-02-03 15:04:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769008","meta_rev_id":"061f9476ddc9365e66ed234d6c54d5e9f5e730e2","_number":769008,"virtual_id_number":769008,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-03 15:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2021-01-18 14:17:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2021-02-03 11:47:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2021-02-03 11:47:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-03 09:49:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 14:17:17.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2021-02-03 11:47:07.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"ca7cae18bd285eb868509dd1408aa51adacaafa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-01-03 08:11:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24cc2c3ab0d6013bbbf6a22d44180a856c5206d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-03 09:49:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64cd6da7489041f488822f293b765cc5 : SUCCESS in 1h 21m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b8db841deda4dfb98d94f191d77ed5d : SUCCESS in 8m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/207922a789cc44b9bc192fd0b849cd98 : FAILURE in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75477c301584f78ad6f760f4440a3ce : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5d30ce98eb449ce859ad1c8057436ad : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/752885bc05e641a58b80405ecde6c586 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55ef8f58c530443bbc85f3e5d80770da : FAILURE in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3fb6dd6430b43b8996d5e051dd3ce9f : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a4c8ea4323e4580817053e9b91c1987 : SUCCESS in 5m 01s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/5fe36d6d482c4ac89ddc8c2d1a04dede : SUCCESS in 7m 06s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/cbe95bee800242828cf54501197e506c : FAILURE in 5m 51s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/231264f8ed4846ec88ded97b74d29ee8 : SUCCESS in 6m 07s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/7c211395368f42928255d01f5a59fb39 : FAILURE in 6m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3e4d45d861e341448dc11c4e8200dc6e : SUCCESS in 1h 25m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d8ddb73d4b9e43488fa679c598685b95 : FAILURE in 1h 34m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/fc3e3d79a7ef4bc19338a8786f8b052f : SUCCESS in 1h 23m 16s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/db59cf3998d54af9b281466e7ce02e3a : SUCCESS in 55m 45s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d7cfbc8d4024a94bbed758c7a031862 : SUCCESS in 1h 06m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12383c8119a841502fbd471cad82bbdd3adb9a12","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-01-05 07:36:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b013d8f2d8211972406b89d17771c25e7c5dfb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 09:43:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47098b8f80114d7a920fa69b1c9a2495 : SUCCESS in 1h 19m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/272623df525549fa89bc2854c82070fb : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4e0ba9e0727492f941317fe1ae92509 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0611b790a84b2fb1d4ab595e98af00 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/698b49450ebe49d0a42cbaa0014ed716 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43a26f531f324b8fb73f2f9a6600f111 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c5d379759a34cf1b249192b4820ad68 : FAILURE in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643c4c8b87804c2b985046911385be0b : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aae646b93c614f84adf80d8f725145ff : SUCCESS in 4m 12s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/52a4ad0d12a14c4e9a38b5c217d60341 : SUCCESS in 5m 33s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/f3f20ce9b84641acbd985bbede270d2e : FAILURE in 5m 50s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/2e2870d2f6b545c5a47ce51bdce8ad5d : SUCCESS in 5m 32s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/e09abd2b6dda4ccbb8308c2f88eef225 : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4a81fcfe0c7c4bcf909780bbb077e25e : SUCCESS in 1h 38m 31s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ef3ea30f89084690a28b8ffab0a97e04 : SUCCESS in 1h 32m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/63be898d27024dbb914757cc36abcc82 : SUCCESS in 2h 05m 19s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5326221703a64527a6c7345dfe245b99 : SUCCESS in 1h 08m 11s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac3366a8e5184fb79e8857cee875cd58 : SUCCESS in 1h 11m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12404cf5bfb01d153acc6ffb554fc6b3266bfce2","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2021-01-06 05:46:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b29b005776283cb0f3d061d854ec61a4da71e254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 07:29:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/716a74800c92457dae15757a763f4630 : SUCCESS in 1h 31m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a199b47caee49609fb4948a3f32085a : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/531156da664f49e5af7b39ca94e02cf3 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f5ed4d36c1447dcba04df2f2e90eac1 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56b1d929f67541928b9c39acb3511b23 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f0d7bb5247840e9b3b13fadcbee9860 : FAILURE in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5831b56bdcee40189411e985f6cef37b : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94e39a983a814fada0e2696eb16de349 : SUCCESS in 3m 02s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/91a061a592b54722a50068d606204080 : SUCCESS in 4m 16s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/05c9a40ca92a44b8942ef12a94c95e66 : FAILURE in 6m 11s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/bbf36512fc6e4927be285f89fafbc303 : SUCCESS in 4m 58s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/05229efb255d49eb90999be5b28e29bc : FAILURE in 5m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/39963107ecb44d0ca6dd42024240bafd : SUCCESS in 1h 22m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ae6ee2b7d8734d9c9cfa2e04b5f4b1b6 : SUCCESS in 1h 35m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6d41ae3e070b4b34b1735cbce1226f7c : SUCCESS in 1h 37m 19s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/499b57a98ce347eea2680c8763a0576e : SUCCESS in 1h 04m 35s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4054dcf6bac4435ba37823345e7baf0 : SUCCESS in 1h 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e361fac60edce5e2d0652335a8e103228da28c3","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-01-18 14:17:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a1234d869a3c53bffd6c6af359c00f503168a3ab","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 11:47:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b45fdd819c4e603fa9d27a8d498f66ee6813db0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 11:48:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d4bdab47b75e08c0719e852bc619dec309b5244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:04:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/596d2763ed5d4d3695c7436be92bfa6f : SUCCESS in 1h 26m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3f37d1c37144e02820f41b29c8d36e0 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81a81c52c88d45608e0f47c5fc183c05 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70d9af34435448598e7799c775b3610d : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb75e300be74f3e908a5dbf23746eaa : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b01c7160a59b4d799dd1505be70c4795 : SUCCESS in 4m 15s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/141d3f77a85c484392c1a9f08d4b9c57 : SUCCESS in 5m 22s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/58d005d70bb844889aba0cc2b1441300 : SUCCESS in 1h 44m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ecc1cd8cbeb526bfd3299e4c9019b6f4526d118a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:04:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0fe5bc2620609d9049696d217a2e71f773f5dc32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:07:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff9e9070830b47d08861661447133038 : SUCCESS in 1m 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fca3ac5cf81419d829b2f57ad1250a0 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":1},{"id":"061f9476ddc9365e66ed234d6c54d5e9f5e730e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:08:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff9e9070830b47d08861661447133038 : SUCCESS in 1m 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fca3ac5cf81419d829b2f57ad1250a0 : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"642367cdfd14bf9507116eb263c454b8b4e2d6e1","revisions":{"642367cdfd14bf9507116eb263c454b8b4e2d6e1":{"kind":"REWORK","_number":1,"created":"2021-01-03 08:11:54.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/08/769008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/08/769008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/08/769008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/08/769008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/08/769008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/08/769008/1"}}},"commit":{"parents":[{"commit":"5560e8e6717b2056b727626ef252d5d3b3f4f1a0","subject":"Remove six","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5560e8e6717b2056b727626ef252d5d3b3f4f1a0"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2021-01-03 08:11:46.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2021-01-03 08:11:46.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: Ib2b816728307166450a4cea2ccdb3c4b550a0713\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/642367cdfd14bf9507116eb263c454b8b4e2d6e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/642367cdfd14bf9507116eb263c454b8b4e2d6e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
