)]}'
{"id":"openstack%2Foslo.messaging~771232","triplet_id":"openstack%2Foslo.messaging~master~Id5cddbefbe24ef100f1cc522f44430df77d217cb","project":"openstack/oslo.messaging","branch":"master","topic":"bug/1905965","hashtags":[],"change_id":"Id5cddbefbe24ef100f1cc522f44430df77d217cb","subject":"Correctly handle missing RabbitMQ queues","status":"MERGED","created":"2021-01-18 15:12:15.000000000","updated":"2021-02-09 18:04:14.000000000","submitted":"2021-02-09 18:01:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":19,"has_review_started":true,"submission_id":"771232-bug/1905965","meta_rev_id":"f8e7083d040c2a0e0251e227cf3f38f19f47568b","_number":771232,"virtual_id_number":771232,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-09 18:01:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":2,"date":"2021-02-09 16:27:59.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2021-02-04 09:48:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-09 13:23:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2021-02-09 16:28:52.000000000","_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 15:28:48.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2021-01-18 17:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 14:16:51.000000000","updated_by":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"reviewer":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"state":"REVIEWER"},{"updated":"2021-02-02 12:04:38.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2021-02-03 11:44:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"bd4f25e4be131c503863f953315ab7739fcbc6e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-18 15:12:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f57e231ce73eb11f17b0617845954c17f66e2ad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:00:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6c1c850d04f4eaeb4bd4451cbd224e9 : SUCCESS in 1h 26m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff8236c1b33746b68336fc18d2980ce0 : TIMED_OUT in 51m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d85d11901a3c4e1282b86f9f01dd1fdb : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d3e7d0f3dc246fe8e17a65fcf78896d : TIMED_OUT in 41m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eea1c223389f43c08517e1dd731eea11 : TIMED_OUT in 41m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e062af11c78e45b9902270683bb9755a : FAILURE in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fbf2aca08714f56b565c56f5784e2b6 : SUCCESS in 4m 54s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/aae36f32d22a4ade95199a6c4201565a : SUCCESS in 5m 30s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/85f952b16445494c814f087a39dea618 : FAILURE in 7m 58s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/979ccf8a7ab042bda8d33400dedf96f3 : FAILURE in 6m 26s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/182b8281445949178282052552afe190 : FAILURE in 7m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7dc634a50d1543d0b78174e366421e50 : SUCCESS in 1h 39m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4ae586e72c0447bc9e404820fa40b8aa : FAILURE in 34m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/dec9ec4a49c7437f9d4b8beafcafad98 : SUCCESS in 1h 31m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/58b15bf1c4fa4371b93a4388d896a6d4 : SUCCESS in 1h 03m 50s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e55232f2b8344e592e2d7360bff7807 : SUCCESS in 1h 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edabf70263f3b559f9af663730fdfae9ef8f9692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-21 09:18:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76b7cf71155291fd996d171645ffb9b0f47f2b91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 10:55:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/728f74fbf5bc4723ac262a4fc1aaf0ff : SUCCESS in 1h 25m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31f8737a1226442daa2dcf54b869e469 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bab0400179149af86dccc1786e7e1c5 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2502f593d8fb41909069ecfc1682a6b0 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0835e36151ca49beba901ff9d1017509 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61acfa288faa43b28ffec917057c5028 : FAILURE in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859bc9fa37a94c23a8270723acb67d32 : SUCCESS in 4m 23s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/9171633c4bd747598fa745bbf99c0bfb : SUCCESS in 5m 47s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/56b28902915d44f5aa6ad6694a619349 : FAILURE in 6m 42s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/c840df0db85d4ba1a7e84075f0d8cf70 : SUCCESS in 6m 41s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/723bcc58a27041a49972388f57507eb2 : FAILURE in 6m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/16bee12dd5154bf984cc662087253997 : SUCCESS in 1h 30m 22s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7d6236b5622844c1b405028cc416fc46 : SUCCESS in 1h 23m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b7dfc44b56cb46bf9c7931f45c775a9d : SUCCESS in 1h 28m 26s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/886ced7620404783b5b141a1d3df4aa0 : SUCCESS in 1h 05m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/280024496760426dbd65392806fe6e00 : SUCCESS in 1h 10m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ac5f26a3356669364741f22f57f6f8b1e30dfbd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-21 13:37:05.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nHello,\n\nI\u0027ll try to add related unit test.\n\nNotice that these changes are mostly to open the debate on how to continue to use the mandatory flag to get informed about the missing queues.\n\nSee my inline comments","accounts_in_message":[],"_revision_number":2},{"id":"40689fe867ffcb96214bd5a55ea9b6376828be34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 10:16:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3bde5b32e61fcb5dc764d0777da7ccca14684b3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 10:21:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"845348ce7b212799b82fc0de1ebad5ce90350746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 10:27:59.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"1f5b4da718b01b291b95986dd187e90ae126af3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 10:53:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"00fa24ad3a5bf8675775875544dfdf99b9cc4f33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 11:07:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"529dc5330f744007f30b87caefe077884930a2c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 13:15:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ccbc7916d6294eac89d79674847d2319 : SUCCESS in 1h 37m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0bac1e3f1c64da7afea02762eaaefc4 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6caf0d77dc91436f9b44a8df6b305a92 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c801ff255bdd40ed9cb246ad84643530 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6f96c3a996f4f87821df857102b1a14 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d2e6c76ff694414a7283fef3ad33b3c : FAILURE in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e111f5d62e43de94a9446465ce982d : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8823440b1c1742ee972db12221119485 : SUCCESS in 5m 12s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/7371308cf9644d9d94c8e7b59d9a799e : SUCCESS in 6m 29s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/da1208f71fd645da95c3884820f52446 : FAILURE in 6m 22s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/9fd21db2529541c7baee38283e1e7217 : SUCCESS in 6m 36s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/c3373713bf8d49dfb0bda877b7378ccd : FAILURE in 6m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/58a7da57b9f2423cae329a2901ca7d10 : SUCCESS in 1h 31m 58s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/905b50c55af148eeb3e22d24b12775cb : SUCCESS in 1h 23m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/536a0705a4a4451d95c96dfc062fcf93 : SUCCESS in 1h 49m 21s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/dfde8ac02097441a8b81c9ebf5388480 : SUCCESS in 1h 02m 56s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ec4facf72a34b288564009b3e1adcdc : SUCCESS in 1h 09m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"39a1b9d074fbef595b045a5ebe09ddb15de836f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 13:36:57.000000000","message":"Patch Set 7:\n\nYou can test these changes by using the simulator (cf [1]).\n\n[1] https://bugs.launchpad.net/oslo.messaging/+bug/1905965/comments/8","accounts_in_message":[],"_revision_number":7},{"id":"56b3dacfc9b282f2b839bd24b9c34cf8e7ed650d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 13:40:31.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"925ba5fa6c545b112431e80223aa67c2d8e52cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 15:29:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d021f06a32d740538dddfa946c22212b : SUCCESS in 1h 21m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ded64eaa38db4860931f25bf5cdc4051 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0917438cbfdb4ffbb74c54d1fc43bcf3 : SUCCESS in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b927e7026b0444909139d9f4dbc9b452 : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d0e5c487a3a41c8b36743ae9a609ae8 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec4bdaadeba74ff9b351625002739ad5 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d3694b39a644351a62fecc93ab6a1b4 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34ff9a88f61940209ba95c3ce0060652 : SUCCESS in 3m 46s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/ef9ce97e739d47c385632a3636a0649b : SUCCESS in 4m 48s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/ee7321a2ad0b40a59edc9bba7c11bbbe : FAILURE in 6m 47s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/6e38b2ed024f4d209219552b7e847277 : SUCCESS in 7m 02s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/6ac4a077806e4f10a91e56c741a94f33 : FAILURE in 7m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b568ae66f4234132ac3067b4cf7f09ff : SUCCESS in 1h 43m 40s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3ff3c411637d4637b5d80f707fe89d8e : SUCCESS in 1h 31m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1db35c47d6d74658950de59c7cad7f67 : SUCCESS in 1h 28m 14s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/df2b924299994efebff1295380c5bea3 : SUCCESS in 58m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b927955c52d34565b64b429f273f76c5 : SUCCESS in 1h 14m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"eb4050c9237b9d7b6c1e5f42799aba8435dc0a3d","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-01-25 16:17:43.000000000","message":"Patch Set 8: Code-Review+2\n\nI\u0027m fine with using the obsolete reply queue list for exchanges as well.  Seems like the least risky approach.","accounts_in_message":[],"_revision_number":8},{"id":"ca9dc78cbea80217b2c026c044209c3fe7785122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 20:04:38.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"81846391b7c0d9a7c697d19a18d739b2aac651c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 20:06:13.000000000","message":"Patch Set 9:\n\nPatch rebased to deprecate the option by using oslo.config is deprecation features.","accounts_in_message":[],"_revision_number":9},{"id":"29cb7459d091098052d4a996006dd5f16b840c83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 22:24:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f27ba82a3bfa4d169f6b596ee6847045 : SUCCESS in 2h 02m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e7b32ab70ef4346a300115a79457bc2 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06ac3ae5a1974d899be074bab79a97cc : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6596ff803c9a4bb2b6c75a5e234248fb : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/752d32f220dd448f9267ff8e5cc06c51 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db899a9b30a243eea822204560b6093a : FAILURE in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad2f3ec173d6442cbde14b3168822412 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7f7e447450c4df3aeec8d12bf293746 : SUCCESS in 3m 30s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/f23efb4cf4c941bbaf7158ef5ce2e138 : SUCCESS in 6m 15s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/19e98c110edf48ee8aae38b2ae84141d : FAILURE in 7m 05s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/9d638a5a3bab4674a7dbba9378acfd9b : SUCCESS in 5m 15s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/e61ccbf158f24db2b42b4832e0a9a117 : FAILURE in 6m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e06937907d7e4b92b1fd03d72ab35497 : SUCCESS in 1h 21m 12s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d1b1cc8f7ddd4efb8b7f9a0efff9c9d4 : SUCCESS in 1h 35m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8fa8017eab294de09769b83f3525d9e2 : SUCCESS in 1h 37m 48s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f3ad027a0d5644dcb1e4023f78204659 : SUCCESS in 1h 02m 59s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/51312980a8e14c3ea129fe28294f6f25 : SUCCESS in 1h 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"bd8d04bcd0bb0fe3dea2f238206151e2b4ded832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-01 09:36:23.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent, and commit message as Patch Set 9.","accounts_in_message":[],"_revision_number":10},{"id":"40d9a14ac22b235237f6f9c905d86a1e85381c81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 11:36:42.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55b0601314814b53b4cd6cccb7778da2 : SUCCESS in 1h 33m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94ec7814973d4d70878e77f4edc7629c : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b1632b88db443ffaa7e8f5f21de63c9 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3a6e49bb54c46ae81dea97adefbe054 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d227d32ad71645d9b3a83574b81dea12 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db1b0c50c8e94224a65336c9d29c3370 : FAILURE in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e093c133640e463b8f04b647ed567211 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d42d4be328b4c1dbd23c747ea27b9f6 : SUCCESS in 4m 21s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/ab64b11ea9f744e7b826fff1b0cf11e9 : SUCCESS in 5m 34s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/191bee5efca8462791e8b215bab619a7 : FAILURE in 7m 00s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/9bbeec5ee02d46ffa38e58771e7764d1 : SUCCESS in 5m 57s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/0376267cc977479b9280331f1e16c1e4 : FAILURE in 6m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/dff756e765cf4b6a84f7719a70b98f9d : SUCCESS in 1h 31m 11s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1657bbacb03b408091969942a9af9c10 : SUCCESS in 1h 22m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/da53e4347b3d49f2986a174cbd919cb1 : SUCCESS in 1h 55m 34s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fbeb818154c54b578df2ddeefd4fdbf4 : SUCCESS in 1h 02m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/482cb2ba5ba6487fa405cab7de8bff3e : SUCCESS in 1h 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"74c69432820f14d36f7c5929092d266133564ee2","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-02-01 14:23:43.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"193745595e33c452cad4eb957abab68cd6b5ae88","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-02-01 16:08:28.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"a79eb1260880a044b0d9a81c8fe150cddf2b9851","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-02 12:04:38.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ea117be19f5b173e2c3154293d017b40db10ca70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-02 15:29:47.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d83031074d3f53096204083b135da25fcb25f601","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-02 15:31:38.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"a184c2a56ef513f69d5863d689937e6fee32052b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-02 15:34:16.000000000","message":"Patch Set 11:\n\nThanks Daniel for your comments","accounts_in_message":[],"_revision_number":11},{"id":"7a551c3d2822d767990b1d9400787975b535d263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 17:19:07.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab12b9f100e1451d92ca6798e219de5a : SUCCESS in 1h 28m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f675b6fac15549579711d5153b879b95 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a8b5a62d8b41ceb48e5fe5109e67ef : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db9e8f22146d460db8951c30904afa80 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e7f5d28fdcf46c1a5cfd6fd83d4c398 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/442fedee6b614372af0ca8d25bc8653d : FAILURE in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa9a7032dc44176908422b7a4efc86f : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2648b2652144ecd96b0df8f0d2677ac : SUCCESS in 3m 39s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/5dd9b8873a11438385b511a82c3087b0 : SUCCESS in 5m 07s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/32096f176deb4b85937766dafafb36bd : FAILURE in 6m 21s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/99923d977e8e4b0fb3af755faa4de6a7 : SUCCESS in 5m 03s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/7a51e8e90c8f4a519479cc577713e0c9 : FAILURE in 5m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7acb3852fc344a50b02f23705c94ccc7 : SUCCESS in 1h 42m 33s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ad20574b679b4986bf016b542810d45f : SUCCESS in 1h 24m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8d6fc8050e7b4ad7947ac13725ca3521 : SUCCESS in 1h 30m 33s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ac54deb2e426441b84318728a2386678 : SUCCESS in 57m 55s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/99980febc1864b28abf4d8c76389ed3c : SUCCESS in 1h 10m 13s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"05b8a5eafd4e514833aff5021e04820d122f2ca3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 11:26:01.000000000","message":"Topic flag removed","accounts_in_message":[],"_revision_number":11},{"id":"fc53956cc2618cfd89b6b13a1ba3ad7b54650f16","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 11:26:07.000000000","message":"Topic set to bug/1905965","accounts_in_message":[],"_revision_number":11},{"id":"f8dc9c25ae3e35554934c93c4500f4b521c39f9c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 11:44:56.000000000","message":"Patch Set 11: Code-Review-2\n\n(10 comments)\n\nThis is really a -1 but I\u0027m using -2 because I don\u0027t understand why we\u0027re removing that config option so soon after we deprecated it. If I\u0027m missing something, please add the context to the commit message","accounts_in_message":[],"_revision_number":11},{"id":"d51254af8e1d74fd6bd5d59967e8cd424b5e6551","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-03 14:31:45.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1b3e22e5fabedd1f975e3e14a2ab462142e142b8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 14:43:06.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)\n\nI still don\u0027t grasp why we must remove this option immediately. Comments inline","accounts_in_message":[],"_revision_number":12},{"id":"ea5867197620960bbdf628b920a621211834dcd5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-03 15:04:48.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"04e325cb40cf16e220e623eb2b429eb3cc66299d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-03 15:09:32.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"fd36a26f3f2f1395a7d64d9d373332dfe9d9d4e5","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-03 15:14:38.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"c0943202725f00a008c3613a49848a2874f68c47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-03 15:22:35.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"759d6454534599609aeafb9efc12a0f521427884","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-03 15:23:43.000000000","message":"Patch Set 14:\n\n@stephenfin: Let me know if PS14 better fit your expectations.","accounts_in_message":[],"_revision_number":14},{"id":"f0b2610c0f80a32a7357c5706daee919bfeaad1f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 15:44:07.000000000","message":"Patch Set 14: Code-Review-1\n\n(2 comments)\n\nOkay, I think we\u0027re almost there 😊 Let me know what you think","accounts_in_message":[],"_revision_number":14},{"id":"55b4cc25d3f6d90a86fa859a992369d66f28013c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-03 16:23:18.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"2d68f430d51432c4091af175e78a66da3f7cb9bb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-03 17:02:26.000000000","message":"Patch Set 15:\n\n@stephenfin: finally I decided to leave the option with a doc/description updates.","accounts_in_message":[],"_revision_number":15},{"id":"1b5f3bcd8c5612e4080fe5ff181cfba655450abd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 18:16:36.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90aa223cd2f64e4c9735338c57be7571 : FAILURE in 1h 20m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beaebad20e4f47db8f2956c730f915c8 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/668ad706ac514ebd8b03328386b2dcb3 : FAILURE in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6b1d6d641d3427eadb87a1ac52e8681 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e3cafcdf35e46b3b294625254f208b5 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/634f7f260af74d5eacd3b45813a65052 : FAILURE in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ddea4be85346bda43a124e051911ec : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fff2334e4e6c47aa98ba3002b0611c8a : SUCCESS in 3m 54s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/2082eee883324bc59fc3da31765b9776 : FAILURE in 3m 24s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/15cd2d82b6fa4acca229aad39c2d48e0 : FAILURE in 4m 02s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/b11d93cbd17947d08c1dabddf7448e83 : FAILURE in 3m 52s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/b213c759d2c4439ab0ab7a42e2ed03db : FAILURE in 4m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/95106544afcd422dbc00a34ce1ab000f : SUCCESS in 1h 38m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/67f3f4ea29914891b4114fe0d8cd351e : SUCCESS in 1h 24m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a78e517a17254866942a4a6a83433564 : SUCCESS in 1h 25m 09s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b2106f0a5a094b8c9f1a570bc4a2558a : SUCCESS in 1h 08m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b912eae7879b4cfb9f6a58c3997e3904 : SUCCESS in 58m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"52c944360c3302ea0f338972e8cf578f28185f7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-04 08:56:24.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"51a580f5d5f2f87ddbe0e2a7e598f93cecc5d299","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-04 09:40:36.000000000","message":"Patch Set 16: Code-Review-1\n\n(2 comments)\n\nI\u0027ll fix my nits","accounts_in_message":[],"_revision_number":16},{"id":"b44fb3705a5b936e329d6b65c184f0163f39562c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-04 09:47:15.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"26388dc7406c99c47e2599667c6b9e8c51c2472a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-04 09:48:09.000000000","message":"Patch Set 17: Code-Review+2\n\nFixed my own nits. Thanks!","accounts_in_message":[],"_revision_number":17},{"id":"875bc0fcc76b2d64c17305d1fdca56b8c7a8ec41","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-04 09:49:14.000000000","message":"Patch Set 17:\n\nThanks Stephen!","accounts_in_message":[],"_revision_number":17},{"id":"1bcb3a985bc6fc7ec5e0cef814631956de0061ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 11:34:33.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c733e44dcb66429a8ff02608f0bd8337 : SUCCESS in 1h 34m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67ecab3c31cc481f84c39b3e18dd1f2a : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e85b1391c7b43fab468aa7adb35469a : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91fc4ed0be394766b74068397d297b07 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0447b25ea1544d06837048e5cbad04e4 : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97c635bec85140c9a61e42a9e78f31f5 : FAILURE in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada68a7e910f4606a4de928033db5a27 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eae2121df8974ee286c49f0ac8878a0b : SUCCESS in 3m 27s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f21390279b5741acaf3c1eec7777d2d3 : SUCCESS in 5m 04s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/371682ec46894019be8c4d8d987152df : FAILURE in 7m 04s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/a5125837c013410ab0afe7a37dfb1086 : FAILURE in 5m 10s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/3186fb146ec546ae82cc0fbf0d50c94d : FAILURE in 5m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/dae418813bcf458f94835af7602dc619 : SUCCESS in 1h 26m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/39df0407520541e88fbe8bc4ca848fb5 : SUCCESS in 1h 20m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/79a6e0123b0440cfb532e08701c8fabb : SUCCESS in 1h 40m 08s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/79d37ffefd9647dcb4bf7cc02b1646c6 : SUCCESS in 47m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e19c7f9380874907afc0ae3769573ba7 : SUCCESS in 1h 14m 27s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"eed7ec283a39f4b63aa65fc4f9ca536b61e78544","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-09 13:23:43.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"6243c19e43a47fc1498c3afdeb22aaa9a4770a5a","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-02-09 16:27:59.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"f39d48c0dec01125fe897b94f2364ef123a93032","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-02-09 16:28:52.000000000","message":"Patch Set 17: Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"6d840ef61c299095753f30cc13cab302e98398d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 16:29:08.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"dc0d7bdb7bf6e27c1da083b2074305178cae9238","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 18:01:28.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f8ea03c8d3841758a5f1b46ebdb36e6 : SUCCESS in 1h 25m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41836609e6b846f38ccb126b08f7dabf : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17cc7b009929442fa6591bbf01383195 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bee9b772230f4750973cdd8635d7c81d : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58b2f91e21b444d8823a742a958b93a : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/506109a19dd54015aed39b774a9ea926 : SUCCESS in 4m 08s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/34e46453e9c44edb98fed12f607752fa : SUCCESS in 6m 06s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/93a71bce8e5c463e9af340f608809dec : SUCCESS in 1h 13m 01s","accounts_in_message":[],"_revision_number":17},{"id":"e9cbfc2242b1addd69fabad56c95af4f0903e178","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 18:01:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"f8e7083d040c2a0e0251e227cf3f38f19f47568b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 18:04:14.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120b57d0995747d98ab0b3296331bcbe : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0171e5e0a0d0418a9878a484d483197d : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"4937949dffecdf8863a7876e5a6b0b18e811c3ac","revisions":{"ee758f0b00960a08167d4dcaf591930ab3bb013c":{"kind":"REWORK","_number":1,"created":"2021-01-18 15:12:15.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/1"}}},"commit":{"parents":[{"commit":"46a971bda31444faa8ec2dfffda61200bc1c1eb7","subject":"Disabling the mandatory flag by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46a971bda31444faa8ec2dfffda61200bc1c1eb7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 15:11:53.000000000","tz":60},"subject":"[WIP] Allow to loop for timeout when a MessageUndeliverable is caught","message":"[WIP] Allow to loop for timeout when a MessageUndeliverable is caught\n\nMessageUnderliverable exception have been introduced by recently added\nmandatory flag feature, however clients aren\u0027t prepared to handle\nrecover on their side and so it doesn\u0027t lets a chance to the requester to\nrestablishes itself.\n\nThese changes leave the mandatory flag feature as it was implemented\nhowever they add the possibility to also define `at_least_once` in\nconfiguration to allow to loop for a timeout while a message remains\nunderliverable, as did with AMQPDestinationNotFound exception.\n\nIf the client explicitly set at_least_once\u003dTrue, then when\nMessageUndeliverable is raised, we re-raise it back to the\nclient, with the assumption that it knows how to handle it.\n\nIf the client did not explicitly ask for it (the current case where nova\nand such do not use this API), then we can still use the\nfunctionality (the mandatory flag) within oslo.messaging, but we catch\nit and not re-raise it, the same way AMQPDestinationNotFound works currently.\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee758f0b00960a08167d4dcaf591930ab3bb013c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee758f0b00960a08167d4dcaf591930ab3bb013c"}]},"branch":"refs/heads/master"},"f3b875c35305dc418e08c3cdabfeae90bfd1b115":{"kind":"REWORK","_number":2,"created":"2021-01-21 09:18:42.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/2"}}},"commit":{"parents":[{"commit":"46a971bda31444faa8ec2dfffda61200bc1c1eb7","subject":"Disabling the mandatory flag by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46a971bda31444faa8ec2dfffda61200bc1c1eb7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-21 09:18:10.000000000","tz":60},"subject":"[WIP] Allow to manage missing queue","message":"[WIP] Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f3b875c35305dc418e08c3cdabfeae90bfd1b115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f3b875c35305dc418e08c3cdabfeae90bfd1b115"}]},"branch":"refs/heads/master"},"306a9e53b1137eec0a112dd99fd51a86c51313cd":{"kind":"REWORK","_number":3,"created":"2021-01-22 10:16:02.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/3"}}},"commit":{"parents":[{"commit":"46a971bda31444faa8ec2dfffda61200bc1c1eb7","subject":"Disabling the mandatory flag by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46a971bda31444faa8ec2dfffda61200bc1c1eb7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-22 10:15:45.000000000","tz":60},"subject":"[WIP] Allow to manage missing queue","message":"[WIP] Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/306a9e53b1137eec0a112dd99fd51a86c51313cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/306a9e53b1137eec0a112dd99fd51a86c51313cd"}]},"branch":"refs/heads/master"},"5f840d5984390389e809ac6cb41e96495c5e145e":{"kind":"REWORK","_number":4,"created":"2021-01-22 10:21:12.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/4"}}},"commit":{"parents":[{"commit":"46a971bda31444faa8ec2dfffda61200bc1c1eb7","subject":"Disabling the mandatory flag by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46a971bda31444faa8ec2dfffda61200bc1c1eb7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-22 10:20:59.000000000","tz":60},"subject":"[WIP] Allow to manage missing queue","message":"[WIP] Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5f840d5984390389e809ac6cb41e96495c5e145e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5f840d5984390389e809ac6cb41e96495c5e145e"}]},"branch":"refs/heads/master"},"1f1bbe16543fd36e7b99bfe4fae7bdc4e611262d":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-01-22 10:27:59.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/5"}}},"commit":{"parents":[{"commit":"46a971bda31444faa8ec2dfffda61200bc1c1eb7","subject":"Disabling the mandatory flag by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46a971bda31444faa8ec2dfffda61200bc1c1eb7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-22 10:24:26.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f1bbe16543fd36e7b99bfe4fae7bdc4e611262d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1f1bbe16543fd36e7b99bfe4fae7bdc4e611262d"}]},"branch":"refs/heads/master"},"37c405c5af381b9a6c69645f0652239146855a59":{"kind":"REWORK","_number":6,"created":"2021-01-22 10:53:59.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/6"}}},"commit":{"parents":[{"commit":"46a971bda31444faa8ec2dfffda61200bc1c1eb7","subject":"Disabling the mandatory flag by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/46a971bda31444faa8ec2dfffda61200bc1c1eb7"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-22 10:53:42.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/37c405c5af381b9a6c69645f0652239146855a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/37c405c5af381b9a6c69645f0652239146855a59"}]},"branch":"refs/heads/master"},"e6b2805915aeffbb884619006331d12314c11920":{"kind":"REWORK","_number":7,"created":"2021-01-22 11:07:28.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/7"}}},"commit":{"parents":[{"commit":"af47f18e82aba43b256074a7d49d482b5e295b80","subject":"Deprecate the mandatory flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/af47f18e82aba43b256074a7d49d482b5e295b80"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-22 11:07:09.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e6b2805915aeffbb884619006331d12314c11920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e6b2805915aeffbb884619006331d12314c11920"}]},"branch":"refs/heads/master"},"c1cf0ff4050a883a0f9bfa9b925d6890c4ad8c33":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-01-22 13:40:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/8"}}},"commit":{"parents":[{"commit":"2a3b3f8975526ecd6b141b9367f3ee7fc29db90b","subject":"Deprecate the mandatory flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2a3b3f8975526ecd6b141b9367f3ee7fc29db90b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-22 13:40:07.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1cf0ff4050a883a0f9bfa9b925d6890c4ad8c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1cf0ff4050a883a0f9bfa9b925d6890c4ad8c33"}]},"branch":"refs/heads/master"},"7e5f7dc5b5eb4d85398bf2172c95295d08a50eb7":{"kind":"REWORK","_number":9,"created":"2021-01-25 20:04:38.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/9"}}},"commit":{"parents":[{"commit":"5d3387181e0f11d205ffec9b39046be12ccebc18","subject":"Deprecate the mandatory flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5d3387181e0f11d205ffec9b39046be12ccebc18"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-25 20:03:53.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7e5f7dc5b5eb4d85398bf2172c95295d08a50eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7e5f7dc5b5eb4d85398bf2172c95295d08a50eb7"}]},"branch":"refs/heads/master"},"55fc4ec21eb06e82640056fde6640df4281f16bc":{"kind":"NO_CHANGE","_number":10,"created":"2021-02-01 09:36:23.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/10"}}},"commit":{"parents":[{"commit":"2b89d97888c9b6cd8b4f23e5ae7ce3103d5f9e5b","subject":"Deprecate the mandatory flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2b89d97888c9b6cd8b4f23e5ae7ce3103d5f9e5b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-01 09:32:26.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/55fc4ec21eb06e82640056fde6640df4281f16bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/55fc4ec21eb06e82640056fde6640df4281f16bc"}]},"branch":"refs/heads/master"},"61470013d64a287d4ba9382786758282195c8f77":{"kind":"REWORK","_number":11,"created":"2021-02-02 15:29:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/11"}}},"commit":{"parents":[{"commit":"e18553f5058ef6fd1200ac76e1f8e548afb1c043","subject":"Merge \"Deprecate the mandatory flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e18553f5058ef6fd1200ac76e1f8e548afb1c043"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-02 15:29:35.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/61470013d64a287d4ba9382786758282195c8f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/61470013d64a287d4ba9382786758282195c8f77"}]},"branch":"refs/heads/master"},"57e6ba879ec1a19f142c50134367b4d512f223e7":{"kind":"REWORK","_number":12,"created":"2021-02-03 14:31:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/12"}}},"commit":{"parents":[{"commit":"e18553f5058ef6fd1200ac76e1f8e548afb1c043","subject":"Merge \"Deprecate the mandatory flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e18553f5058ef6fd1200ac76e1f8e548afb1c043"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-03 14:31:09.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\nThe mandatory flag is deprecated since oslo.messaging 12.7.0\n(the last version of oslo.messaging c.f the depends-on below).\nThese changes want to drop the mandatory flag feature ASAP to avoid\nadoption as services aren\u0027t not really ready to handle MessageUndeliverable\nexception on their own. Also we want to abandon it quickly to avoid\nbacks and forths with related config and avoid to mislead users.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nDepends-On: ihttps://review.opendev.org/c/openstack/releases/+/773895\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/57e6ba879ec1a19f142c50134367b4d512f223e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/57e6ba879ec1a19f142c50134367b4d512f223e7"}]},"branch":"refs/heads/master"},"81edc2267a305f04d6ec21dd21def6f8a1a6f7f7":{"kind":"REWORK","_number":13,"created":"2021-02-03 15:09:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/13"}}},"commit":{"parents":[{"commit":"e18553f5058ef6fd1200ac76e1f8e548afb1c043","subject":"Merge \"Deprecate the mandatory flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e18553f5058ef6fd1200ac76e1f8e548afb1c043"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-03 15:09:17.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\nAlso these changes drop the deactivation mechanism\n(direct_mandatory_flag) as it doesn\u0027t really make sense to ignore\nmissing queues.\n\nThe mandatory flag is deprecated since oslo.messaging 12.7.0\n(the last version of oslo.messaging c.f the depends-on below).\nThese changes want to drop the mandatory flag feature ASAP to avoid\nadoption as services aren\u0027t not really ready to handle MessageUndeliverable\nexception on their own. Also we want to abandon it quickly to avoid\nbacks and forths with related config and avoid to mislead users.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nDepends-On: ihttps://review.opendev.org/c/openstack/releases/+/773895\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/81edc2267a305f04d6ec21dd21def6f8a1a6f7f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/81edc2267a305f04d6ec21dd21def6f8a1a6f7f7"}]},"branch":"refs/heads/master"},"96315f51b2167902a7f7413fab939abb7c38f573":{"kind":"REWORK","_number":14,"created":"2021-02-03 15:22:35.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/14"}}},"commit":{"parents":[{"commit":"e18553f5058ef6fd1200ac76e1f8e548afb1c043","subject":"Merge \"Deprecate the mandatory flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e18553f5058ef6fd1200ac76e1f8e548afb1c043"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-03 15:22:20.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nThe MessageUndeliverable would normally by raised only at a more low\nlevel than a 404 error, if a queue is missing we will catch a\nMessageUndeliverable exception, and if an exchange is missing then we\nwill catch a 404 errors, so I think they can live together.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nDepends-On: ihttps://review.opendev.org/c/openstack/releases/+/773895\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/96315f51b2167902a7f7413fab939abb7c38f573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/96315f51b2167902a7f7413fab939abb7c38f573"}]},"branch":"refs/heads/master"},"17b36bc1ceb1b4fa43caf36873aa54dc13dbc703":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2021-02-03 16:23:18.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/15"}}},"commit":{"parents":[{"commit":"e18553f5058ef6fd1200ac76e1f8e548afb1c043","subject":"Merge \"Deprecate the mandatory flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e18553f5058ef6fd1200ac76e1f8e548afb1c043"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-03 16:18:34.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nCurrently, setting the \u0027[oslo_messaging] direct_mandatory_flag\u0027 config\noption to \u0027True\u0027 (the default) will result in a \u0027MessageUndeliverable\u0027\nexception being raised when sending a reply if a RabbitMQ queue is missing.\nIt was the responsibility of the application to handle this exception,\nhowever, many applications are not doing so. This has resulted in a\nnumber of bug reports.\n\nStart handling this error condition, using a retry loop to attempt to resend\nthe message and work around any temporary glitches. Since attempting to send\na reply will will no longer raise an exception, there is little benefit in\nretaining the \u0027[oslo_messaging] direct_mandatory_flag\u0027 config option, since\nusers setting this to False will not benefit from the retry logic and\nimproved logging added here. As a result, This option is already deprecated\nand will be fully removed in a future release.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nDepends-On: ihttps://review.opendev.org/c/openstack/releases/+/773895\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/17b36bc1ceb1b4fa43caf36873aa54dc13dbc703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/17b36bc1ceb1b4fa43caf36873aa54dc13dbc703"}]},"branch":"refs/heads/master"},"d6c1405cab5dc973232cf177b8108543dac62355":{"kind":"REWORK","_number":16,"created":"2021-02-04 08:56:24.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/32/771232/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/16"}}},"commit":{"parents":[{"commit":"e18553f5058ef6fd1200ac76e1f8e548afb1c043","subject":"Merge \"Deprecate the mandatory flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e18553f5058ef6fd1200ac76e1f8e548afb1c043"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-04 08:56:11.000000000","tz":60},"subject":"Allow to manage missing queue","message":"Allow to manage missing queue\n\nMessageUndeliverable inform us that a queue is missing so we can handle\nthat to provide more fine grained retries.\n\nUntil now our retry strategy was based on 404 errors that could\nreflect a missing exchange or a missing queue [1]. By using the\nmandatory flag/MessageUndeliverable we can isolate queue issue and keep\nAMQPDestinationNotFound/404 errors to handle missing exchange case.\n\nCurrently, setting the \u0027[oslo_messaging] direct_mandatory_flag\u0027 config\noption to \u0027True\u0027 (the default) will result in a \u0027MessageUndeliverable\u0027\nexception being raised when sending a reply if a RabbitMQ queue is missing.\nIt was the responsibility of the application to handle this exception,\nhowever, many applications are not doing so. This has resulted in a\nnumber of bug reports.\n\nStart handling this error condition, using a retry loop to attempt to resend\nthe message and work around any temporary glitches. Since attempting to send\na reply will will no longer raise an exception, there is little benefit in\nretaining the \u0027[oslo_messaging] direct_mandatory_flag\u0027 config option, since\nusers setting this to False will not benefit from the retry logic and\nimproved logging added here. As a result, This option is already deprecated\nand will be fully removed in a future release.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nDepends-On: ihttps://review.opendev.org/c/openstack/releases/+/773895\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d6c1405cab5dc973232cf177b8108543dac62355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d6c1405cab5dc973232cf177b8108543dac62355"}]},"branch":"refs/heads/master"},"4937949dffecdf8863a7876e5a6b0b18e811c3ac":{"kind":"REWORK","_number":17,"created":"2021-02-04 09:47:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/771232/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/32/771232/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/32/771232/17"}}},"commit":{"parents":[{"commit":"e18553f5058ef6fd1200ac76e1f8e548afb1c043","subject":"Merge \"Deprecate the mandatory flag\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e18553f5058ef6fd1200ac76e1f8e548afb1c043"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-04 09:47:08.000000000","tz":0},"subject":"Correctly handle missing RabbitMQ queues","message":"Correctly handle missing RabbitMQ queues\n\nCurrently, setting the \u0027[oslo_messaging] direct_mandatory_flag\u0027 config\noption to \u0027True\u0027 (the default) will result in a \u0027MessageUndeliverable\u0027\nexception being raised when sending a reply if a RabbitMQ queue is\nmissing [1]. It was the responsibility of the application to handle\nthis exception, however, many applications are not doing so. This has\nresulted in a number of bug reports.\n\nStart handling this error condition, using a retry loop to attempt to\nresend the message and work around any temporary glitches. Since\nattempting to send a reply will will no longer raise an exception,\nthere is little benefit in retaining the \u0027[oslo_messaging]\ndirect_mandatory_flag\u0027 config option: users setting this to False will\nsimply not benefit from the retry logic and improved logging added\nhere. This option is already deprecated though and will be fully\nremoved in a future release.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nCloses-Bug: #1905965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4937949dffecdf8863a7876e5a6b0b18e811c3ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4937949dffecdf8863a7876e5a6b0b18e811c3ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
