)]}'
{"id":"openstack%2Foslo.messaging~771240","triplet_id":"openstack%2Foslo.messaging~stable%2Fvictoria~I01a38754a31c891f2b3b9c7f8135690693df5d13","project":"openstack/oslo.messaging","branch":"stable/victoria","topic":"drop-LC-victoria","hashtags":[],"change_id":"I01a38754a31c891f2b3b9c7f8135690693df5d13","subject":"Fix type of direct_mandatory_flag opt","status":"MERGED","created":"2021-01-18 15:57:24.000000000","updated":"2021-02-17 17:31:38.000000000","submitted":"2021-02-17 17:29:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771240-drop-LC-victoria","meta_rev_id":"07140cd4773ad4072a0f97d26169a08b93f96bb6","_number":771240,"virtual_id_number":771240,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-17 17:29:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-02-17 17:31:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2021-02-17 17:31:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2021-02-17 13:33:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 15:57:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2021-01-18 16:11:40.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2021-01-18 18:35:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-17 13:33:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"7de70a3c4c20d09c44575ba2bd52c97698419447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-18 15:57:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"446478a72c94d9a10b9b56e5717ca40ba14f4396","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-01-18 16:11:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ae645cedb1530b7a1d4fc61b5fb3e1e733821da7","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-01-18 17:00:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nNote that backports proposed by cores only need one +2 to merge.","accounts_in_message":[],"_revision_number":1},{"id":"f63e0d72d236d2d9e05f59ea924caa26cb57c82a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:35:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1dbde1a831704933a117cb8a260db4a0 : SUCCESS in 1h 11m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37cc023e3d854fe2a2c052483dad3da6 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31f63acc7f0c40279a8df12bd9ac885b : FAILURE in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a04471dfe254b6e848e7f9db5c9a72b : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d34cb6e43764573ba2ce271cf1eaee5 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51ae381dbe2648828e434a886978b134 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792056e9e52b40f38821c3367d73cda9 : SUCCESS in 5m 50s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/c80434849edc4b6f90373b0d156fb0b7 : SUCCESS in 6m 57s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/bb020bada493480aac712542d1bc999a : FAILURE in 12m 34s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/70e48b57dca54fda984df398cc14ae94 : SUCCESS in 11m 21s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/aa092831633849f48aa9ce46d7476933 : FAILURE in 11m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/88212897f887405182f651d757275e2a : SUCCESS in 1h 33m 25s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a3c675a5f43545e48904b0bde31da723 : SUCCESS in 1h 39m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/72e22f19ad024a1cbaeeb344dce98391 : SUCCESS in 1h 50m 59s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/61e14dd7f9ed4139ace88fffbe8e8936 : SUCCESS in 1h 19m 50s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c5780c1a5304126bd3a187a9317725d : SUCCESS in 1h 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0673e4bbcf54286ae38e91ef4a2a24c38b7ada51","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-19 08:48:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fe99d45de79ba102b1aecf692f61d05325bbf041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 10:38:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ddfb9fe829e149e5bb31a293372acdfe : SUCCESS in 1h 25m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2c00eba10af4ce5ad43c4c5b38fcf0e : SUCCESS in 6m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78f0745e18f04918a2ece029050710e8 : FAILURE in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a865555fdb274e139799ef43f2349be6 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fce210d3f74a4cf2b06bb17b79aba433 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/634c34ded90349a891e060702c0e2918 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2de05c46b53d4e33af865bad06428ca4 : SUCCESS in 5m 00s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/aa7eec7cc0b545a4bd26a6b5f5e79a46 : SUCCESS in 5m 02s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/0bcc811da85442de84a147cb9d1a6ed1 : FAILURE in 5m 59s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/81a627b631044013bd6782e6d44c2a7d : SUCCESS in 5m 02s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/12464c0220c346ed864ac08983f7ab31 : FAILURE in 6m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/663f337569bf403cbb94600ff3cb764d : SUCCESS in 1h 27m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a683cb4a83434616a309ba038c8428a8 : SUCCESS in 1h 30m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ddb8f60d53914f539196ccc047104cdf : SUCCESS in 1h 31m 50s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fb1636b42af54e2d9d455543ddd5cd5d : SUCCESS in 1h 00m 53s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fd2a96ed7564a63ba1772b878d5178b : SUCCESS in 1h 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91c840670e72114a057121702d2f1abb1f1d552d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-19 12:29:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e933514a2edcf443ccc93f8ea9dea1c538ca3fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 14:12:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/476e824725434d6982cd69580b98ed00 : SUCCESS in 1h 30m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76e48c71da0546a1bdbd17c263f3cc05 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b492f12860a4730af5e2f891f96c5e8 : FAILURE in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8bfde6a54794671870da168f263b32e : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77888e3571ba4ec0bf1daf981f75b4ac : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fd83a23908e4f2c95dad94d95cf490e : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9e7d63b3879403aa34069ac864d32f4 : SUCCESS in 3m 59s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/d9500e121804446295f98278ce1c07b8 : SUCCESS in 5m 39s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/1721d58e0af54656b597e3e69dc1e0f0 : FAILURE in 6m 05s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/ea4c0019f2d5430f90e478a27fc9dedd : SUCCESS in 5m 41s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/5c1467083aa74acbb0fc7206be1a33ae : FAILURE in 5m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3ad64ce7c345420185f41e75a33c3da2 : SUCCESS in 1h 34m 08s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/78aa8b7ddc094d6ba37ab7bc36f42ee1 : SUCCESS in 1h 31m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/cef45291c3c74a1d9c781b6efaac48f9 : SUCCESS in 1h 27m 49s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6f3e19985f994a308e98360bd6c92555 : SUCCESS in 55m 18s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3423b7750a94718b04e104445fee082 : SUCCESS in 1h 06m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44ecf1f0f3bbce490cf7b20327caf485fade6e9e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-20 10:47:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"259a77c8a17a447f054386dcc48ea11142abe088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 12:33:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8c321db11654528982c5381a0d18340 : SUCCESS in 1h 01m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62a0cc27d4514fb69cae72b3fdb62548 : SUCCESS in 5m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/deece427a52e4716b71ff58658fa954b : FAILURE in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95c4d1528b9e47eebdddb64b66ccbfab : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e476f0b47bf049e7aa89acd3cef659d5 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d799e0346ab744e5a92d54fd9bf38784 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6245fd9880b47da827e4ae105bc11fb : SUCCESS in 3m 46s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/821be6edb5c2464488240d52d14aec97 : SUCCESS in 6m 23s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/fdc0210dcffd453ba6234efe9a78fb6f : FAILURE in 7m 11s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/4e9fbef4ed434c91a476535bee054f3b : SUCCESS in 5m 45s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/ccf81aedb28240b3bc1f31c4b452a5b3 : FAILURE in 7m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a7872d0526874367b9c6aa836982f5f7 : SUCCESS in 1h 21m 52s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6b67c5064616401ab4c0af1e587950a2 : SUCCESS in 1h 36m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a4407a0cc79044c3996d6bc2f5d7c66f : SUCCESS in 1h 30m 12s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/476f095b54844ccab11c34594a6fdadc : SUCCESS in 1h 03m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8e87ce6225f44b8ab30749233d4912e : SUCCESS in 1h 09m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47b9f023bf3a6dc3949088de8305f082487a82aa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-21 09:19:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f3ccca1c8577d3e633edcd0502bde9c8bf3ee1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:38:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a12a3b5b4f794d0db8e6f40d56dcab46 : SUCCESS in 1h 27m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/890a7e8accda4e3f8e63bf602c1ac7d4 : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e7116986ec04a008e66d4c3d8c33103 : FAILURE in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f979492b4e4e0bb0e4fa913773a014 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c23213d1fd24f72a766bb3e3f0f4de7 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/801ed61a2b524f1d92188378c4556875 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3d021c537ad4deea9c9166454fe1865 : SUCCESS in 4m 18s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/ccc23d3b7ee0403d8cecca0c8a50a9a0 : SUCCESS in 5m 12s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/0e67cabd2da24cedb98a4a674eee307a : FAILURE in 6m 19s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/1bde749a789847d6a99349e22c6f0c63 : SUCCESS in 6m 57s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/da3660e6b0694fbabe0ba689f1c6307e : FAILURE in 7m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0f7b5cf0180c47de8793d4c13f81ffad : SUCCESS in 1h 46m 54s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f75a488aabdf467a836fc41079a592f5 : FAILURE in 2h 12m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f28cccc5ad994dccb8f54c6ef11e6b92 : SUCCESS in 1h 42m 59s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/1d4979bd04574d5b865fdd059d256048 : SUCCESS in 1h 03m 07s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8830ad21d9604328b721ffbb6c8f962a : SUCCESS in 1h 07m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e96097c827045aecc14983493d39f591e5baacc7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-21 14:32:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"43ac2c7c235a22cb5fe8d2897013fbe5790678a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 16:25:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2abed35485ca46fcad4e6e1eecbe2d42 : SUCCESS in 1h 18m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12d2be4e1f6543ffbf4066d202f5efaf : SUCCESS in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4ec4454b6344e4e92089f9131b5aad1 : FAILURE in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c2006891da41eb8ab9b5f568ca88dc : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57e16b4c232e4f1baf14ec4163dc1f5c : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce78ff7da9b9453887c75e96e694f58d : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1432ce62b65497b82b75f2edf8d6c1a : SUCCESS in 4m 09s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/bbdcee514c04421a94f4aa9e38dbc049 : SUCCESS in 6m 07s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/af2aea893c00414199ca500f15dd0b5d : FAILURE in 6m 37s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/799b7e9a173346c9b42ce6e626737c0b : SUCCESS in 5m 28s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/848a79f39ad74bebad8ee950d3b0172a : FAILURE in 7m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/730a9715c1ad4d63a11a13ad7cf29be8 : SUCCESS in 1h 40m 35s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f26a41f947604876859f81131ad2fe0f : SUCCESS in 1h 27m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8045efbe42ed4cbd90fa33495fe8aed7 : SUCCESS in 1h 29m 58s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/cb3c0c1edd9b4ba3babff40fc2830263 : SUCCESS in 1h 08m 55s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/219a9f96f1684961b90cedc8b36921fd : SUCCESS in 1h 08m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4c37bfc04899e72d8de43266f25c34578e09580","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-22 09:03:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1dd08c273182dbd8647ce5ba0024d3b0d7110057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 11:23:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5963c5d98e6443e489a72a0ffe4bb3c9 : SUCCESS in 1h 15m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52d67743c4bd4beb87586876517675f8 : SUCCESS in 5m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e68a33afd9348c29954e00456175660 : FAILURE in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a537c8c05214880ad46bcb3f85ce476 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a417b433e3d45e58986979f2a88adc5 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b13aca73720447abcac1eae99a25570 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b504eabaf6a6484b9eef5581142efa59 : SUCCESS in 4m 40s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/fe7a850c074a48998c409fd3d291a95b : SUCCESS in 5m 47s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/4d9c0ec8619a414dba75856470dad06f : FAILURE in 7m 07s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/b141ab8869d8417e8342d5a330ecfd01 : SUCCESS in 5m 59s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/92592248e5a94b3bbeafa0a62f6bae76 : FAILURE in 7m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/11451326f3be4529a330dd47371cf749 : SUCCESS in 1h 29m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b573f3516e39430f9fb1bc624d1531b5 : RETRY_LIMIT in 15m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6405453c0b124211a69a3e6fedb0dff6 : SUCCESS in 1h 28m 35s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/76555115ab1c406393dc6fc4477e569c : SUCCESS in 59m 44s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/5595a61f7fa44065b827bfe751b122bb : SUCCESS in 1h 07m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"034098b79db8a3d58442daf4e5b9b5342a4abc58","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 10:43:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"59755a5efcbc57fc839f5adc4cce71c5a51876ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 13:37:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41c994a8b6aa4d3cbff7bbadd97dbbd2 : SUCCESS in 1h 16m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25e70601a65d48dea787f3f3c5bb6fac : SUCCESS in 8m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c39e20f18874b348f87bce9c08d617a : FAILURE in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a87b68eeb734024b6abe7c798bc9a66 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a695aaa7b2214f74a8223692d54dfdf7 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b073148b4214a94a26cdd47d7b6d28b : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbada4885e144a8a33cae2b40f7e013 : SUCCESS in 4m 21s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/f55da10b9aaf4cc7b2568025e57c8502 : SUCCESS in 5m 31s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/3f13ed2d68b54285b04a6b14db4e3d57 : FAILURE in 6m 39s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/537460df7ed64300a07555a2017d65cc : SUCCESS in 5m 21s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/6b9b700220ab4ef9bf12bc47918558fe : FAILURE in 7m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/468d89feb79443bab50d4ca90209faf2 : FAILURE in 2h 47m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7148ec3f14834176a00b62e15b6b8d98 : SUCCESS in 1h 23m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1bfb69ed3c304a4a860d85248e83ab41 : SUCCESS in 1h 32m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/851ee22eb59d45998a6b98e11371817c : SUCCESS in 1h 07m 47s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/17b7ef6575da4eaeb43120db56a96949 : SUCCESS in 1h 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b108509b07ded84fa320b23b4c29522c07dc0ca3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 19:46:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"078a661b3a650e201427f486cbb1ca54e7eca3f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 21:29:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4a3deeb78c54ef4a09bb7ec64e4e949 : SUCCESS in 1h 24m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27224e5826bd4aa089bd71403bc73d09 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/daad984c102449958e92335fa7996861 : FAILURE in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568844a8679a4afc9aa76692fbe4b55c : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae50ce19c9e14ee6b43c2c7a745df10e : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ebad680af6644efa99d9b36e71ee715 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee02f10263ca444084fbff1b8fda267e : SUCCESS in 4m 24s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/05a54d95a7ee4ae180fa69b863bfd214 : SUCCESS in 5m 17s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/88eaeecffac84b99ad5b224a1996e122 : FAILURE in 8m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/ec572619d3834925857d7a1601fe3452 : SUCCESS in 5m 20s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/b12746931dd74462a92d651efb2378fe : FAILURE in 6m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7b80847146fc4e2ea974ce1c6ce09ab9 : SUCCESS in 1h 35m 23s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/be9bd3b04b9a42caa3cae6aca05709b5 : SUCCESS in 1h 21m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b7c8f4b982cb497b9eefa4a2273d4b99 : SUCCESS in 1h 23m 25s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e675099acdfd452d9849e317a40c4f41 : SUCCESS in 1h 05m 53s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/17ee8c976d754e3c91ee2956dcf0dcbf : SUCCESS in 1h 10m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"250f3e827e8a71782beb855fe03085231e17ca41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 09:05:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"9efa4bb6c9db7e4602045f1ad93ce2bd6613b95e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 09:07:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"978956b5bfe9251514525b0de070a40686148236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 11:00:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8eebc11cf1f94479a69cda2d4ad6879a : SUCCESS in 1h 36m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6350846f403c40cba5b7968131859c41 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e76c647272094e66883f69d2ed9f7e78 : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89899126e98b471592301e38e2e5e48d : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16811245179649f7b2de3d2013f06ac2 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21257e3a485744a686320d4250177922 : FAILURE in 3m 55s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/eb280fb029994c71b516c66c473a0db2 : SUCCESS in 5m 46s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/c1baddf6197c4b8db52b0ce19d507c98 : FAILURE in 7m 23s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/e283068ecc1f4b6aa7d2173a4318f70b : SUCCESS in 7m 41s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/0653e5feb3f146cabb8615825e30a907 : FAILURE in 10m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/212be7e89bf24104a9f2009d3c984f44 : SUCCESS in 1h 27m 32s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f408cf0360d54cc2a296ce59c0def990 : SUCCESS in 1h 30m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/331a19d93e604b2b919ee783278cc215 : SUCCESS in 1h 31m 58s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d9b9e917eb68462b894dc189410a6dfd : SUCCESS in 1h 06m 10s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9252221f49a44bba215177903f10489 : SUCCESS in 1h 07m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3f169415a272d36c4151dd5357043fb844e36c2e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-01 09:37:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6c33e261ecbec2c3fb43edd7295534692e60818d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 11:13:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bde779aba4f43fb86c95d17b9552eed : SUCCESS in 1h 19m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bee06dc3bad4ff3ba46979abdcc65a4 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/820fb2af83014e0097a4833c9d54acda : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dce860fbd7d64b679c900cf43bdb6a69 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06727e10c56440f7bff9c8c0752421ec : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6946c153924031b10d4deffc588f01 : SUCCESS in 3m 46s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/9668ac8d323b4d6684027d5596409887 : SUCCESS in 5m 19s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/172dac3ffa9a480bbd8ace274b0d3d6e : FAILURE in 6m 51s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/76aacefd484444ed893d85609ee5a034 : SUCCESS in 5m 48s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/1e07cfc1f94f4469aa43e1ced0182517 : FAILURE in 6m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5a12a859edfd4a81b91883e1262c48e6 : SUCCESS in 1h 31m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/69ea9bd9d5994ee2b8424e73a92ea527 : SUCCESS in 1h 28m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1b23217ed4ad4172bde8607d1bde26b5 : SUCCESS in 1h 30m 48s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/116860a6842642ddba1b92db86590428 : SUCCESS in 54m 34s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b20992de53a4b35bc80d45070f78bcb : SUCCESS in 1h 06m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"95c019e59c0606546095b907e762dc3f4fabd516","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-02-11 21:48:17.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e7d825dc53eb5a3538992fdd14d919c66b6368ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-17 13:31:45.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"fcd9b4752a0c23e867fd8887711e3fd8da989d85","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-17 13:33:06.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"05caee64f032d027f3670d0589fbed7a314255be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 15:32:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/887fc97e7e9b4968bc33d5725a00fb1c : SUCCESS in 1h 15m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbf00e02460e47f0800ae3f7018457d9 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/803a3f65bbb7425e9b21e219121c8162 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/958c3f64d650415aa1dce6d57fbace16 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70b25e634dee494c92eaae7a26ca4447 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84a69d4744c4c91a70b6e964f08f71a : SUCCESS in 4m 20s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/b51cd3d60cbe4d5daa2f5c02efa17a5b : SUCCESS in 5m 06s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/1b9edb50c6484bb58a90f3c27e1c5af0 : FAILURE in 6m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/1aa7fa049cbc480ebcc5e6d6800faa2f : SUCCESS in 5m 30s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/8b9e92d1e5e9421cbadd7749a9173cd3 : FAILURE in 6m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a28e85c2fb1c4c5e8174dc30de2e9dae : SUCCESS in 1h 29m 48s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/28adabe57b16476dbef6d61839e64ec5 : SUCCESS in 1h 39m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ad94f31d68624223a4ab642fac9fcf25 : SUCCESS in 1h 32m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/be2c3bc2dd034293958211fec5a34243 : SUCCESS in 1h 05m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ddd9847361ea488e9b0b458cfc398780 : SUCCESS in 1h 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b96db6537047fcc2d6f8aeb46cb72bbd150a641f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 15:32:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"624ed9f193b08b0275ca86ec51a191c964f99abd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 17:29:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d64bd5c6249148478409228d266c89ce : SUCCESS in 1h 50m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026a596daacc4d9296c26fb9c1b8c890 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef305c5439094be5842f8efa1e0502d1 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1a5935dbc19422faa490d5656523c21 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa1b75f487d647cfb0a45372d40b1d04 : SUCCESS in 5m 44s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/ba42bc044f774f2c8e5f0364594332e1 : SUCCESS in 5m 52s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/056a4d3cb67b44c29f8a463798d69385 : SUCCESS in 1h 37m 48s","accounts_in_message":[],"_revision_number":4},{"id":"b54089a36509aeb974d6e9275d61666cfd01c65a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 17:29:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"82256843c39e8ffdb8aca50458c217bfb09cf4ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 17:31:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d737863bec844756a36640e830634c70 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a0329a3f86e8bf275dd4e16c1f354911a6ff2eb6","revisions":{"49a52f97d6fc732e4fe35a3bc8efd536d9406003":{"kind":"REWORK","_number":1,"created":"2021-01-18 15:57:24.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/771240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/771240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/1"}}},"commit":{"parents":[{"commit":"8f5847dcbe5144272ff7295fce2b50a9f544e5b0","subject":"add min of 1 to rpc_conn_pool_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8f5847dcbe5144272ff7295fce2b50a9f544e5b0"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-12-22 19:34:16.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 15:56:59.000000000","tz":60},"subject":"Fix type of direct_mandatory_flag opt","message":"Fix type of direct_mandatory_flag opt\n\nAn IntOpt with a default of True is invalid. I\u0027m a little surprised\nthis doesn\u0027t fail a defaults check somewhere, but it needs to be\nfixed regardless.\n\nLooking at where it is used, it appears the boolean type is correct.\nThis just changes the opt type to BoolOpt to match.\n\nChange-Id: I01a38754a31c891f2b3b9c7f8135690693df5d13\nCloses-Bug: 1909036\n(cherry picked from commit b8f8b17030e93f6fe64e986ebae1de4011f007d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/49a52f97d6fc732e4fe35a3bc8efd536d9406003"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/49a52f97d6fc732e4fe35a3bc8efd536d9406003"}]},"branch":"refs/heads/stable/victoria"},"c109a6f9430d852c3d4e59329c9ba709309f0a2f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-01-28 09:05:41.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/771240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/771240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/2"}}},"commit":{"parents":[{"commit":"85f05564ad94035d8edccea4038d0f2b43b8c112","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/85f05564ad94035d8edccea4038d0f2b43b8c112"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-12-22 19:34:16.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-28 09:05:23.000000000","tz":60},"subject":"Fix type of direct_mandatory_flag opt","message":"Fix type of direct_mandatory_flag opt\n\nAn IntOpt with a default of True is invalid. I\u0027m a little surprised\nthis doesn\u0027t fail a defaults check somewhere, but it needs to be\nfixed regardless.\n\nLooking at where it is used, it appears the boolean type is correct.\nThis just changes the opt type to BoolOpt to match.\n\nChange-Id: I01a38754a31c891f2b3b9c7f8135690693df5d13\nCloses-Bug: 1909036\n(cherry picked from commit b8f8b17030e93f6fe64e986ebae1de4011f007d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c109a6f9430d852c3d4e59329c9ba709309f0a2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c109a6f9430d852c3d4e59329c9ba709309f0a2f"}]},"branch":"refs/heads/stable/victoria"},"6c79e07d3af1182ca10fdafc623f103618630008":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-28 09:07:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/771240/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/771240/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/3"}}},"commit":{"parents":[{"commit":"7187f1bf056daceab6ca62bac8ca060dd4b86a8d","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7187f1bf056daceab6ca62bac8ca060dd4b86a8d"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-12-22 19:34:16.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-28 09:07:27.000000000","tz":60},"subject":"Fix type of direct_mandatory_flag opt","message":"Fix type of direct_mandatory_flag opt\n\nAn IntOpt with a default of True is invalid. I\u0027m a little surprised\nthis doesn\u0027t fail a defaults check somewhere, but it needs to be\nfixed regardless.\n\nLooking at where it is used, it appears the boolean type is correct.\nThis just changes the opt type to BoolOpt to match.\n\nChange-Id: I01a38754a31c891f2b3b9c7f8135690693df5d13\nCloses-Bug: 1909036\n(cherry picked from commit b8f8b17030e93f6fe64e986ebae1de4011f007d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6c79e07d3af1182ca10fdafc623f103618630008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6c79e07d3af1182ca10fdafc623f103618630008"}]},"branch":"refs/heads/stable/victoria"},"a0329a3f86e8bf275dd4e16c1f354911a6ff2eb6":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-17 13:31:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/771240/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/771240/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/771240/4"}}},"commit":{"parents":[{"commit":"1682ef69ae30732071fb4296f947047c37c4622e","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1682ef69ae30732071fb4296f947047c37c4622e"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-12-22 19:34:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-17 13:31:45.000000000","tz":0},"subject":"Fix type of direct_mandatory_flag opt","message":"Fix type of direct_mandatory_flag opt\n\nAn IntOpt with a default of True is invalid. I\u0027m a little surprised\nthis doesn\u0027t fail a defaults check somewhere, but it needs to be\nfixed regardless.\n\nLooking at where it is used, it appears the boolean type is correct.\nThis just changes the opt type to BoolOpt to match.\n\nChange-Id: I01a38754a31c891f2b3b9c7f8135690693df5d13\nCloses-Bug: 1909036\n(cherry picked from commit b8f8b17030e93f6fe64e986ebae1de4011f007d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0329a3f86e8bf275dd4e16c1f354911a6ff2eb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0329a3f86e8bf275dd4e16c1f354911a6ff2eb6"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
