)]}'
{"id":"openstack%2Foslo.messaging~772655","triplet_id":"openstack%2Foslo.messaging~stable%2Fussuri~Icf48ef72fbfff88dda2714b386dbbfe355dc1acb","project":"openstack/oslo.messaging","branch":"stable/ussuri","topic":"drop-LC-ussuri","hashtags":[],"change_id":"Icf48ef72fbfff88dda2714b386dbbfe355dc1acb","subject":"Dropping lower constraints testing","status":"MERGED","created":"2021-01-27 11:44:26.000000000","updated":"2021-02-17 17:45:27.000000000","submitted":"2021-02-17 17:39:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772655-drop-LC-ussuri","meta_rev_id":"cc5f35aaf5075a302fc0ecda3a28b89dd44fcad5","_number":772655,"virtual_id_number":772655,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-17 17:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-02-17 13:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-02-17 13:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-27 13:24:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 11:25:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-02-05 12:20:31.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"e49bd07e4882db82044c0a55bd8c7fafa67557a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-27 11:44:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b64d399a4ba3fe17976824f8581de84ce280fdc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-27 11:52:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c58aee544bcf58de54ec4a0c962cef03a62645c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 13:24:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/262263b4ce68425c9c820cf63fea5abe : SUCCESS in 3m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9383cb704cb421191da2852f1c1df40 : SUCCESS in 1h 20m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/516bae176e714c14a968755329c13bb9 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/062a8a3bfe024f458b5764ddfce2d461 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5125ef2fec47448ab0d5df2551c7c5ae : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da0b2ad97fa24078a8b327b054d8f621 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0573c44c67e4ff4af10af4cc36ed703 : SUCCESS in 3m 34s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/457027693f8a449089ca0eaa0436cdb0 : SUCCESS in 6m 01s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/330f8a0919834372b0be0a1ace6b384d : SUCCESS in 8m 38s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/79d6f0545a6d41caab670be0ff2e0a6d : SUCCESS in 5m 54s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/11437bf6193347a8b3428fa6a945abfe : SUCCESS in 7m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8e2b4902b5804b449c8ba943c33a7e03 : SUCCESS in 1h 14m 25s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/90567152201a43daa4f379261bdee1b1 : SUCCESS in 1h 20m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3649c88446f44bb7875a2bc4b393d91e : SUCCESS in 1h 28m 17s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/9180b07cc64a406bbec1ea46090c1419 : SUCCESS in 1h 02m 32s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/9a54f04ed29c478b9ab4e630d88661cb : FAILURE in 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1351fde63c46fbb561fd46762e443fac4fb82a18","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-03 11:25:45.000000000","message":"Patch Set 2: Code-Review-1\n\nThis is missing the tox.ini changes. Can you squash them into the victoria backport and then update this?","accounts_in_message":[],"_revision_number":2},{"id":"58009008d28615ec58107c5d513703f58d036e5b","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-05 12:20:31.000000000","message":"Patch Set 2: Code-Review+1\n\nYes right I have made a patch to remove from tox in master. But I think that can be two different commits.","accounts_in_message":[],"_revision_number":2},{"id":"2375e72f68d059f43b124a4a3d8bbd868dc41f23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-16 12:38:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8875012217140227211132468c7a534a94fedd19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 15:01:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a30375f17bf64a7aa31766f6eeb9256b : SUCCESS in 3m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/140e33c017914f52bef59bf0cd4c6cce : SUCCESS in 1h 21m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf3ac839f7cb418ab3959c37076d90d2 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64795c3a226a49f38b4eb74be88e23ab : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c135fb87f0e448b995a5c4d9bc4ac514 : SUCCESS in 6m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f39f67dfea9475d8cab8b68490c8ac1 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4dfbd76be6944f1acc49f82dab21c6a : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/217b227a65094d34aaf0319c2bcf0742 : SUCCESS in 4m 00s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/08a0df55e4264276a04baf1600afea76 : SUCCESS in 6m 04s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/32fd198a35e1474babc36a50308ffcd9 : FAILURE in 6m 07s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/6250e00fd57544c28048b2cb257a50b4 : SUCCESS in 5m 14s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/0f3687e66e094ac2ba5a5f6f9cc87c23 : SUCCESS in 6m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/03ab438c21ce468d9748ab2dd58c1356 : FAILURE in 2h 08m 28s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2dd0143a79024adbb20ac9553ca45824 : SUCCESS in 1h 57m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9884faa69c9042ee841c4cce9deba732 : SUCCESS in 1h 29m 11s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/cd6be109a1234b86a538bf18773fc570 : SUCCESS in 1h 24m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/ee5ae18ed3d74dc6b70749a2f3a93045 : FAILURE in 22m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"30b4f12e9c2df92b12d47667e1af18643dcaae85","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-16 15:03:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d2e09e399c55d1f953ad26dada2b5af8616c4328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 16:45:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7cbf90caf9fa465282b31af8a77d467c : SUCCESS in 3m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0dce06608094725a58cde5765b6b069 : SUCCESS in 1h 16m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c60ec63bec8246f0afea7e438f6cda4f : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0e543ba83bb4b6e8b0b7d83d1a12d06 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30ad1f9e30db4c2caa5e88b5da13f572 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ea3492c3454495fb119dd6068789499 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27bf5cfd66334583ae3bb6a7c75e84c9 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9de40271fbc74e51bff5227044aff438 : SUCCESS in 3m 32s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/f48f42f2634e45fab78869504cb1c437 : SUCCESS in 5m 19s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/13dfb7f48080431eac2a285eb8cf48df : SUCCESS in 7m 04s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/81c1e210d7f74f67a317ade867c4aed3 : SUCCESS in 5m 16s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/1bd6dcea4b764d64853b9addb9070b6d : SUCCESS in 6m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e397821d5b2c4ffa8ee13fe05a83de81 : SUCCESS in 1h 23m 24s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/26b8e82425f142a89b6799dc63dbeeb3 : SUCCESS in 1h 29m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/40114178f0be4af2b7a566f578b2a36c : SUCCESS in 1h 32m 27s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d71e7327fd2d4d8ca3e7bd74d204fd20 : SUCCESS in 1h 19m 55s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/f126089f6215439db06b4c572e051f93 : FAILURE in 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b7090340ac5f732ac7ebe6e56095142eae76e24","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-17 13:29:48.000000000","message":"Patch Set 3: Code-Review+2\n\nThis is missing the Conflicts section. The changes are pretty self-explanatory here so we can let this slide this time, but in general you should include this plus an explanation of why there was a conflict to help reviewers. The end result should look like e.g.\n\n  Conflicts:\n    .zuul.yaml\n    lower-constraints.txt\n\n  NOTE(hberaud): Conflicts are due to different base jobs in zuul and different\n  constraints in lower-constraints.","accounts_in_message":[],"_revision_number":3},{"id":"dbc13dd270eae5bd4f9d2ba2968f88da1c6242aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-17 13:35:35.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0a10aeef50edafa6551375d375dc344782291807","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-17 13:36:07.000000000","message":"Patch Set 4: Code-Review+2\n\nLovely, thanks","accounts_in_message":[],"_revision_number":4},{"id":"3fc0be24178e016a4f8a86402cef6b5f754e3397","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-17 13:39:21.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f23f1be3e7ac8785168595255e600da4673e0bee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 15:46:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/12c1c9b1c02847a7b994022759ad4d33 : SUCCESS in 3m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6c0a8e9c11b4db6a40f9d4c9bed1582 : SUCCESS in 1h 29m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26a6f78de8a34d3e98dd360706107591 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e3a8dec5e941369d6c4b612d776cfb : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e42b659b75240aaa932472a7d8cd212 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c87d7068f60244678a864a8e59c29884 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f861e889644a459c97a13a876ba40db0 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/837f83c2517540879760b1b470d3e18c : SUCCESS in 3m 54s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/b48ff05934be489d81c920e619d4b943 : SUCCESS in 5m 43s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/6df8cea6adaa428998a92986b429d508 : FAILURE in 7m 22s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/28d96eaa1ab6444b8585389d190aac91 : SUCCESS in 5m 56s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/5a7b5146c34e475aa3ac462391850a0a : SUCCESS in 6m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0d0552d5d27346d0aadffb55227514d2 : SUCCESS in 1h 24m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1a9a9f4a63d44e1f9be2ef6d149f590a : SUCCESS in 1h 18m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9eab968eaef549e6a0e6800b27e9fd92 : SUCCESS in 1h 24m 40s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/80a04254543548a08cef5921ba59b687 : SUCCESS in 1h 15m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/bb8bcebbf57b40aa9002d70d7e31d4ad : FAILURE in 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c747b5465f791758b22b84d1cde785dd0506ad0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 15:54:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2b261c7df66c17886311bb2dc26be7d9572044fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 17:39:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/263df4cc7065456aa3eb245703332a79 : SUCCESS in 3m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73f6a8bc11824b5da73f3f4070da5a1f : SUCCESS in 1h 24m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ffbe85bb5d456095b8a7332273109a : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/592d7938a5ba4a42a4e4d9288d458c17 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/965869068a8040259a7cdb038ce98d42 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90d82cd36fe945b0895b88a03b34e231 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed052b8121fa4d859f39369566d96266 : SUCCESS in 4m 56s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/c6dab992c4c844598e8e2f4fe6809976 : SUCCESS in 6m 23s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c88634ea23d647be862adf947c14ade5 : SUCCESS in 1h 37m 25s","accounts_in_message":[],"_revision_number":4},{"id":"6499ba301d40cefb0e50a17d46e1af80ec7020fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 17:39:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"cc5f35aaf5075a302fc0ecda3a28b89dd44fcad5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 17:45:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/514d1e9aae3742b292ee63095ec3750a : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/426d498f33d54d50b5daf8e4711063fc : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7e151b15b949e00e5d6a1b6ba93c4fbbb38274bc","revisions":{"cb30fcb986273c66b9be8423f3b1dad038aa6328":{"kind":"REWORK","_number":1,"created":"2021-01-27 11:44:26.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/55/772655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/55/772655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/1"}}},"commit":{"parents":[{"commit":"7c9b070f5bc07692c837f9915eb0d8525065d0f6","subject":"add min of 1 to rpc_conn_pool_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7c9b070f5bc07692c837f9915eb0d8525065d0f6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-18 14:09:42.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-27 11:44:16.000000000","tz":60},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain needed\nto maintain that now exceeds the benefits provided by this mechanismes.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nChange-Id: Icf48ef72fbfff88dda2714b386dbbfe355dc1acb\n(cherry picked from commit 686a8b61826d9bed0a989e645a13c708c3e6343c)\n(cherry picked from commit 85f05564ad94035d8edccea4038d0f2b43b8c112)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cb30fcb986273c66b9be8423f3b1dad038aa6328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cb30fcb986273c66b9be8423f3b1dad038aa6328"}]},"branch":"refs/heads/stable/ussuri"},"499e53cd0404a15cd65eb819f30194d3e050cda7":{"kind":"REWORK","_number":2,"created":"2021-01-27 11:52:42.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/55/772655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/55/772655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/2"}}},"commit":{"parents":[{"commit":"7c9b070f5bc07692c837f9915eb0d8525065d0f6","subject":"add min of 1 to rpc_conn_pool_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7c9b070f5bc07692c837f9915eb0d8525065d0f6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-18 14:09:42.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-27 11:52:06.000000000","tz":60},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain needed\nto maintain that now exceeds the benefits provided by this mechanismes.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nChange-Id: Icf48ef72fbfff88dda2714b386dbbfe355dc1acb\n(cherry picked from commit 686a8b61826d9bed0a989e645a13c708c3e6343c)\n(cherry picked from commit 7187f1bf056daceab6ca62bac8ca060dd4b86a8d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/499e53cd0404a15cd65eb819f30194d3e050cda7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/499e53cd0404a15cd65eb819f30194d3e050cda7"}]},"branch":"refs/heads/stable/ussuri"},"d10e7caa697f349397c211bf7c711c66d44aefde":{"kind":"REWORK","_number":3,"created":"2021-02-16 12:38:14.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/55/772655/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/55/772655/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/3"}}},"commit":{"parents":[{"commit":"7c9b070f5bc07692c837f9915eb0d8525065d0f6","subject":"add min of 1 to rpc_conn_pool_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7c9b070f5bc07692c837f9915eb0d8525065d0f6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-18 14:09:42.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-16 12:37:30.000000000","tz":60},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain needed\nto maintain that now exceeds the benefits provided by this mechanismes.\n\nAlso removing lower constraints and related tox environment as they are\nno longer needed.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nChange-Id: Icf48ef72fbfff88dda2714b386dbbfe355dc1acb\n(cherry picked from commit 686a8b61826d9bed0a989e645a13c708c3e6343c)\n(cherry picked from commit 1682ef69ae30732071fb4296f947047c37c4622e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d10e7caa697f349397c211bf7c711c66d44aefde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d10e7caa697f349397c211bf7c711c66d44aefde"}]},"branch":"refs/heads/stable/ussuri"},"7e151b15b949e00e5d6a1b6ba93c4fbbb38274bc":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-02-17 13:35:35.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/55/772655/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/55/772655/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/55/772655/4"}}},"commit":{"parents":[{"commit":"7c9b070f5bc07692c837f9915eb0d8525065d0f6","subject":"add min of 1 to rpc_conn_pool_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7c9b070f5bc07692c837f9915eb0d8525065d0f6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-18 14:09:42.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-17 13:34:44.000000000","tz":60},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain needed\nto maintain that now exceeds the benefits provided by this mechanismes.\n\nAlso removing lower constraints and related tox environment as they are\nno longer needed.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nConflicts:\n  .zuul.yaml\n  lower-constraints.txt\n\nNOTE(hberaud): Conflicts are due to different base jobs in zuul and\ndifferent constraints in lower-constraints\n\nChange-Id: Icf48ef72fbfff88dda2714b386dbbfe355dc1acb\n(cherry picked from commit 686a8b61826d9bed0a989e645a13c708c3e6343c)\n(cherry picked from commit 1682ef69ae30732071fb4296f947047c37c4622e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7e151b15b949e00e5d6a1b6ba93c4fbbb38274bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7e151b15b949e00e5d6a1b6ba93c4fbbb38274bc"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
