)]}'
{"id":"openstack%2Foslo.messaging~774231","triplet_id":"openstack%2Foslo.messaging~master~Iefa00bee6f67e9e82f0c703a7aff144c0dcbe1ec","project":"openstack/oslo.messaging","branch":"master","topic":"remove_lower_constaints","hashtags":[],"change_id":"Iefa00bee6f67e9e82f0c703a7aff144c0dcbe1ec","subject":"Remove lower constraints.","status":"MERGED","created":"2021-02-05 12:00:52.000000000","updated":"2021-02-12 11:22:37.000000000","submitted":"2021-02-12 11:21:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774231-remove_lower_constaints","meta_rev_id":"c52a54360044ba238359ab01043a1f5aa1796650","_number":774231,"virtual_id_number":774231,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-12 11:21:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2021-02-05 12:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2021-02-12 09:53:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 12:04:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-05 13:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"29da9b063ddd1a33959ead3729de8972ac922767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-02-05 12:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b0053185c521e18d6f0b8de223209257f6c0e75","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-05 12:04:06.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"48e877f55da4a70ea2b8f333be8d57a3baa79a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 13:46:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/34a1d689c367431bbfb11ee366312343 : SUCCESS in 2m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eda1f2baf5a44f57af69291c72a73d2e : SUCCESS in 1h 21m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e82c5050bfec4c16ab750de3b5a81238 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb7a2235319f467398e53eb5f5b56c07 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22325ee4f41a4873be9a069d4004c196 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/884bbea57a524612b4503a2443b9901e : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca4a5f0564ee47769e281cdb4d6aa227 : FAILURE in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecfab9efc4e1469091f695455743e25b : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ff574f0db0f4f3ea617f17c5986feb0 : SUCCESS in 4m 20s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f928fbb4773f41538195ab3901b95583 : SUCCESS in 5m 35s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/f88f6456ceb64f10aeca3e30fa31cf2a : FAILURE in 6m 47s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/aa7290404ed049679a46242de6f10aaa : FAILURE in 4m 59s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/08c60fe5de9144f29c0060ac9210dd04 : FAILURE in 4m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/23ef88a692e74ae38ceb835705281cbf : SUCCESS in 1h 29m 35s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/57e7c0eb07e94a62af69d8d75a23bec9 : SUCCESS in 1h 33m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/34fc01254ee44711a1a12c357ebae433 : SUCCESS in 1h 37m 17s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6588543a2aee4dae82a35830c4cab88b : SUCCESS in 1h 03m 30s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7acf7205cd34d5294e4b73d612a8731 : SUCCESS in 1h 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12844bda8ce8aa44cd761d19a2299873439659f9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-12 09:53:43.000000000","message":"Patch Set 1: Workflow+1\n\nFast tracking these changes.\n\nDuring our meeting we agreed that these patches required only one core approval.","accounts_in_message":[],"_revision_number":1},{"id":"f5437b6662ab4fed3ecf54d5228ac25892fd2762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 09:53:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ada7bc4a3349f951b54cd319c1cddc00e0b5b2af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 11:21:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0aa0245f7d234d9cbe9503fc56de5741 : SUCCESS in 4m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9f73bfe10174458a701655735e66525 : SUCCESS in 1h 09m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e07bc4755d5842c0b89424d4d5a0b5e9 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dd7334c0f7e44468c311ee4db1ea4de : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4df9d39c72614aada36bfe613039e4e4 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f599c8a8e34e4ea02ecc335529bdd8 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/441f9953c2c04dde8d209aa916a1c565 : SUCCESS in 4m 30s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/3127e7310e0d46fa99e5512ecd7b40ad : SUCCESS in 5m 06s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c89493142f514813ab7f13e41aa2fc0c : SUCCESS in 1h 22m 43s","accounts_in_message":[],"_revision_number":1},{"id":"476468ab7083cc2d212b93ded5b13f4fe350fdcf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 11:21:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c52a54360044ba238359ab01043a1f5aa1796650","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 11:22:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c85f16a99e44d04a18d433a1ff761f4 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5946c6baf57044648e59326da8d2c418 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e52dc2accfc16b57c9e712448edf8e1dd6ea66ba","revisions":{"e52dc2accfc16b57c9e712448edf8e1dd6ea66ba":{"kind":"REWORK","_number":1,"created":"2021-02-05 12:00:52.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/31/774231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/31/774231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/774231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/774231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/31/774231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/31/774231/1"}}},"commit":{"parents":[{"commit":"38a06028f2b9176130cfde3f73367c62d9ba28fd","subject":"Merge \"Move jobs to py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/38a06028f2b9176130cfde3f73367c62d9ba28fd"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-02-05 11:26:33.000000000","tz":60},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-02-05 12:08:03.000000000","tz":60},"subject":"Remove lower constraints.","message":"Remove lower constraints.\n\nRemove the lower constraints file and the\ntox environment for lower constraints.\n\nChange-Id: Iefa00bee6f67e9e82f0c703a7aff144c0dcbe1ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e52dc2accfc16b57c9e712448edf8e1dd6ea66ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e52dc2accfc16b57c9e712448edf8e1dd6ea66ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
