)]}'
{"id":"openstack%2Foslo.messaging~776416","triplet_id":"openstack%2Foslo.messaging~stable%2Ftrain~I01a38754a31c891f2b3b9c7f8135690693df5d13","project":"openstack/oslo.messaging","branch":"stable/train","topic":"fix-option-train","hashtags":[],"change_id":"I01a38754a31c891f2b3b9c7f8135690693df5d13","subject":"Fix type of direct_mandatory_flag opt","status":"MERGED","created":"2021-02-18 10:34:50.000000000","updated":"2021-03-01 21:00:02.000000000","submitted":"2021-03-01 20:58:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776416-fix-option-train","meta_rev_id":"b164f723a79e359e63411ba39f340c2424cb672a","_number":776416,"virtual_id_number":776416,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 20:58:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-02-18 16:41:42.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2021-02-18 16:41:42.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 10:34:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2021-02-18 12:25:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 04:58:43.000000000","updated_by":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"reviewer":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"state":"CC"}],"messages":[{"id":"52c5a6166144fca77443323ae5c10fc302101dd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-18 10:34:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"541bc1e69691f34e6267e43255cbf81e109d9fc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 12:25:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c582808dde764066b7ca1694cd401baf : SUCCESS in 1h 27m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ec2ec8ddeac4f15b1d95bfb011bbfef : SUCCESS in 1h 43m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b08bdae0ae046f9ae743ac5315e8941 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3f786d5bb4e4caeb54ac0506663817b : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14f2fe8070c1495bb5e02ed82f024ca3 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/535e51381dd743ba9c9948470e542187 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4843f75bc4b43068f74dce8f108751b : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6faea4731cb34ae2af849729bcea028b : SUCCESS in 4m 46s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/20231eb5d17547ca8c64233013981c19 : SUCCESS in 6m 24s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/6a0a644bc90547f48f6003367fca7b16 : FAILURE in 13m 58s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/bd6cb7587a724b6e8751b8d866ea75f6 : SUCCESS in 6m 38s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/a91dcd2c7c5d4c5295e18d59f23173ae : SUCCESS in 5m 47s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/92d573bbf9af4af7aebfc89e2996accb : FAILURE in 13m 18s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/2eecd4f646ad4ac6914709f24e5e5f7d : FAILURE in 7m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9cf70ed3127c478ebd5f1285d200997a : SUCCESS in 1h 37m 54s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/53b2bdf08b11421d91398397742f7bed : SUCCESS in 1h 30m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/8c49c7bc596b4ba7bddd6d02e1ace849 : FAILURE in 9m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c00d7dc8501345bb8c140ede8fee1b1b : SUCCESS in 1h 45m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/5be05a6262eb4d87a7ab055560551ced : FAILURE in 14m 52s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/8ef1a5df78b8465daab8cf4920826540 : SUCCESS in 1h 00m 39s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/20ca6835c454492397811bb97520158d : FAILURE in 27m 30s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/f3cec88021ef471b95cba779c1b7af40 : FAILURE in 46m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/e4753fe3492a4e77878b4cb3c634a329 : FAILURE in 43m 36s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/b764fa629d6248088689e152ccc5abe1 : FAILURE in 42m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ede999f3f1aa535d53b6c57817a660fe737da3f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-02-18 16:41:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nBackport proposed by a core, single approving.","accounts_in_message":[],"_revision_number":1},{"id":"25d8427f384bc802e6c2ae12fcbd7b88d136db75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 16:41:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"509814a22c8db750e7f0022602151458c71b68df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 18:32:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/ac00481685f94fc789429708c8fb1970 : SUCCESS in 1h 45m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fd6adc8a81745e5af04bfdd8eb9d0af : SUCCESS in 1h 23m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9cfc8e7aa6346a89bd19b59f465e8a8 : SUCCESS in 9m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f27da8479804c1eb39f0ff208f63bbf : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0f2205519774274a52ab560a5ca8a3f : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce8b255284004e1a9dcaa13a4ea78b3c : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe169c8e5ec84811b323961b57edaf98 : SUCCESS in 5m 12s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/0da8bbc62b784a62af96e64d58ac012e : SUCCESS in 6m 12s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5bdd35a9ebca450b9d639f1ebf94f531 : POST_FAILURE in 1h 34m 28s","accounts_in_message":[],"_revision_number":1},{"id":"651d729ec95ada040501a5ebf12767df735e5b4b","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2021-02-22 04:58:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"121bd49b25a16733b682b6536177c192ae8d4cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 06:53:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6186e9cc4958447aa7e42f515807f053 : SUCCESS in 1h 25m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4382703e0bab43988dd1e81799962475 : SUCCESS in 1h 31m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d3d634df4b5433e893e460f4639f576 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f7caace7aa942e9af7b88bfcb91a379 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e13cff44d5c4145a37ff7a56b3a183b : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1d6cd0bc5334d82951d1d026b7b13ab : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5cfe4bb88be74fdb82ff4b81f1833408 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca0e670e6dc49d2aac2f678e2c8070b : SUCCESS in 4m 39s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/c4fe34eb00cb454482bbc417271ed3da : SUCCESS in 6m 06s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/ccbde318e1044a1b92b736cd40f31979 : FAILURE in 13m 38s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/fbc661bfd3f14fe78c771e2424174d9c : SUCCESS in 6m 09s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/4559c50914474dd99de174cbc98dfdc8 : FAILURE in 6m 27s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/a65cdcbb6437417bb3f6b3327e21a04f : FAILURE in 12m 34s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/ce6273ed7b344cb4806bf21faf214793 : FAILURE in 7m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0571d2e0b6794a69b304a07ee7042226 : SUCCESS in 1h 20m 50s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/80e35ac424bb4c56bbddb028b98a4526 : SUCCESS in 1h 52m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/0a3c531c394140a781d8123d12d4203e : FAILURE in 10m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f98ce4b76d0c4ad2b019ce7b5d66a00d : SUCCESS in 1h 39m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/e88d827f33c64589a0b5c210f3a91401 : FAILURE in 11m 01s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a7f2206a842c44259cf84d72a0e971bf : SUCCESS in 1h 13m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/c94113f439db48be93eac41adbd78375 : FAILURE in 21m 15s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/22692f18191f4852b588d8253dd88d38 : FAILURE in 54m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/664b977c93af4a30ac2e21f5afa5d062 : FAILURE in 45m 05s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/c47a0ac8d5194d6d8d68bb3bd9fa71cf : FAILURE in 56m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c9dfff84a687f07c78993fa84591f765ce54de14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 06:54:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00460f3ba1b8425381768e0b8488c0d326cf825b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 08:23:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/627e9a6fd85243718d52eb354ca85ecc : SUCCESS in 1h 28m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/163aea8704ea47c7a72e06ae031a1791 : SUCCESS in 1h 23m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27028305b97f4615bcbfbc4feb23c29a : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fcabf90b26b462a85b60d5961e9fe33 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f97da496eac54eac84b56c321734ebb4 : SUCCESS in 9m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b45e7ff1779c401faa4d6eee4f512294 : SUCCESS in 9m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4c5cd43c1a747c4addda5162d607966 : SUCCESS in 4m 07s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/56fef7f6dc3c4bae834ec84c9a8f6ac5 : FAILURE in 5m 58s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/34800107379a4c5fabec0b54fa86c3fa : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0fce9e618277fcea7ea95adb805b75e63063a3cf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-01 12:33:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ed663799c1aa06b83256a5118bc8b5125caa47da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 15:31:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/58bddd720b2d4f17aa8307a9250aaace : SUCCESS in 1h 27m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7784aecdd57472fb317db72fd5c508e : SUCCESS in 1h 14m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bb6f5613ceb4aacaf8da69e09d3610c : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a558da66f4824e858cb279b6c19c4269 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/969bb46dc60143d8b795f088486aa1ef : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/707993ac4a924cd2b51fd4437b86398c : TIMED_OUT in 40m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/600a00eb68844ea99512b96713d20410 : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c67bd8e2d9ad4d6c81f37895f47685dd : SUCCESS in 5m 04s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/5656eaf13c664794a6480b1a6ecef825 : SUCCESS in 6m 09s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/5b78c93822f54ed9aad4843d1f109133 : FAILURE in 14m 13s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/267d43e2fc7d4e949d7b4207381ecb0b : SUCCESS in 10m 17s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/3f35dbfcb4d64979b1ef09e60ffcefd7 : FAILURE in 7m 49s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/5c34bd1b11db48d88a4fdd506bbb5b32 : FAILURE in 21m 05s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/4a859134eab44bddb85d22ddf615ecf4 : SUCCESS in 9m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bae3a3dc4a6b43628f28030d591efb19 : SUCCESS in 1h 31m 48s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5c1f8339dadb4c478250854573c12962 : SUCCESS in 1h 22m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/eb3e309610524b349e21320127ef0cdd : FAILURE in 15m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b8841e2296b349e58f709c94d9582330 : SUCCESS in 1h 37m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/899b13629148434ab21355b503cc4fd8 : FAILURE in 15m 04s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/9cbacca7dae646febddb69c2d4d34589 : SUCCESS in 38m 05s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/4cc7be9d3a494fbe9a04aaeb347a3596 : FAILURE in 22m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/ef549665f9314032bca6822d6581d64c : FAILURE in 2h 51m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/89c03610fd7749c78560501389292166 : FAILURE in 2h 55m 09s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/73b603efd34d482aaf960a93a57b2d96 : FAILURE in 2h 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8903b923622711aed70820869a2ebaea82cc684d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-01 15:41:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ec70db3c3594b61e27961c4b8e0bad126343498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:49:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/02cc7acf7d1844cda9230c9ed5e805df : SUCCESS in 1h 37m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/230d0f4611d54b69a88c0c81dc89f50b : SUCCESS in 1h 23m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d7438917ff44bbaa82e96d41c139e30 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6416525176747f69ea8334315ae5bca : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f83bc95342f41e88586d83769417f16 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46aab30ca74747b2ab5765dba0950b6e : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4a7e36574fb4c22825c7bbd23a4dcac : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62089fe0063547bdb8b8523c7dfe06e4 : SUCCESS in 6m 00s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/e76280f6d5904f368a6976e1bdcb0aae : SUCCESS in 5m 21s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/471b7befe3ae40838dcf7f8e4b0c680d : FAILURE in 16m 23s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/adc36d514cf54c48b6cab7aab1850bfa : FAILURE in 13m 36s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/8b77381ebed141ab965f2c8644f82167 : SUCCESS in 7m 11s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/994f15e2c4b44918990ec37c4ac9396e : FAILURE in 16m 51s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/f19a2621e73c4838be84add3430d5fbe : FAILURE in 11m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9be9d077f711471caa03bee7b99f38e0 : SUCCESS in 1h 38m 03s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/260ec268df024e45a7333a8fdd76396d : FAILURE in 1h 06m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/02a276bb6296468f95997cd57d27060f : FAILURE in 11m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b3fb7785f6b8497188665abde868eb13 : SUCCESS in 1h 29m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/3017059f8aa5432a9871a0cd95bbd45b : FAILURE in 13m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f18261bb1f0242e4b9fb754dee485e78 : SUCCESS in 54m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/a872d02173684a049cec906ff4744990 : FAILURE in 23m 42s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/1789b9a9255542eab0a390c2e9a2c173 : FAILURE in 2h 56m 11s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/517bb49af49e495ebc412bb45cdc72ca : FAILURE in 2h 56m 23s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/d90317c1c9894520aa31127d7b873664 : FAILURE in 3h 00m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a89e648073785017399b4dacaaef03e25868b42c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:49:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26e93640f3390c67b5cba53222a1a7965c26776a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 20:58:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f30b6d50847444028e9363cdc95c86b9 : SUCCESS in 2h 04m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f937859f10474c32a501d714f4589af9 : SUCCESS in 1h 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b37e8fc2b9e4393bb45d32dc48f01e3 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b537ad90be24f2aaefecf9c73fabf62 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0be51ce672e4f1ea8d52de4a581f5e3 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34da0766a869401abaeb054744a9f146 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/911aa6a392f844e5b997483d9bed2b59 : SUCCESS in 5m 34s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/ed77a0c0effd4967a23ce68706f9ff61 : SUCCESS in 8m 38s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/313f32549669421fbe72d0c9e22b1d60 : SUCCESS in 1h 41m 11s","accounts_in_message":[],"_revision_number":1},{"id":"30c1e3b0f3f877e26b0e9615182a4a76cff9c0ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 20:58:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b164f723a79e359e63411ba39f340c2424cb672a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:00:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39eaa1e6a5ab49e7b637b8909d3e6b79 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56fc332608028ba4c671255e8bd9223a0bae762d","revisions":{"56fc332608028ba4c671255e8bd9223a0bae762d":{"kind":"REWORK","_number":1,"created":"2021-02-18 10:34:50.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/16/776416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/776416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/776416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/776416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/776416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/776416/1"}}},"commit":{"parents":[{"commit":"a430ad873f26e6aa1863587ddc66ed7916c0d5df","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a430ad873f26e6aa1863587ddc66ed7916c0d5df"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-12-22 19:34:16.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 10:30:16.000000000","tz":60},"subject":"Fix type of direct_mandatory_flag opt","message":"Fix type of direct_mandatory_flag opt\n\nAn IntOpt with a default of True is invalid. I\u0027m a little surprised\nthis doesn\u0027t fail a defaults check somewhere, but it needs to be\nfixed regardless.\n\nLooking at where it is used, it appears the boolean type is correct.\nThis just changes the opt type to BoolOpt to match.\n\nConflicts:\n  oslo_messaging/_drivers/impl_rabbit.py\n\nNOTE(hberaud): Conflicts are due to a patch [1][2] not yet backported on\ntrain and which added a new option on other branches.\n\n[1] https://opendev.org/openstack/oslo.messaging/commit/196fa877a90d7eb0f82ec9e1c194eef3f98fc0b1\n[2] https://review.opendev.org/c/openstack/oslo.messaging/+/747892\n\nChange-Id: I01a38754a31c891f2b3b9c7f8135690693df5d13\nCloses-Bug: 1909036\n(cherry picked from commit b8f8b17030e93f6fe64e986ebae1de4011f007d5)\n(cherry picked from commit a0329a3f86e8bf275dd4e16c1f354911a6ff2eb6)\n(cherry picked from commit 3bf9ba550454719f1d56dbbf8bccbabab11a16f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/56fc332608028ba4c671255e8bd9223a0bae762d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/56fc332608028ba4c671255e8bd9223a0bae762d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
