)]}'
{"id":"openstack%2Foslo.messaging~776417","triplet_id":"openstack%2Foslo.messaging~stable%2Fvictoria~I1cbafff03349f7da9224de46285707fbf2a81a68","project":"openstack/oslo.messaging","branch":"stable/victoria","topic":"backport-mandatory-to-victoria","hashtags":[],"change_id":"I1cbafff03349f7da9224de46285707fbf2a81a68","subject":"Deprecate the mandatory flag","status":"MERGED","created":"2021-02-18 10:47:17.000000000","updated":"2021-02-22 08:19:31.000000000","submitted":"2021-02-22 08:17:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776417-backport-mandatory-to-victoria","meta_rev_id":"bfc02412fb0664f6cacbe0421f28a4e2d052e143","_number":776417,"virtual_id_number":776417,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-22 08:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-02-18 16:23:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2021-02-18 17:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2021-02-18 17:20:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 10:56:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2021-02-18 10:56:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-02-18 12:38:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-22 04:59:37.000000000","updated_by":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"reviewer":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"state":"CC"}],"messages":[{"id":"2ebdfec3cca14101c0f17746a9962363d1e93763","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-18 10:47:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ca8542cdd834101f4c1a160245899ddf3598e8f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-18 10:56:16.000000000","message":"Patch Set 1:\n\nI\u0027m not sure that we really want to backport this deprecation too.\n\nMy goal is mostly to backport https://review.opendev.org/c/openstack/oslo.messaging/+/776418 maybe that an adapted version of this patch where we drop the deprecation would be better.\n\nTechnically it\u0027s just a deprecation, the default value remain the same and we don\u0027t remove existing feature. We didn\u0027t give a date for the deprecation, so that match well with the reality and that could help to inform user sooner that potentially this feature will be removed in next series, I think it can\u0027t hurt. For a temporal point of view I don\u0027t think that backporting the deprecation is an issue.\n\nFor now I cherry-picked the both patch to allow us to discuss about this but if we decide to drop the deprecation I\u0027ll abandon this patch and I\u0027ll rebase the other on the HEAD of victoria.\n\nLet me know what do you think about that.","accounts_in_message":[],"_revision_number":1},{"id":"ccc8e833f9568f7f65870b77833ffef720f518d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 12:38:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48c39451e1314368a9a4ba95bde781b3 : SUCCESS in 1h 27m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdde3d1700b74c8783964c270b9f458a : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d476c9ceec4002861e6378915cfaf8 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b7ff648e0f8481980bce08fd19350db : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6aff55a8ac91455187fbffca7884e0c4 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49c8a2ed96544fdaf631e3b2ced97f8 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b2fee6b8299444f91bf14c14baa4a27 : SUCCESS in 4m 19s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/c265e1d01a5b4978b9d69fa3ae359666 : SUCCESS in 6m 48s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/16c4476edd8f479a9a21d78c8605ebbb : FAILURE in 6m 48s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/9e0bdba0f0d34639815e3df0133005a0 : SUCCESS in 5m 46s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/1e01dab9b48a4ce682950656edf79508 : FAILURE in 7m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/66985154f0744abc821cbd71a734a8bd : SUCCESS in 1h 31m 53s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/225ccf67294f4ff7ad8e66e0a6bdd5ff : SUCCESS in 1h 27m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ce5aa80de1e64951bdcc2ccfb8e44131 : SUCCESS in 1h 32m 51s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/68f2fdba25054438992737702b43fef8 : SUCCESS in 1h 03m 52s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/17229bb20c324a41bcd2c142229c13d3 : SUCCESS in 1h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e497b32c740ba6af428c2b1747c024faa7f0805","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-02-18 16:23:05.000000000","message":"Patch Set 1: Code-Review+2\n\nIt\u0027s kind of weird to backport a deprecation, but I agree that it\u0027s harmless and makes the other backport clean, so +2.","accounts_in_message":[],"_revision_number":1},{"id":"0f88324e1fa7b658899a627b5ebbf22587e503b2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-18 17:20:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nBen\u0027s happy so so am I","accounts_in_message":[],"_revision_number":1},{"id":"ac45c92aa6641731e3089fd9c0d6beb0dd4c9f05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:21:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d10c8d5839ed25db43531e6d3ccb64488d48a9cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 20:30:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e84f38d91aa9490f9c95db76e4cd72a2 : FAILURE in 1h 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a04633c125c4536966fee734bbdcfc1 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3c6968224a9418db8cf79db64784da8 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e45d964121c4e6d85af77e1e758d6e0 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4f250546f946c08333497f9335e20e : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0a988ef4a3e41609b22d4edfb60c6c9 : SUCCESS in 4m 21s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/3f255007f8f14d2c87e7a1a8865bc2ea : SUCCESS in 5m 32s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3ad319f47d734d5cadd4005b0055d3fc : SUCCESS in 2h 05m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a4b5676e864676f5c84e8727f57791ff84f785eb","author":{"_account_id":30095,"name":"Michele Valsecchi","email":"mvalsecc@redhat.com","username":"mvalsecc"},"date":"2021-02-22 04:59:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f7f89d031d52e1d19ee37f43472eace358ff7230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 06:51:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/749bcb3f60d14e31981c06f79b57f944 : SUCCESS in 1h 16m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e5572f707d24a7a8ff71b02bc6eb8e4 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d85fac88cc1b4039b035217dac1f8e5d : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e2714cb04d34b7ab6ebd60c93bf77ee : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2754d90b54654c3892239fdce8df558f : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/792d93d605d0407da26444b109369036 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57aeaadfdb1347cf9296c325dbc0064a : SUCCESS in 4m 06s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/55ff07cc312a43a499f33efe0d180c5c : SUCCESS in 6m 00s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/af8585005a954d4eb7d8cd5aa4b5ec09 : FAILURE in 7m 08s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/b4f76b69ba904e86b1b2ea0ee0cbf155 : SUCCESS in 6m 29s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/d54bd09bc5ec48e38e4fd831ee9c79e0 : FAILURE in 6m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f46203a54a234bf89011b5db5e215e6c : SUCCESS in 1h 40m 25s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2935c7d4a45f4f5e8ae4a5c1a1098b65 : SUCCESS in 1h 22m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d9c3ccfac2da456e92c84a79bb0aaabd : SUCCESS in 1h 16m 34s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/28bd092b5ad34896bcb770ee39087f3c : SUCCESS in 56m 10s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d81277cea0304823bd771bdcf1778b8f : SUCCESS in 1h 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17aeb78a9b31ef6a1a49ce2b8ae947f34ce00317","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 06:52:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"718054a4752a75e86cd3d3fc47fd6a3b5752a866","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 08:17:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0ccb60a5b3e42ddad1e610ec6f9bb6d : SUCCESS in 1h 24m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/937db3e43a624afc873203feeb0e4149 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39ebc37678204fdaa1ce00a11af92b5c : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99bb30d29e0349a79e3b322717e4d79a : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f926bb5d3e44eb4b5fd6d1d33de3ef5 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cac533ef961a4740a7836a78f07f79cc : SUCCESS in 4m 08s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/25bfaf90ef8e4eae965da214b64e10b3 : SUCCESS in 5m 46s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/cc4ad4df8bb14af1abc475971a112d01 : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1ebb5ced8ce4ba2a26e7588d8f68ca79764ff6fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 08:17:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bfc02412fb0664f6cacbe0421f28a4e2d052e143","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 08:19:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7279ff6b3c24315aecebfd78f46df3f : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e832a872de1c4d0fa439aa0cb9407d41 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06ad070cc6dbe72d1f96c529f7bd193dbae383f5","revisions":{"06ad070cc6dbe72d1f96c529f7bd193dbae383f5":{"kind":"REWORK","_number":1,"created":"2021-02-18 10:47:17.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/17/776417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/17/776417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/776417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/776417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/776417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/17/776417/1"}}},"commit":{"parents":[{"commit":"a0329a3f86e8bf275dd4e16c1f354911a6ff2eb6","subject":"Fix type of direct_mandatory_flag opt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0329a3f86e8bf275dd4e16c1f354911a6ff2eb6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 13:14:21.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 10:46:41.000000000","tz":60},"subject":"Deprecate the mandatory flag","message":"Deprecate the mandatory flag\n\nIt will not be possible to deactivate this functionality anymore.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\n(cherry picked from commit 2b89d97888c9b6cd8b4f23e5ae7ce3103d5f9e5b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/06ad070cc6dbe72d1f96c529f7bd193dbae383f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/06ad070cc6dbe72d1f96c529f7bd193dbae383f5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
