)]}'
{"id":"openstack%2Foslo.messaging~776418","triplet_id":"openstack%2Foslo.messaging~stable%2Fvictoria~Id5cddbefbe24ef100f1cc522f44430df77d217cb","project":"openstack/oslo.messaging","branch":"stable/victoria","topic":"backport-mandatory-to-victoria","hashtags":[],"change_id":"Id5cddbefbe24ef100f1cc522f44430df77d217cb","subject":"Correctly handle missing RabbitMQ queues","status":"MERGED","created":"2021-02-18 10:47:17.000000000","updated":"2021-03-01 16:04:11.000000000","submitted":"2021-03-01 16:01:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776418-backport-mandatory-to-victoria","meta_rev_id":"7e9abe238efbca3b1ff044cb8e1816a452281a4a","_number":776418,"virtual_id_number":776418,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 16:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-03-01 14:26:30.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2021-02-18 17:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2021-03-01 14:26:30.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 12:55:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-18 17:31:14.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-03-01 14:26:30.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"}],"messages":[{"id":"0b73bf449aed475ccd30aa4b7d9406481f93fc90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-18 10:47:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ae66426e60cbba1804b3b68a18919c74d58d244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 12:55:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd51033e37444b9fa142103468e92194 : SUCCESS in 1h 09m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4875ada0447d40338828a05673983ee6 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd188a2fefca49aaabfe37a0f4438118 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6dbec687ca64028bfe6981dd40c545c : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfd8193cc1a14047965918c295182878 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489ff83302d94d90ba4b6668378dc83a : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2655b14557c34f48a96554a339194083 : SUCCESS in 4m 10s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/8779c7c3daf74e7d8c9d843a7d41b4c6 : SUCCESS in 5m 49s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/2eb30d8a8ba8401c89fb496d831d0fbe : FAILURE in 7m 12s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/98589624e36c4ebd9000588123ba8598 : SUCCESS in 6m 37s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/05ebf039a6de411c8a00d5cabde77d69 : FAILURE in 6m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/df07467d57254110b6c01ea28f6e14e2 : FAILURE in 1h 12m 19s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b578868c4fc146f193651b235cab68c2 : SUCCESS in 1h 47m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/99e7a215707c4d01b5b6e064c648e1bb : SUCCESS in 1h 26m 52s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/43b2cc2457784c38aa1ae0d7cd436810 : SUCCESS in 1h 06m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5a77c232e484f6691db4343091c3604 : SUCCESS in 1h 14m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7e648cd19e2c62279b97b71cffd2a6c33fb46fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-02-18 14:21:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5918460f8c8c7a8450fd4eda934d18e3bac9fc54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 16:08:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfca39fabe204c2a879bd83231f9befd : SUCCESS in 1h 37m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b63d9b1c59f14b71969925f42c1858f3 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eabbee9597a747d4b8cf22dfa15a0d23 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/556937df85cb4d9faf53c77d1e5091d5 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f56ef9973337416796adc6bb0e4c6962 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b7f496bae4a44e1a713fdb86bb1cf6f : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40b7f2d9e8b14cd7a56977b65e8c1fa8 : SUCCESS in 4m 05s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/f9af7a23441c4ae9b2badccca715363b : SUCCESS in 6m 17s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/7a60abcc825f4dc482dd8204d7eee9e2 : FAILURE in 7m 10s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/a03b39c0f1454eb3a8d6f44549dc34af : SUCCESS in 8m 26s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/18624cf4c17747619176b988165c5c96 : FAILURE in 6m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4acf72c969974f82a24776eef212cd52 : SUCCESS in 1h 23m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e41a1bf7bdf74ccaaccb381275308a89 : SUCCESS in 1h 34m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d5084c2f0bb04e69ac396c88c2cf57e4 : SUCCESS in 1h 32m 01s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a5b7511e464047e2a997d02f7567964e : SUCCESS in 1h 16m 04s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f67d415625cb41c5ba827bcd3551cab9 : SUCCESS in 1h 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0867968dcf88761d47e8f3475ee044d6b0539a6a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-18 17:31:14.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport. Sensible fix. LGTM","accounts_in_message":[],"_revision_number":1},{"id":"dde61a9830f8f2d4e2e846151c3b9ca4573c757a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-03-01 14:26:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d03df6b6e4fa52baa3cfc43cd3419026b470e85e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 14:27:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6823a6ab649e545b43ff66cf245d7da29837d29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 16:01:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa93c09739a4409cbf956b0f1ff0c6ef : SUCCESS in 1h 30m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42a7c9f50c49414d85e11b79f0873bfc : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ee88964142e4aaeb72e24fe1e119392 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f17f3521f6243479815dc027af5b442 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e0aa68ff3ff4d228d2ab0c13a6e28d7 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a55ddf253f674733a6272d798810fc3d : SUCCESS in 5m 14s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/476038fba33844169c3648a8f5118bba : SUCCESS in 6m 12s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a8523303868c4205903e19dfe4cdd731 : SUCCESS in 1h 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e07035ff9bf22aa1b297f0f3b7eb546436aad457","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 16:01:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7e9abe238efbca3b1ff044cb8e1816a452281a4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 16:04:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5caf123af69405aa20d5cf11b2d0212 : SUCCESS in 1m 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc012ed88fe340048522e97d830f83bd : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"391ce7fc69adc8d713d8ca64e76d901eb3e65df1","revisions":{"391ce7fc69adc8d713d8ca64e76d901eb3e65df1":{"kind":"REWORK","_number":1,"created":"2021-02-18 10:47:17.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/18/776418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/18/776418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/776418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/776418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/18/776418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/18/776418/1"}}},"commit":{"parents":[{"commit":"06ad070cc6dbe72d1f96c529f7bd193dbae383f5","subject":"Deprecate the mandatory flag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/06ad070cc6dbe72d1f96c529f7bd193dbae383f5"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-08 14:54:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-02-18 10:46:49.000000000","tz":60},"subject":"Correctly handle missing RabbitMQ queues","message":"Correctly handle missing RabbitMQ queues\n\nCurrently, setting the \u0027[oslo_messaging] direct_mandatory_flag\u0027 config\noption to \u0027True\u0027 (the default) will result in a \u0027MessageUndeliverable\u0027\nexception being raised when sending a reply if a RabbitMQ queue is\nmissing [1]. It was the responsibility of the application to handle\nthis exception, however, many applications are not doing so. This has\nresulted in a number of bug reports.\n\nStart handling this error condition, using a retry loop to attempt to\nresend the message and work around any temporary glitches. Since\nattempting to send a reply will will no longer raise an exception,\nthere is little benefit in retaining the \u0027[oslo_messaging]\ndirect_mandatory_flag\u0027 config option: users setting this to False will\nsimply not benefit from the retry logic and improved logging added\nhere. This option is already deprecated though and will be fully\nremoved in a future release.\n\n[1] https://www.rabbitmq.com/channels.html\n\nChange-Id: Id5cddbefbe24ef100f1cc522f44430df77d217cb\nCloses-Bug: #1905965\n(cherry picked from commit 4937949dffecdf8863a7876e5a6b0b18e811c3ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/391ce7fc69adc8d713d8ca64e76d901eb3e65df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/391ce7fc69adc8d713d8ca64e76d901eb3e65df1"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
