)]}'
{"id":"openstack%2Foslo.messaging~778175","triplet_id":"openstack%2Foslo.messaging~stable%2Fussuri~Ibbb811e988cc048e6c39a0c2336c716a1a0ddb30","project":"openstack/oslo.messaging","branch":"stable/ussuri","hashtags":[],"change_id":"Ibbb811e988cc048e6c39a0c2336c716a1a0ddb30","subject":"Move legacy grenade jobs to Zuul v3","status":"MERGED","created":"2021-03-02 11:41:11.000000000","updated":"2021-03-15 18:15:20.000000000","submitted":"2021-03-15 18:10:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778175","meta_rev_id":"e351b883cd0295f0ecce875b7387ee7186a587fa","_number":778175,"virtual_id_number":778175,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 18:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2021-03-15 16:38:26.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2021-03-09 14:54:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2021-03-15 16:38:26.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"}],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 11:41:11.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2021-03-02 13:22:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 11:04:48.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-15 16:38:26.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"}],"messages":[{"id":"9118cd68cad9f0ad5a2c94cc449b98503d1df3b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2021-03-02 11:41:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3519e114a562a0abce73c06248782ed3d579bc67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:22:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42184ebfbf884865af6e14e21b29bfee : SUCCESS in 1h 28m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f4b49f7855d4e488bf65b7155e0171c : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6897ae6fb3da4c81a04d0315901943cc : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1da9e97a657947549d7bc4496a3db307 : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7e54cf8f8eb4be0bf91fb0d8aea3772 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f19b1058e8413db9630418c0a80070 : SUCCESS in 5m 40s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/002589ec91f24107aa1d33750bb9b73f : SUCCESS in 6m 00s\n- oslo.messaging-tox-py36-func-scenario02 https://zuul.opendev.org/t/openstack/build/98f3d5ae2fed4f6a896417642f6179ea : FAILURE in 13m 31s (non-voting)\n- oslo.messaging-tox-py36-func-scenario03 https://zuul.opendev.org/t/openstack/build/0b5733a7ec474ce795e80d67170e1430 : SUCCESS in 10m 20s (non-voting)\n- oslo.messaging-tox-py36-func-scenario04 https://zuul.opendev.org/t/openstack/build/463786e0407645c1a9912bb77bc2ec29 : SUCCESS in 12m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3d1953638cd145479bde5b70e4ae56d3 : SUCCESS in 1h 32m 07s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4611cf0cfba7479bbf0b0f26817f1504 : SUCCESS in 1h 29m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/27cacfef6d71429eafb0c684f6dc1fb8 : SUCCESS in 1h 33m 51s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4bd92e06bde9461cb199341b9279a70a : SUCCESS in 1h 07m 48s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/71d00b7ca25248f88358ef9b1c1267e3 : SUCCESS in 1h 27m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6971060b4081883774b1a993ba1887f6abcf7b32","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-09 14:54:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"491278bd90bc04f077ba9eaff87eae2156e5fab3","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-03-15 16:38:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2df891166d56cd411fcbe13024c6d1b425b41559","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 16:38:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bca707481327841ca5de965ebed06398e4771bb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:10:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf6cd111402d43fc8d5945a8b2654a62 : SUCCESS in 1h 30m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2467cae9cdac47e7b1b1e3b0d7df7bc2 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0e87b63ed284ca0be54b2a78a41f77a : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51d87668f3cb4835a2f7e651bc7e7acf : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d632075d294f10b46ae878fa366a21 : SUCCESS in 6m 01s\n- oslo.messaging-tox-py36-func-scenario01 https://zuul.opendev.org/t/openstack/build/7db68fd22d2a4c6c93711d080434651b : SUCCESS in 6m 42s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/17233362bd86435c9d845cb092dde6a4 : SUCCESS in 1h 13m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6139fbdba82b75bdef94f27f453d0613159b9407","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:10:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e351b883cd0295f0ecce875b7387ee7186a587fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:15:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dfbfed3b939459e84287bb99454ff83 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb3f9bd6f0ba47120d2c88b40e328104dda78450","revisions":{"eb3f9bd6f0ba47120d2c88b40e328104dda78450":{"kind":"REWORK","_number":1,"created":"2021-03-02 11:41:11.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/75/778175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/75/778175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/75/778175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/75/778175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/75/778175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/75/778175/1"}}},"commit":{"parents":[{"commit":"3bf9ba550454719f1d56dbbf8bccbabab11a16f6","subject":"Fix type of direct_mandatory_flag opt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3bf9ba550454719f1d56dbbf8bccbabab11a16f6"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2020-07-20 13:51:15.000000000","tz":-240},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2021-03-02 11:41:08.000000000","tz":60},"subject":"Move legacy grenade jobs to Zuul v3","message":"Move legacy grenade jobs to Zuul v3\n\nAlso deletes obsolete playbook files for telemetry tests that have\nbeen removed.\n\nChange-Id: Ibbb811e988cc048e6c39a0c2336c716a1a0ddb30\n(cherry picked from commit a5ae9f4aa0e18787b4bb676b3e725c225deae96d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/eb3f9bd6f0ba47120d2c88b40e328104dda78450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/eb3f9bd6f0ba47120d2c88b40e328104dda78450"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
