)]}'
{"id":"openstack%2Foslo.messaging~778865","triplet_id":"openstack%2Foslo.messaging~stable%2Ftrain~I1cbafff03349f7da9224de46285707fbf2a81a68","project":"openstack/oslo.messaging","branch":"stable/train","topic":"backport-mandatory-to-train","hashtags":[],"change_id":"I1cbafff03349f7da9224de46285707fbf2a81a68","subject":"Deprecate the mandatory flag","status":"MERGED","created":"2021-03-05 09:44:04.000000000","updated":"2021-03-17 11:09:27.000000000","submitted":"2021-03-17 11:07:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778865-backport-mandatory-to-train","meta_rev_id":"591a6390da9723a0a83924055ff6aa0ee32f3d23","_number":778865,"virtual_id_number":778865,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-17 11:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":2,"date":"2021-03-17 11:09:27.000000000","post_submit":true,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2021-03-17 11:09:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-17 11:09:27.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2021-03-15 16:21:53.000000000","_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-05 12:48:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 14:50:42.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2021-03-12 12:52:18.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-03-15 10:53:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"a1f69d4a15778a5dc3c29a7e6121b73f23a9f4ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-05 09:44:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85ffd2f2acd9a18550f872a02fc6d2ca262bc077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 12:48:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/3235adfc9ff941b1a36c9de5aebc176f : SUCCESS in 1h 51m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba5fbccd795746c094e91d5e49c2d7cf : SUCCESS in 1h 32m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a70eb2729fb4db4a4a26690a9386977 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d910193e33504909ad6ecc8237bc25a7 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3e12033c49454190ac6abe589db2957b : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d205d3b2f36748b9ad9d991c2687b391 : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d902fb16b6e949b5a6890eeb4719be7d : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0f7dc340c0483581eb5e26b5f88b74 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ed7d2efc18448f39e3d136b7ceb2b2d : SUCCESS in 4m 33s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/294b70a796824d2789fc08928a0b07e4 : SUCCESS in 6m 10s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/ef52c8356c9743b0a68425b8e6d96453 : FAILURE in 12m 54s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/87c5d1f20f4943acaace3f7f1f9d9aa5 : SUCCESS in 6m 39s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/fef738d3937349a3b805f2fa4bb1161d : FAILURE in 7m 13s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/fc779d13edc544d285e5a65b697c2888 : FAILURE in 13m 45s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/1ee0cad2f4bb469e99dd579e48520d74 : SUCCESS in 6m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bfa83fad09534f6eb48d6a610ca542b5 : SUCCESS in 1h 40m 46s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7cfd31a547844884a534a32cb28092f2 : SUCCESS in 1h 43m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/e9ab60bb37f04b9e80f5d7eced9adcfc : FAILURE in 15m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c00bde96b2614f45916f573735c7a190 : SUCCESS in 1h 44m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/b717706c7878444cae5914bdc18db292 : FAILURE in 13m 20s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/178ec5c3159e42f895cd44f8df508381 : SUCCESS in 1h 00m 34s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/1775d3c23e764070906bf63c23294c5a : FAILURE in 22m 00s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/24d65491f0204e2b8b0983f02d0f385a : FAILURE in 2h 51m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/8680d3c05f844bc2a6216be21075aa64 : FAILURE in 2h 51m 49s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/470b023d014f4e3da1593e95cbd8cd2e : FAILURE in 2h 54m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e734dd2db4ae813e3f9dd708effbde549d772d4b","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-03-11 14:50:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0c1478e7809d5a347fc22cfaac365be1840d38ac","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-03-12 12:52:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a82f4b194ef6108b1ac50f133bde72d027442246","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-15 10:53:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport. LGTM","accounts_in_message":[],"_revision_number":1},{"id":"05535e4ab0664704f8b633a2dd94d8522b623afc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 12:13:54.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e1dfcb2d5ed7cb1c092f1e952ff20517fdae6587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-15 12:17:43.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a31774bf7872e3db34c42c8dc2749a9c5cc3d781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 15:22:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/2888534b3e2f4248957dc9a928c580d0 : SUCCESS in 1h 15m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0aa3d461242c4bccb84e66bb1d5e763a : SUCCESS in 1h 29m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6b4ce22c9d045a8b1ce84bab8e8d73e : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bff16a100ee46d8936a29dee315724d : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/acb01b3b4dcb4a879c1ab78e95f1e281 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d33c67995421438792a6874e2b583d58 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7add6e7309284deab67b91d0494ba2a1 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045632a798d241db9b6f9c39dbaf864c : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9a0c0b387e04bb1b32634819c8f071f : SUCCESS in 5m 04s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/833e5a86ccae4528896134875c228de2 : SUCCESS in 6m 32s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/7b55629dbc914a1e93e6c34d6c1bad5a : FAILURE in 18m 08s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/6d285a94685b49e3848fe8ba48f6e232 : SUCCESS in 11m 23s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/cd381256b69d4aeaa9c75c2923445656 : SUCCESS in 6m 17s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/325054bf81e040249a36ddc1af039ef3 : FAILURE in 17m 35s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/dfb4caf16d7f4b1d988fd6acedb3eb1f : SUCCESS in 13m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e11db2ebc7f543f1ae66b79c57372cec : SUCCESS in 1h 34m 43s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1ef992bd9af94d5498e1d21e7303a1e8 : SUCCESS in 1h 22m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/8a16a38ef6f844dfa07d4c62c9a6365f : FAILURE in 16m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5ee9f3d63d214d0ea829c9122f761be5 : SUCCESS in 1h 34m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/ffd5987c7c4d4635b50802c9e49fc45b : FAILURE in 12m 21s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/6857e6ef9fd54a3595f2c77c3b2b0596 : SUCCESS in 57m 11s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/23eede1aed624c49857ea01078ae40b4 : FAILURE in 26m 35s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/3896bb27cf514a40837e75ce70f02b97 : FAILURE in 2h 58m 11s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/025a4aaf11394b5a8d515a494de72db7 : FAILURE in 2h 53m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/67b13ceb86ef409cbddeb796052311d9 : FAILURE in 2h 58m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c3582a38b4d6580f5c511d2ee0f440bfa1695ea","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-03-15 16:21:53.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b9751c407c469c667eb0839c4cc6fc8f22ba652d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 16:22:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9d072f29b0436ad249987f127d9174a13211f09c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:18:20.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"e210040afec46757dc2729be667f81ff4c5c0940","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-16 08:31:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b26da001b18144cb35640de51732a1e971460b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 11:30:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/20bbe011ea564a88bddc87c682656514 : SUCCESS in 1h 37m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a047ea4aa724a3dbb47997f93fcec00 : SUCCESS in 1h 23m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1a8c2137e8445e6ae605eb095e917ef : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b9b6dedad4b4d2c9395ca2e1b45caf9 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4141158e8fe48858e18131b1c93c3e5 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/772bb43c83ad4466af8fbe4fa30db175 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d481bf6aa8884224b3d68c08874fb7e9 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c845fead211141eb9363f8993508c9f2 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60dc5fdbe712459d829428c2d0706596 : SUCCESS in 4m 09s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/74113c0368824f92954369edf61c463c : SUCCESS in 5m 31s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/4c03dcaac34e4582a0381d528149277e : FAILURE in 12m 53s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/5b0a297df9b24730af8cfd2fc997b91f : SUCCESS in 5m 45s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/e8a75c6a541545e5bc391ff4585f8f75 : SUCCESS in 6m 01s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/fc4e0df3c2704c5092524c4a7cdc4283 : FAILURE in 12m 30s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/d6b2113fab09495e8c91723cae6dd683 : SUCCESS in 6m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/678ad506421a453b8672e5043e8bda72 : SUCCESS in 1h 36m 30s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/71b38a8c353a4046956a29caa42a1a1e : SUCCESS in 1h 27m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/7330321fce624624a01540fc8bdf9fd2 : FAILURE in 12m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a1897c695c184ed6b41879012d207320 : SUCCESS in 1h 29m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/9ec3c532a23f453f8005271b1abde682 : FAILURE in 10m 42s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/b2b725c13a714c1fbaa7450651af9375 : SUCCESS in 58m 29s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/6ea5f6679dd841089461fa0aa8728d96 : FAILURE in 20m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/039a30b3d1a9413cb5cd0a368a35c46a : FAILURE in 2h 52m 24s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/afdeeda899844f1b9a617c2359643b98 : FAILURE in 2h 53m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/a9a3b16feff24b07852be4c486e9089e : FAILURE in 2h 53m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cfe77e91b5ad115028f174c51c345b1e9d00cda1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 11:33:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2ca4d025226721c1404c48a419e036fba06e1d67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 13:25:22.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"82f01c10eaba5fd27b8983baf673e44aa9fed8be","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-16 13:33:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"233c0c4c3fcd6cecc7a069b9fcf3caf430441c46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 16:34:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6218f1c82c394c4fa2502a38d4b3f33a : SUCCESS in 1h 21m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/966d2e4aa06b4508b4108f79ffdc0e28 : SUCCESS in 1h 21m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b7543a0ca344d4f8049669182742a2f : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5ac6d71ade447e6a719d7bade9dbb2a : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0fc223f31ff345a6b61e615da50d63d3 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0126cc7922f84cc5b45f336969c90048 : SUCCESS in 7m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35e27f4f78f647aca27f17254fadb38f : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca9d845ae9c44c06a65886c2dc51db5d : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd98f0d09f8b4435805b0ceccb75f79e : SUCCESS in 5m 25s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/dc39aa3f8101420ea7aee6ee9a8ccbcc : SUCCESS in 6m 34s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/fa8f137cc98b492bb25e09ba0a1792d4 : FAILURE in 13m 37s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/69cd87eb67a44fa4a19dec859e4e0908 : SUCCESS in 8m 36s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/3840add72cd1491ebb495cdf174cd519 : SUCCESS in 7m 59s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/48f9afa5ef5f4e93bc82729819ba5bb4 : FAILURE in 14m 06s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/fd3cbf224e2e458286115e50615af6e6 : SUCCESS in 7m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bfb8517abc7c4b65b55ece5fc7cdcb61 : SUCCESS in 1h 43m 57s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7f88ba4ee6a64530a5ae9bc19d29ff90 : SUCCESS in 1h 26m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/04b1cc006af84b369021b14a46e224ca : FAILURE in 16m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/13a1b79946a54d67845be49ad56e7202 : SUCCESS in 1h 50m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/4d9a737cdb07450dbb8b1593fe7d5f71 : FAILURE in 14m 17s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/1aed19c5cfbd4facb32f8bc3c2f83c4f : SUCCESS in 57m 14s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/c75e3f10128f4423a746d5daa7af26b3 : FAILURE in 22m 12s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/6f1965bea4844129aa1026e10fbf7097 : FAILURE in 2h 53m 43s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/faef5973484b44899684c6017781d0db : FAILURE in 2h 53m 29s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/52e707241c7b41c19320784cc72dc690 : FAILURE in 2h 54m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92a9153539d8f3bfc44a056d086bcb9b230f513e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 16:35:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a817f1112bbfb728520d2983433c075477ab027","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 19:08:12.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b2cc3f31a0494f439d0e54b5f7a7848a : TIMED_OUT in 2h 06m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ca77ab396654ffc885d5e2a6b0a308f : SUCCESS in 1h 28m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bababdcfa044f42b23f884f6d115956 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f11ac0b11f14e71bf9807f1337c88fd : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9dd57d541884207ad6b74fc1cc4fb91 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/388feb15cf9744d1889af490ef1e9f17 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25ba61c5f5734baaa5274d741160c5db : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75773e9a2f904420b4dcf24638a3ab55 : SUCCESS in 5m 12s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/92a98fd34d5f4b118fff111a1bd58ba4 : SUCCESS in 6m 35s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fa740dcc21e544d0b45b621101300d99 : SUCCESS in 1h 34m 09s","accounts_in_message":[],"_revision_number":3},{"id":"f98e4ba7242f77425cd91a18cfda5477eeaa1e25","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-17 06:22:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"49ca7e4450668b9c8d3e707678e56dff2d6903bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 09:24:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b010efa8c9434c0db5cf8d4348a8a0bc : SUCCESS in 1h 25m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf0adf7192044096b878012519afe42d : SUCCESS in 1h 19m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7378548f1df14d76b3884026ab7049cd : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a77a9d9bb124f06a7cb4580b5e12660 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b479ef64b0d746d789816ec44176325c : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e40274e9a3c641eaa4438c159d87e464 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a53c464c1078413a9dc5f876b848dee9 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a77166f277444878fb473e26f2fcb0d : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3213aa795f9f404a98f994899e11b6a9 : SUCCESS in 5m 14s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/a92667d785ca4509ac70ba50577460e1 : SUCCESS in 7m 09s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/cb25fd82710b442a90e404e9edfe697e : FAILURE in 15m 24s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/fdf1324890074cee80b095b08649b951 : SUCCESS in 6m 32s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/0fd7dd6eafeb470b946b10c8a87606bc : SUCCESS in 7m 00s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/3fedfd67e0c9487aa415bd0caccb5ef0 : FAILURE in 15m 03s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/2acf76c16de2421d98cae87413855a17 : FAILURE in 8m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/49b11806292a41ea8fb7e932df0a51fa : SUCCESS in 1h 28m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4bae6dc2d8774b34a36937e3f804393d : SUCCESS in 1h 15m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/57e53b3da6b64f758cbc3bb4e689390d : FAILURE in 10m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/15997db0859b45fc96c19d43fd5b2177 : SUCCESS in 1h 23m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/46a06f31575143ca8186b662d0964042 : FAILURE in 19m 13s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d287a1dbe7814d059272c0a9cd871441 : SUCCESS in 54m 58s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/5b7e0fc65f4143b7a25d5e82aac21150 : FAILURE in 27m 21s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/876d027945fc479db54cff0ca6681c4d : FAILURE in 2h 52m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/64cee1325c87421e903b7f0138178642 : FAILURE in 2h 53m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/6fd96b3035174a6f959aa497e2e7fb8e : FAILURE in 2h 52m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"228753210f355c6a5fa255f1103f79e11b97d30c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 09:24:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a2e55c8f1edffd29da788efafa8da47cd02ffdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 11:07:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f1a5e0aa66044097948619d21ae416d7 : SUCCESS in 1h 36m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59e7ace83006480baf6b9b27bc355f7b : SUCCESS in 1h 25m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f00e2b016a0c4ca5afa331d0a2b36935 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/441ba47507e743b2a0b88ee015f8eb9d : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bd35f09837c4fa9a3572eb75e343ac1 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0dad5da0baf4939b0f218b7e3d11cb0 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a11f2a474b8b45b59a796e1942f0164b : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/900f736d3f214db0b98097683d09e9fb : SUCCESS in 4m 11s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/80c2b31135f5493dab660c460f0fdd33 : SUCCESS in 5m 49s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e706e5543e8d4a0a8610af5437e6fe53 : SUCCESS in 1h 29m 55s","accounts_in_message":[],"_revision_number":3},{"id":"08183c60c8f6bfbb053c7c5706e6d075c1d5920a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 11:07:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d71868adf04a2e75c0c9fe311dfc316f1b627c5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 11:09:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67cda536f9c343b9a99bb7e777b010db : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c05b555888214bd4baf5781370e682a4 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a0aa3a88a370c33baf733fff39de8722e4c7e3e3","revisions":{"c6f51fc3619408326e85c08a8f6660790193ac02":{"kind":"REWORK","_number":1,"created":"2021-03-05 09:44:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/65/778865/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/65/778865/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/1"}}},"commit":{"parents":[{"commit":"53061647f508482eb0c4011badd76887bc930da8","subject":"Fix some typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/53061647f508482eb0c4011badd76887bc930da8"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 13:14:21.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-05 09:43:32.000000000","tz":60},"subject":"Deprecate the mandatory flag","message":"Deprecate the mandatory flag\n\nIt will not be possible to deactivate this functionality anymore.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\n(cherry picked from commit 2b89d97888c9b6cd8b4f23e5ae7ce3103d5f9e5b)\n(cherry picked from commit 06ad070cc6dbe72d1f96c529f7bd193dbae383f5)\n(cherry picked from commit 5f2a4f21d0a9988e4a939f2092a523956b4d4583)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c6f51fc3619408326e85c08a8f6660790193ac02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c6f51fc3619408326e85c08a8f6660790193ac02"}]},"branch":"refs/heads/stable/train"},"afc7e8b11ff337e4b1383907a0d769fe277c80e8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-15 12:13:54.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/65/778865/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/65/778865/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/2"}}},"commit":{"parents":[{"commit":"750c59ddaac307ca9082be36cef5527b2bc75241","subject":"Fix some typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/750c59ddaac307ca9082be36cef5527b2bc75241"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 13:14:21.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-15 12:13:22.000000000","tz":60},"subject":"Deprecate the mandatory flag","message":"Deprecate the mandatory flag\n\nIt will not be possible to deactivate this functionality anymore.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\n(cherry picked from commit 2b89d97888c9b6cd8b4f23e5ae7ce3103d5f9e5b)\n(cherry picked from commit 06ad070cc6dbe72d1f96c529f7bd193dbae383f5)\n(cherry picked from commit 5f2a4f21d0a9988e4a939f2092a523956b4d4583)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/afc7e8b11ff337e4b1383907a0d769fe277c80e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/afc7e8b11ff337e4b1383907a0d769fe277c80e8"}]},"branch":"refs/heads/stable/train"},"a0aa3a88a370c33baf733fff39de8722e4c7e3e3":{"kind":"NO_CHANGE","_number":3,"created":"2021-03-15 12:17:43.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/65/778865/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/65/778865/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/65/778865/3"}}},"commit":{"parents":[{"commit":"72ad2823b28a73eb1cd1285d29de20a645d81b00","subject":"Fix some typos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/72ad2823b28a73eb1cd1285d29de20a645d81b00"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-01-18 13:14:21.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-03-15 12:17:12.000000000","tz":60},"subject":"Deprecate the mandatory flag","message":"Deprecate the mandatory flag\n\nIt will not be possible to deactivate this functionality anymore.\n\nChange-Id: I1cbafff03349f7da9224de46285707fbf2a81a68\n(cherry picked from commit 2b89d97888c9b6cd8b4f23e5ae7ce3103d5f9e5b)\n(cherry picked from commit 06ad070cc6dbe72d1f96c529f7bd193dbae383f5)\n(cherry picked from commit 5f2a4f21d0a9988e4a939f2092a523956b4d4583)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0aa3a88a370c33baf733fff39de8722e4c7e3e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0aa3a88a370c33baf733fff39de8722e4c7e3e3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
