)]}'
{"id":"openstack%2Foslo.messaging~784417","triplet_id":"openstack%2Foslo.messaging~master~I3d2e95c783cb0ca88ea8657acd9f262ea9244e9d","project":"openstack/oslo.messaging","branch":"master","topic":"fix_hacking_version_pre_commit","hashtags":[],"change_id":"I3d2e95c783cb0ca88ea8657acd9f262ea9244e9d","subject":"Move flake8 as a pre-commit local target.","status":"MERGED","created":"2021-04-01 13:38:51.000000000","updated":"2021-04-15 16:50:03.000000000","submitted":"2021-04-15 16:42:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784417-fix_hacking_version_pre_commit","meta_rev_id":"b5b00572fcd45bc0ad1b9baa3c01e99b3712ff6f","_number":784417,"virtual_id_number":784417,"owner":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-15 16:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"all":[{"value":1,"date":"2021-04-14 17:33:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2021-04-14 12:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2021-04-15 14:49:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"reviewers":{"REVIEWER":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 13:41:40.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-01 13:42:48.000000000","updated_by":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-04-01 17:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-13 10:18:06.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2021-04-14 12:21:35.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"a5f2a5d14f58eb025e1f4a7f1a6bafe6b1fc8a58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-04-01 13:38:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"715a4243b7d301c6f5e14c8dcfd2bbcaf648f640","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 17:32:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50448d050c6a463d95c9d5850781e8fb : SUCCESS in 1h 27m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66a11cb607a74fe69537b768ef751b89 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b141df8f9c2d4306960e1d6f3bd106a8 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75117f5e54064b8aafde9bbb3d65f5b6 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4d4fefc79d04e15aec81155fad2674d : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebdc442bd0064a19b314db98c5ffe98d : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84e27cc689f54ce9bb95eedebb32c8da : SUCCESS in 5m 20s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/24c494d8e31c484e9148d751fff48f24 : SUCCESS in 6m 27s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/345560c4482a49b8887efafee1e46cba : FAILURE in 7m 47s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/d31d636d0d5249fea727de5ce0848d31 : FAILURE in 6m 10s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/57eefc3ee47e4e5dad15239e99c4b8a6 : FAILURE in 7m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0187d0ff10614ea4a5b8e8cab113cb72 : SUCCESS in 1h 24m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2072a81ec74a47baaf8bb1cbc4b8b71c : SUCCESS in 1h 37m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/107e55d323b24c399c5d20f63545e6cf : SUCCESS in 1h 38m 03s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/13ef68675a824518a1dd8cc8a4fe995c : SUCCESS in 1h 05m 40s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/33afe6eedb3c477fb19689058db055c7 : SUCCESS in 1h 23m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b7a4dd48ffe70880c6ed8528f441b5fccf071f6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-12 12:59:18.000000000","message":"Patch Set 1: Code-Review-1\n\nFirst, thanks Ken for these changes.\n\nThis issue should be fixed by following this method \u003d\u003e https://review.opendev.org/c/openstack/oslo.db/+/781470\n\nThe problem here isn\u0027t the hacking version but the problem is that we determine the version of flake8 to use instead of using the one pulled by hacking.\n\nPlease see this ML thread \u003d\u003e http://lists.openstack.org/pipermail/openstack-discuss/2021-March/021251.html","accounts_in_message":[],"_revision_number":1},{"id":"5a9f7ff228d51d332c9cacf6290d759ee11a7e9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-04-13 10:18:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"384f14f0e9b05052470af1d07f99c2020f18aad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 12:13:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/456af543963e4ecdb44a647281e9565f : SUCCESS in 1h 31m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcbfb38d0ebe411a9580226cffdae857 : SUCCESS in 9m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63555fb46cc940c591a28a4babed59bb : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/947e0e8997a34661907c0005f35ab1f2 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/150ecabaa3d844838ac92aaa0f31c0fb : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1b811476246453f9a2546560e68851c : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a218cd29ab3406385355c2367f290ac : SUCCESS in 5m 21s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/ca8dfb66973849aa86aa615595a05530 : SUCCESS in 7m 22s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/8532ae18b6304a0ea995a4d9f59fdb29 : FAILURE in 7m 25s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/b623bf9cea7e46708a3dc7b6d3b49bfa : FAILURE in 4m 49s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/290ab943dae0484c8ed3dd4b843a9771 : FAILURE in 5m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/15c8ed2da0c349159a5dce865230fca9 : SUCCESS in 1h 23m 31s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c4c00dbf066a4b8aade5a4f559995cd7 : SUCCESS in 1h 11m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7d9369200aa74d49978cff3a151b1f35 : SUCCESS in 1h 53m 10s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fe5629c9e5624be5a024e982c9863a15 : SUCCESS in 58m 04s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b29151376a74f988a64768f45789e9b : SUCCESS in 1h 16m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"70396fabe8ddeb32022eb9217d4c993dc139f05c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-14 12:21:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d550c3f0ef7e132f80ab3daf8172ce2dee306a21","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2021-04-14 17:33:45.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks Daniel, lgtm.","accounts_in_message":[],"_revision_number":2},{"id":"058284398d75ae7cfeb385b59475ad1d0327ef1e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-04-15 09:53:13.000000000","message":"Patch Set 2:\n\nKen, you\u0027re welcome.","accounts_in_message":[],"_revision_number":2},{"id":"4964daec3a775e5f9f9015ab893b3a397c8fe06e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-15 14:49:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7b16af350391c2cb61fb684ce2f3918fa4c5b89b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 14:56:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df08b9b935b8b8b782a7d8390c8e3c86fd5c1fa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 16:42:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/576ff16e9d1e4a399ae4cd4c067bfcdf : SUCCESS in 1h 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c0b95c115164a029928ab826039e17f : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9509a29d31b34a72b92b17254baf11fa : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cca15ee2adce49578cf17e95e00a6895 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde1f7d2cd2543f8b18b58f9e1662d16 : SUCCESS in 5m 31s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/7ab51ef01fc549c3b6dedc75f04ee0d6 : SUCCESS in 7m 09s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/634261eae5934765932f5b6525dadca6 : SUCCESS in 1h 35m 29s","accounts_in_message":[],"_revision_number":2},{"id":"64ad11f79bcb76d2551a8977d03656929da73378","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 16:42:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b5b00572fcd45bc0ad1b9baa3c01e99b3712ff6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 16:50:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78906aef08984518b4f630c74105e268 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f43a70687cf9e4704612780f2e8d026661d2aae1","revisions":{"e4c456d565cfc4239cc05298e5dfdb8f70601ec9":{"kind":"REWORK","_number":1,"created":"2021-04-01 13:38:51.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/17/784417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/17/784417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/784417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/784417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/784417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/17/784417/1"}}},"commit":{"parents":[{"commit":"5aa645b38b4c1cf08b00e687eb6c7c4b8a0211fc","subject":"Merge \"Remove lower constraints.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5aa645b38b4c1cf08b00e687eb6c7c4b8a0211fc"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2021-04-01 13:36:11.000000000","tz":-240},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2021-04-01 13:36:11.000000000","tz":-240},"subject":"Fix the conflict status with hacking.","message":"Fix the conflict status with hacking.\n\nThe conflict is caused by:\n    The user requested flake8 3.8.3 (from /home/zuul/.cache/pre-commit/repozi7hytk7)\n    hacking 3.0.1 depends on flake8\u003c3.8.0 and \u003e\u003d3.6.0\n\nInstall the latest version of hacking to fix the conflict error with the\npre-commit hook\n\nOriginal fix by Daniel Bengtsson\n\nChange-Id: I3d2e95c783cb0ca88ea8657acd9f262ea9244e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e4c456d565cfc4239cc05298e5dfdb8f70601ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e4c456d565cfc4239cc05298e5dfdb8f70601ec9"}]},"branch":"refs/heads/master"},"f43a70687cf9e4704612780f2e8d026661d2aae1":{"kind":"REWORK","_number":2,"created":"2021-04-13 10:18:06.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/17/784417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/17/784417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/784417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/784417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/784417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/17/784417/2"}}},"commit":{"parents":[{"commit":"5aa645b38b4c1cf08b00e687eb6c7c4b8a0211fc","subject":"Merge \"Remove lower constraints.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5aa645b38b4c1cf08b00e687eb6c7c4b8a0211fc"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2021-04-01 13:36:11.000000000","tz":-240},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-04-13 10:23:56.000000000","tz":120},"subject":"Move flake8 as a pre-commit local target.","message":"Move flake8 as a pre-commit local target.\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach[1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements[2] will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nCo-authored-by: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\n\nChange-Id: I3d2e95c783cb0ca88ea8657acd9f262ea9244e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f43a70687cf9e4704612780f2e8d026661d2aae1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f43a70687cf9e4704612780f2e8d026661d2aae1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
