)]}'
{"id":"openstack%2Foslo.messaging~786442","triplet_id":"openstack%2Foslo.messaging~stable%2Fstein~Icf48ef72fbfff88dda2714b386dbbfe355dc1acb","project":"openstack/oslo.messaging","branch":"stable/stein","topic":"drop-LC","hashtags":[],"change_id":"Icf48ef72fbfff88dda2714b386dbbfe355dc1acb","subject":"Dropping lower constraints testing","status":"MERGED","created":"2021-04-15 13:58:47.000000000","updated":"2021-04-16 12:22:37.000000000","submitted":"2021-04-16 12:16:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786442-drop-LC","meta_rev_id":"d423430a116b92dd72c38e6cbeaeef8ee9fb53b9","_number":786442,"virtual_id_number":786442,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 12:16:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-04-16 10:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-04-16 10:27:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-15 13:58:47.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-15 15:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d94d7411c67a6552c559d3aab5043f75440ba758","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-15 13:58:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77f443546eaf985c774a36ad8b3fd50496d09387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 15:47:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/28463f0a5a4e48069dd876ede97719f4 : SUCCESS in 4m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ed85b606e41e4bb489ee62ddaf6e5d5a : SUCCESS in 1h 45m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/450b8e75cad7499db8c02354cc899dc8 : SUCCESS in 1h 21m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58ccb1ba835b4c07b78c7485eae7fa32 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c9de251f4cc4d24bdbdd89b25e76683 : SUCCESS in 7m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/114eae615bd84d42b3c6dbba95d57952 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc1f80b1585345708264b20e701e2c4d : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/734b509ed57b45ea9d038c7a404ead20 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13eedbb1c2e94fefa88f09706b7c91fc : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46c3960df9f4412cbaf8ed711a611b35 : SUCCESS in 3m 59s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/8125f2582345427996b2d95db90d6523 : SUCCESS in 6m 39s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/29761f65027e4694b8e1b9493ada4adb : SUCCESS in 8m 14s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/ee87842326ca48b595fd0a975e65c40c : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/ab3123e0ece940aea5efa16154231246 : FAILURE in 4m 48s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/15a05ad57f3e4ed58edb160d069e559d : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/701b0df53896436bae837bfbd0162c10 : FAILURE in 4m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/84da6ac8304e4af8ae5ce820455824af : SUCCESS in 1h 42m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/43caa73fbb6f4b0094fb65a353d66dc5 : SUCCESS in 1h 34m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/af5938d83ba5455ab73db78dc36a9437 : FAILURE in 13m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b868394718534200a92a598dbc751aae : SUCCESS in 1h 30m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/b8900a57c05a48a1a58b43d84718297d : FAILURE in 16m 22s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d5aa44d3e5e248949a58259234bf6ae0 : SUCCESS in 1h 05m 56s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/79fd753a46024b3788c1cefcfa5a69c1 : FAILURE in 1h 00m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/2c1aa3b6a0f043bdb21a0b33b268d82b : FAILURE in 32m 08s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/997f5a9b87d34b75a2b577c9d5a2db8c : FAILURE in 33m 34s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/fee63641c5054cf89ecc10e84763a640 : FAILURE in 36m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e406009b493d8855d6ce78d6c38750403c2fbe19","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 10:27:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis is an obvious fix that will unblock the important fix next in the series. Self approving.","accounts_in_message":[],"_revision_number":1},{"id":"c28247cbcbcf2c55b95b0f76190792ffb9a634a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:28:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aae5afe7251ec662807cb046aa51e9334d12a493","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:16:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c99c320068648cf9adec8a17617fa2e : SUCCESS in 4m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1071d31b215040b3b9d9c1f51675cda1 : SUCCESS in 1h 35m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c927174092a940c7a57ebc759a3bcca9 : SUCCESS in 59m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bacf6a409154d56961d3b2f1ce25240 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/22fe5efac02549828856976818f809a1 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0a06d66549047348961a432ddf87455 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cccd785e5cb44a37b688dc964a5492fe : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a56518d0749409396ee1360f217b8c7 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adb6efe477044e4898bf4757de34c648 : SUCCESS in 4m 27s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/2f8ebfb5cfb643d39df2aa462bcd299e : SUCCESS in 6m 40s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/949e0a9d015d4321a2c24c3ef1949b26 : SUCCESS in 1h 46m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b628ed73a0b24e5d66578db46ac4179e56309817","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:16:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d423430a116b92dd72c38e6cbeaeef8ee9fb53b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:22:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38cefc09d01d45178afcc4e35d86e204 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/610a2572507b4e3093b01314b54e7315 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c47ed4e1b03dd6999a3ae464eb311297a8ac4cd","revisions":{"0c47ed4e1b03dd6999a3ae464eb311297a8ac4cd":{"kind":"REWORK","_number":1,"created":"2021-04-15 13:58:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/786442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/42/786442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/42/786442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/42/786442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/42/786442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/42/786442/1"}}},"commit":{"parents":[{"commit":"a9939f46946a88fd882b6eabb13c51d11269d059","subject":"Add oslo.messaging-tox-py27-func-rabbit non-voting.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a9939f46946a88fd882b6eabb13c51d11269d059"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-12-18 14:09:42.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2021-04-15 13:57:34.000000000","tz":60},"subject":"Dropping lower constraints testing","message":"Dropping lower constraints testing\n\nWe facing errors related to the new pip resolver, this\ntopic was discussed on the ML and QA team proposed to\nto test lower-constraints [1].\n\nI propose to drop this test because the complexity and recurring pain\nneeded to maintain that now exceeds the benefits provided by this\nmechanisms.\n\nAlso removing lower constraints and related tox environment as they are\nno longer needed.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-December/019390.html\n\nConflicts:\n  lower-constraints.txt\n\nNOTE(stephenfin): Conflicts are due to different constraints in\nlower-constraints.\n\nChange-Id: Icf48ef72fbfff88dda2714b386dbbfe355dc1acb\n(cherry picked from commit 686a8b61826d9bed0a989e645a13c708c3e6343c)\n(cherry picked from commit 1682ef69ae30732071fb4296f947047c37c4622e)\n(cherry picked from commit 7e151b15b949e00e5d6a1b6ba93c4fbbb38274bc)\n(cherry picked from commit a430ad873f26e6aa1863587ddc66ed7916c0d5df)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c47ed4e1b03dd6999a3ae464eb311297a8ac4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c47ed4e1b03dd6999a3ae464eb311297a8ac4cd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
