)]}'
{"id":"openstack%2Foslo.messaging~791877","triplet_id":"openstack%2Foslo.messaging~master~Ifbea6ef06e230127bfa16fc3dc0ddd9f236cfbed","project":"openstack/oslo.messaging","branch":"master","hashtags":[],"change_id":"Ifbea6ef06e230127bfa16fc3dc0ddd9f236cfbed","subject":"Upgrade the pre-commit-hooks version","status":"MERGED","created":"2021-05-18 03:14:39.000000000","updated":"2021-05-26 19:02:58.000000000","submitted":"2021-05-26 19:00:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791877","meta_rev_id":"abafdfcd757f11f262b67db5825bd759de33098d","_number":791877,"virtual_id_number":791877,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-26 19:00:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-05-26 17:00:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-05-21 13:25:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-05-26 17:00:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 04:41:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 13:25:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-26 17:00:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"7a6f46fa0cb35ea832103c0dc3d1032c64b3ac33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-05-18 03:14:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b253b9cf48d8a3637bdd7640dfb22668d8885c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 04:41:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7025b799e9f34d7db4abcb89b68e4891 : SUCCESS in 1h 26m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a27f946f867d48caa3d11a280e21a592 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73dfda1764f84bcf83d2129d56507d2b : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7db0a11224b24203a728bc0d69247651 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c13e007dc06441095127eb01ac5248e : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3111aabcf014326a84db3dd645dd117 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d61a0741e6f4d689939c2fbec88d616 : SUCCESS in 4m 34s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f8c627a610214670b7e89afcc30c2942 : SUCCESS in 6m 20s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/9cb4f463e2804af99c0ebb35305c4e7f : FAILURE in 6m 37s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/e68b85fcab7c45ea8286ce3b5b6a2268 : FAILURE in 6m 05s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/7894a36d63c449cdb583104ba34888c6 : FAILURE in 5m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/93ef357030494411a6cf7c636e2f21a7 : SUCCESS in 1h 09m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/23c99bc0f9494db7a11b339ba5651421 : SUCCESS in 51m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3002774330ee42b79e83f986ec014215 : SUCCESS in 1h 17m 52s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/1a862849f777408eb69a7eb3b05f1a0b : SUCCESS in 43m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe2762235e7b44f6ae4f3906a3a26a93 : SUCCESS in 1h 16m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"214e55e750afe697b6bbc13ccbc38b21df94e3f0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-21 13:25:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9a02a13c2d656178d60bf7cc70ad0fa74a499ccd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-26 17:00:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"024a1ad7196399406a9cf748774a9569ea26d185","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 17:01:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68382b7dddfb459245838b725075f2d33ce3108d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 19:00:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcbb75ff47ee4e17a99b4d88aa7fb980 : SUCCESS in 1h 28m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10de783f88dd45b1b004776aedd12608 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82a789f787524c1097140386ae4bf1ce : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4d7425f1665400982c0995316c5f1df : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/377b9b9292204d3a9e1877af2b893746 : SUCCESS in 5m 47s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/fab4c2d08133427485b4a9bcbc318ae4 : SUCCESS in 5m 39s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/abbf72c127f24cd78292ad2804af8be8 : SUCCESS in 1h 31m 27s","accounts_in_message":[],"_revision_number":1},{"id":"db0e9c35b7c910207ac5af605b536a8a39c93fcc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 19:00:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"abafdfcd757f11f262b67db5825bd759de33098d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 19:02:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f616060edc4c08be09df91d656932d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1f9fbeadf5bb164004e8a2c9b6eefad2accace0","revisions":{"c1f9fbeadf5bb164004e8a2c9b6eefad2accace0":{"kind":"REWORK","_number":1,"created":"2021-05-18 03:14:39.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/77/791877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/77/791877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/77/791877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/77/791877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/77/791877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/77/791877/1"}}},"commit":{"parents":[{"commit":"355e0bd79bc2757008ac883874c405b5a952bdf8","subject":"Remove references to \u0027sys.version_info\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/355e0bd79bc2757008ac883874c405b5a952bdf8"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-05-18 03:14:07.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-05-18 03:14:35.000000000","tz":480},"subject":"Upgrade the pre-commit-hooks version","message":"Upgrade the pre-commit-hooks version\n\n[1] https://github.com/pre-commit/pre-commit-hooks/commit/9136088a246768144165fcc3ecc3d31bb686920a\n\nChange-Id: Ifbea6ef06e230127bfa16fc3dc0ddd9f236cfbed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1f9fbeadf5bb164004e8a2c9b6eefad2accace0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c1f9fbeadf5bb164004e8a2c9b6eefad2accace0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
