)]}'
{"id":"openstack%2Foslo.messaging~794716","triplet_id":"openstack%2Foslo.messaging~master~If129b56be47952d018b9f6024d2a192950c1a974","project":"openstack/oslo.messaging","branch":"master","hashtags":[],"change_id":"If129b56be47952d018b9f6024d2a192950c1a974","subject":"Changed minversion in tox to 3.18.0","status":"MERGED","created":"2021-06-04 07:12:42.000000000","updated":"2021-08-10 09:50:01.000000000","submitted":"2021-08-10 09:48:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794716","meta_rev_id":"48fb8972d491be98edae56f32d13926c4f3c62e6","_number":794716,"virtual_id_number":794716,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 09:48:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"recommended":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"all":[{"value":1,"date":"2021-07-14 06:09:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},{"value":2,"date":"2021-06-09 12:25:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-22 11:20:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},{"value":1,"date":"2021-08-10 08:26:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"all":[{"value":1,"date":"2021-07-14 06:09:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 08:41:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 12:25:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-22 11:20:40.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"f914bd11c208dd2b43c440014408c888045cd1c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-06-04 07:12:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3842a5d5e9ccd466f82d933756f2b1253b86175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 08:41:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92cbf644077141d89b3e5073841a7468 : SUCCESS in 1h 27m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e2828320c6143ed8d9ca75c43ca5c2b : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0142e9318132463f8241b16415ef312d : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/475362f0459345288ef26ad31080d756 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c03d471caab7417a935b1dab9fd94389 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bab8162d8f6470eb7f8e38efcd32506 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2858b4b7be774f048de1b63198df48cd : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d4002ad78814d799a1b32da3e346fd6 : SUCCESS in 4m 29s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/95ba65a1e6ab4af59dea4d974fe17d48 : SUCCESS in 6m 37s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/cadc5902e64346e8af85c85093652b58 : FAILURE in 7m 43s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/7b20b79ed92a4ce98e2b11136ea7cd17 : FAILURE in 6m 19s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/d2b931ce05f3419abbc5ff39ea58ea91 : FAILURE in 5m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8c65a0f2504c4ea28d2a792901933aba : SUCCESS in 1h 22m 50s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/31af9d9d571c41cabe8918b10712730a : SUCCESS in 1h 20m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8bb4902f6c46485f8f8972687aaf3e74 : FAILURE in 1h 18m 37s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9412a2b85e91482aa2b5447f68b77ec5 : SUCCESS in 1h 03m 18s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/68e0d39bc94c4a9b83040e9725f89fb6 : SUCCESS in 1h 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4900781662afdc767edc072ab896183145411c43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-06-07 08:59:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22be7565b76baacd0418aaafcda050ae9f29b089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 10:37:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78715a57d5a24e32aac988f2be2c0397 : SUCCESS in 1h 26m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/139861266cd641948b07ca1a95d6f33d : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a830f2cb9d466bb9b668399b87d9a8 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff77526dfb534600b78f81a10bb6dbcc : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/320bdff4fc324e199b00a920ae1f5dbc : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad9b34c8d0984b4295ba7a0768a87a1b : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09380a6fc79f420e9940ac229737a941 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aeeb5d8b392540e0a4839e635c7a8b1e : SUCCESS in 4m 14s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/390835566dbd4179aca86205392a9448 : SUCCESS in 6m 26s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/80251c2a452341779256179bc5dd1611 : FAILURE in 10m 27s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/e0cbf1571e4b46e7b5821db1e89f5dc7 : FAILURE in 6m 15s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/dccbf5d47ba64e14b2d36a851acc2777 : FAILURE in 9m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/dd7ec8d4ad10488780465338d71a1436 : SUCCESS in 1h 21m 02s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/abcd75581fd64d8fbc9a05adc6342679 : SUCCESS in 1h 22m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6929e692e3b24d2794332b6fff95206d : SUCCESS in 1h 36m 42s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/af7f4387830d47c5af9628f8958b060f : SUCCESS in 52m 14s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1dc8407a84c44adbb53f10a80e69db5 : SUCCESS in 1h 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d10626c484cfdae05d4bdc9568d907fdea018b3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-09 12:25:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7b29b5e211aa53d8ccbd3a12ea673e0dcb8f305e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2021-06-22 11:20:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8004054a4990ecc8533ed0e2458b41bcb950a220","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-07-14 06:09:04.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0f63bf841f0b1a8c8c820d737ddc523dcf92dcca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-10 08:26:25.000000000","message":"Patch Set 2: Workflow+1\n\nTrivial changes let\u0027s continue with a single approval","accounts_in_message":[],"_revision_number":2},{"id":"7ced8d661e56059af2c259515e72cb2f7ef40c20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 08:26:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aca943e3bbec17caa03afb0e3388b1897209cf0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 09:48:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84c1b2dca1ba4246b4b5a84b1f0df779 : SUCCESS in 1h 16m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36db5fdaf5d44b98adea0e4af828d6d5 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a7c5b30459c4f65abc4511f481dfe16 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/223790a655f64931b832a6710b60dc84 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33d35f4de784ebdbcf9eb709ae16278 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bca137545ba4a2d8e093db856677122 : SUCCESS in 4m 32s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/07f5178883584f899234d19f5846df6b : SUCCESS in 6m 27s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/96557ff0e05f408f9ed0fd96b1ae630a : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":2},{"id":"855e9685a31b06efebcc1d24c0caec7f9378dbf4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 09:48:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"48fb8972d491be98edae56f32d13926c4f3c62e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 09:50:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b712ec04da4fdf86a8729e5be07a18 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc78cc0570bd4d5d99005ae6fcc93cb0 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3e589edd902a3bb4a1f313ca877314e52f1b0520","revisions":{"e30eb46ee6dc7dd057fea4d638593f4de67cefac":{"kind":"REWORK","_number":1,"created":"2021-06-04 07:12:42.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/16/794716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/794716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/794716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/794716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/794716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/794716/1"}}},"commit":{"parents":[{"commit":"15ad4c2b7a19bd334a79aab6f2bb0e0be29bf371","subject":"Merge \"Upgrade the pre-commit-hooks version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/15ad4c2b7a19bd334a79aab6f2bb0e0be29bf371"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-06-04 07:10:14.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-06-04 07:12:38.000000000","tz":480},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: If129b56be47952d018b9f6024d2a192950c1a974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e30eb46ee6dc7dd057fea4d638593f4de67cefac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e30eb46ee6dc7dd057fea4d638593f4de67cefac"}]},"branch":"refs/heads/master"},"3e589edd902a3bb4a1f313ca877314e52f1b0520":{"kind":"REWORK","_number":2,"created":"2021-06-07 08:59:19.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/16/794716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/794716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/794716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/794716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/794716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/794716/2"}}},"commit":{"parents":[{"commit":"15ad4c2b7a19bd334a79aab6f2bb0e0be29bf371","subject":"Merge \"Upgrade the pre-commit-hooks version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/15ad4c2b7a19bd334a79aab6f2bb0e0be29bf371"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-06-04 07:10:14.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-06-07 08:58:52.000000000","tz":480},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nChange-Id: If129b56be47952d018b9f6024d2a192950c1a974\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3e589edd902a3bb4a1f313ca877314e52f1b0520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3e589edd902a3bb4a1f313ca877314e52f1b0520"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
