)]}'
{"id":"openstack%2Foslo.messaging~797874","triplet_id":"openstack%2Foslo.messaging~master~I50c7de71bfd38137eb83d23e910298946507ce9f","project":"openstack/oslo.messaging","branch":"master","topic":"fips","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-04-26 14:15:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50c7de71bfd38137eb83d23e910298946507ce9f","subject":"Add a new option to enforce the OpenSSL FIPS mode","status":"MERGED","created":"2021-06-24 10:58:37.000000000","updated":"2022-04-26 14:17:23.000000000","submitted":"2022-04-26 14:15:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797874-fips","meta_rev_id":"481cc6b46cf9a5b79c169fa24f1244ace4d9c40c","_number":797874,"virtual_id_number":797874,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-26 14:15:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-03-22 12:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-09 15:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-03-22 12:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-24 12:33:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 15:56:20.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2022-03-22 12:14:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"af348ed00dd4332f90398b50de912dfa1736451c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-24 10:58:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b602da8d2927b6b69b982a05bfa8e73ad1af4635","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-24 11:28:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4603f9dc6b32197c942e498748dcdbd780f35d94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 12:33:00.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c036fb4ac254c259f237ace571c87ba : FAILURE in 23m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2444107bac704dccb609a2878fbe0c93 : FAILURE in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ce28a1c24104df0a43893cd6af1d28d : FAILURE in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd32cd0c030846e5b5cc5177f86b8ce0 : FAILURE in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/610c6adeed7341a691fa864d033abf95 : FAILURE in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29620702c51c42f29a6b912b28082db7 : FAILURE in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f01e2345278449eb24ed1e0e4bbdf24 : FAILURE in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e333c822c074166a4e0362b1877f67d : SUCCESS in 4m 56s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/fa159929d5ed4725a66b94eee65cd695 : FAILURE in 5m 06s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/a7445d1849574511af924f94147b2b52 : FAILURE in 5m 55s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/dfd953f3da6747bbb54814b9373f7182 : FAILURE in 5m 50s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/60cd52448e704ddcac271b39c07ed3c7 : FAILURE in 5m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3101a5f44c0e4f0da0f2207e148d92e2 : FAILURE in 25m 24s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4ce6428eaf104a34b393f95084a9d702 : FAILURE in 27m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/da03b77ba95a42449cde3c8245c67da2 : FAILURE in 20m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/bcf59416c1cf4666b6b861c2709ac50b : FAILURE in 51m 35s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcfa3491c77346709e495a4205a61a98 : FAILURE in 1h 02m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"674b31a07150a029217b5a5ea71aed5fe94270c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-24 13:02:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ffd323722036e2f14d3482f8035270849c45ab25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 19:09:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02f485a7276a464aa78da2fe9fcee095 : SUCCESS in 1h 23m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67a7426e8ccf485b9d2a1745ff1cd077 : FAILURE in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41075f0f077b4020b24a601f4c1bbcec : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db0a8a4ae8e741a19f4c2bc9616357b0 : FAILURE in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ce823fcfb2e4b9093640267db6408e8 : FAILURE in 8m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/782dccc9015940e4a9b7c8cc1c5b62d3 : FAILURE in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c3ec42695240999e0283fcdc4278c2 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27feb871aadc4e2d85a93916d56a47ef : SUCCESS in 6m 08s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a20899740622470fb963df998f00de32 : SUCCESS in 8m 17s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/2d36c26da7dd499d93d5a39acf3dabfe : FAILURE in 9m 45s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/d41204ce06e34711916a95a400b41773 : FAILURE in 7m 53s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/d9291f817a90423e818216e56c44b909 : FAILURE in 7m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/07688b5827084d20a5506e57ef2d2d22 : SUCCESS in 1h 23m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/012c4ef3ef8a4f42b716b7c375771cd7 : SUCCESS in 1h 21m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3328e606309a4a63a209d996f4a88fa5 : SUCCESS in 1h 19m 11s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b48fad96b0154fb8aaab0978fb29e3bf : SUCCESS in 1h 07m 11s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b5f55e690f54e25a23bd6b062e1bd23 : SUCCESS in 1h 14m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ed14087ecee74b540a2d5f732723d2ef28ee0eef","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-08 10:45:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"086a67c23c15346fce272301a947848a0de8f7b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 12:12:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42dc63ecfe7245b0ae379525ffdb49a5 : SUCCESS in 1h 20m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b85b963dd05240aca49ebf2153f6191a : FAILURE in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca9c93c5301e4824a82cce68615b62f6 : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce36c25d72734c7aa32c5425904a6e11 : FAILURE in 10m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/384f6d58cad245beb031cc8c734c4f6e : FAILURE in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/553994eb4650485d934c389f0a593142 : FAILURE in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8f4fa2f61f41878692e313ba090e4c : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3c695c738404ac3b67b03b26a63a28d : SUCCESS in 4m 11s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/35b243a6d6b244afacc831ca2550a577 : SUCCESS in 6m 39s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/d36c8c296c9943039c603606641b2c4f : FAILURE in 7m 08s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/585d11ec7b914f5f8e7b14ec9e6610c4 : FAILURE in 5m 49s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/4dc5f2b6e72b485fba610d73a33d8af0 : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ba688a58865e4f72aae655c949370afc : SUCCESS in 1h 21m 04s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/64fca976da064fa39b598951b2a06f04 : SUCCESS in 1h 24m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b4af1e8990934aaaa94da0cbc3295aab : SUCCESS in 1h 14m 40s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/975769521f3b495aa232b5c5c52050ab : SUCCESS in 54m 48s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c481d164b2854c6f86953236f5f55c7a : SUCCESS in 1h 21m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"34874baf1a1a03e3481baba8d184fa70ebb87562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-08 14:05:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f6d9185935561c3927d634c60949e20dc674d1cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 15:42:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae6feefb7a5a4533bed2db5e188a1426 : SUCCESS in 1h 25m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb2a8654cf4f4f429289412f192b6b85 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e827e0b999749c5b076a544964fe17e : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e637515564944d2a9d7a75398edea5d7 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2f601333adc43309cc2621e571257bd : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66a46f31ff1a4083bcab277d74b45ad1 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56807e5919e94706b9d19de539ac6bba : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9af2e2b2c8ac4063bcdce19ffe76e0f1 : SUCCESS in 5m 36s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/27641cb0d9054890b2a09a1847b28325 : SUCCESS in 5m 36s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/7b36c12229f748359e73d5ae9c0e4544 : FAILURE in 7m 43s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/2e9ab007c3c840e6af7ac4bdc7492144 : FAILURE in 5m 03s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/6a0aa82fa1fd4c88baff37b1a05f7e10 : FAILURE in 5m 53s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/40bf3b7d73bc4c8bb39534669428fb33 : SUCCESS in 1h 28m 11s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/8963761ee96f4986be8fd53f13b0ec89 : SUCCESS in 1h 11m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5b0409b43011438a9c0378b9cf3ae2cb : SUCCESS in 1h 12m 10s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/bc9026888ebe4cc8a91d8500ece02956 : SUCCESS in 48m 17s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb1ae48ce67846938118f0b72773a2cf : SUCCESS in 1h 25m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9b5722f90522e8679dbe1355c04f3b640aa910cd","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-02-09 15:56:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"272a20bbc139cff37b8a59420d1bc704bce5cfbc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-22 12:14:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9eb9479db3e5e28c91c0d40da9b609ebe6a82140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:18:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bf6cf7abbe647a066eaa3d93e86de3e811b3a006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 13:50:05.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e7cf21f06b9437d8e734fa7ced63d89 : SUCCESS in 1h 31m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c19874bbea114c51b9ba7adaeca1de86 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d608731497e94b96882177430f6ed22a : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac2ffdbc55704cd684a4699faa4399b3 : FAILURE in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7cde636475e4e95ac665ded4121cc6d : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32d0098a7d6347f5b0060922d163c0d5 : SUCCESS in 5m 23s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/dcf00588c7484935afded0ea718ebaa3 : SUCCESS in 7m 12s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0864450c39f34e4e9e07fad9d006ff16 : SUCCESS in 1h 17m 03s","accounts_in_message":[],"_revision_number":4},{"id":"be14c80e36f399cfa90f00e119b4fcfc9ab2aa59","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-03-23 12:26:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4792824bc90b1b75e2f3633e53e445eab31ddd46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 14:14:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be3ae8b9af67444280257c396659cbb5 : SUCCESS in 1h 40m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/932bef78c60c40e5907596597cc86494 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4626863076d34a97975c79e7e56a1289 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e75ff8c735a04fefa9ccd83a08118e0f : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f50a950e7f574227bdeb5727f0022466 : FAILURE in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0cf09ece07a4729b895357b2619d2cd : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30390a8bc45a4bed9f1fa389a019179e : SUCCESS in 6m 12s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/aa597891300b4cc0aee41dce16db4394 : SUCCESS in 8m 11s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/4c13ffa0b0ab4773b28ea67b4b26cbe2 : FAILURE in 9m 26s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/6c6ed09b986d47ed96db35ebdcb1dde2 : FAILURE in 7m 27s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/95310b12465f4b8088593f673b5c513d : FAILURE in 7m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/511929e109294a82bfe00007d8d965f1 : SUCCESS in 1h 17m 25s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/11fe3949bb0c479fa69218edb1ef5b1c : SUCCESS in 1h 35m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/88e2753410264580b794de850de6282a : SUCCESS in 1h 40m 22s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a16ea15f0d694a97a23a7572cc8c6e91 : SUCCESS in 57m 36s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/05ee6c9ab88941838a5ffe6624726b9b : SUCCESS in 1h 03m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"81624b855317e9dd81b521ce319108faba1aa992","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-03-25 15:56:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2d8fb4ca97f159ba51c5f9a685335ca298c49720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-25 17:05:19.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4a39c6be7804595b00fd2e13ce5567a : RETRY_LIMIT in 15m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e52eaf78522341e49be36a76829851f3 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5835cb7e72b469489158d00aeda42fc : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aca3538ee858420098e0517f33fbff44 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fff983bcb60f4cd39210f155f362cd19 : FAILURE in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f880d9782c314cea8365dcd4f955643d : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8f8e62bd90f444dab821571ad80453c : SUCCESS in 6m 18s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/022da0dd653b4db3baefabc2947323ba : SUCCESS in 6m 38s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/04fab093c83145abb9db72e1079ec8e5 : FAILURE in 8m 54s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/77ba3c3b6e594a86a9779658d98e716d : FAILURE in 6m 05s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/9fe8b1345bf047d6af42811e0316091b : FAILURE in 7m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/eaf346a9b35c48a18f85047b2234546f : RETRY_LIMIT in 15m 57s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/366d633367174578b1fc6e4763bcd5fb : RETRY_LIMIT in 8m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/28e234c2d10a4e6b8e6d9e22aafffcef : RETRY_LIMIT in 9m 24s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0eb8d4f12f634810a397eb94a4378ecb : SUCCESS in 1h 03m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/f96842389ec04f1e9effa43645ccab35 : SUCCESS in 1h 02m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"49d388f0c3f3eb2327040058ee67fc0b9d19354c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-26 10:41:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a86a449a17dbaa4cd60fd1953e98e41fe85e018f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 12:34:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99ae7a54fb364d1c855cea09ae17dca7 : SUCCESS in 1h 36m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26b25ed651de4436a3fcd35625ce4c88 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b899eb6316b4093b2bb34e73f95896c : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a09b38f075445b58a0e77920e838efe : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e1f230c23eb42af9f8dcac464e19ebe : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198b88eae14545769a5aba54bd0cc6ed : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cd77991f85e47b5adb1bdaee6c9e9fb : SUCCESS in 4m 13s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/b0c17c874a3a4c7b92ed113938c2fcee : SUCCESS in 5m 48s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/550a7aa4023541d4a97d6a00aabc4fd9 : FAILURE in 8m 13s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/9cf369e5c36a4f9185660a38e729423c : FAILURE in 5m 33s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/2ee1072df69b486b973b08c89a911ec3 : FAILURE in 5m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e059d86096bf42d88d7b6785766b1dcc : SUCCESS in 1h 35m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/89262443cb2349d29b07453fed9960c4 : SUCCESS in 1h 21m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/dd0ec9abf6084d478f77a867c9437b44 : SUCCESS in 1h 44m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b1ad54e0a4884a879da641bd462b0972 : SUCCESS in 1h 06m 01s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b48daa0a9e15405281a3055eba2279ce : SUCCESS in 1h 16m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ed2509c5bdcccf44dc0bcfbc7ec7758962def367","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 12:35:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2caa4852babe898ffb22be323bc6256eb41d224c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 14:15:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1809069bcc74bd09c00fddb0c8fb980 : SUCCESS in 1h 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1263b189b59499d9ae529fcf41bec25 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b036dccc8e4d44b68f20f0c082ba9b51 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/549e2801027c45ec9c2227b8a5967510 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e00d14298104b3092c9d2a5267c78a6 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0980a3d3a5344a88223bf0c6482a983 : SUCCESS in 5m 01s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/555efd1f9cca4407adaddff9d11146e7 : SUCCESS in 5m 44s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/045d3f20d36740bdb1f5b507f31f165b : SUCCESS in 1h 39m 28s","accounts_in_message":[],"_revision_number":4},{"id":"9446680d5f4b7512d62050ad5023087753e0af23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 14:15:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"481cc6b46cf9a5b79c169fa24f1244ace4d9c40c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 14:17:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ea9252ac9341129b0839c4925d3b03 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3c2d80970a744a8b91c5e8c181a83e4 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"384738a92d1d840bec762086ec10523c52f7497b","revisions":{"422785e2a9a767a8c54a4d9d069f485f859b1174":{"kind":"REWORK","_number":1,"created":"2021-06-24 10:58:37.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/74/797874/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/74/797874/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/1"}}},"commit":{"parents":[{"commit":"f1b47b6ba8ee4888f6cba1efac0fbf747c5406e6","subject":"Merge \"Add Support For oslo.metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f1b47b6ba8ee4888f6cba1efac0fbf747c5406e6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-24 10:53:55.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-24 10:53:55.000000000","tz":120},"subject":"Add a new option to enforce the OpenSSL FIPS mode","message":"Add a new option to enforce the OpenSSL FIPS mode\n\nThis option ``ssl_enforce_fips_mode`` allow us to enforce the FIPS mode\nif supported by the version of python in use.\n\nhttps://en.wikipedia.org/wiki/Federal_Information_Processing_Standards\n\nChange-Id: I50c7de71bfd38137eb83d23e910298946507ce9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/422785e2a9a767a8c54a4d9d069f485f859b1174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/422785e2a9a767a8c54a4d9d069f485f859b1174"}]},"branch":"refs/heads/master"},"af045fc818e9973f7724969bed743078c9c80c0b":{"kind":"REWORK","_number":2,"created":"2021-06-24 11:28:01.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/74/797874/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/74/797874/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/2"}}},"commit":{"parents":[{"commit":"f1b47b6ba8ee4888f6cba1efac0fbf747c5406e6","subject":"Merge \"Add Support For oslo.metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f1b47b6ba8ee4888f6cba1efac0fbf747c5406e6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-24 10:53:55.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-24 11:27:50.000000000","tz":120},"subject":"Add a new option to enforce the OpenSSL FIPS mode","message":"Add a new option to enforce the OpenSSL FIPS mode\n\nThis option ``ssl_enforce_fips_mode`` allow us to enforce the FIPS mode\nif supported by the version of python in use.\n\nhttps://en.wikipedia.org/wiki/Federal_Information_Processing_Standards\n\nChange-Id: I50c7de71bfd38137eb83d23e910298946507ce9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/af045fc818e9973f7724969bed743078c9c80c0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/af045fc818e9973f7724969bed743078c9c80c0b"}]},"branch":"refs/heads/master"},"ff1914f5f5719127be63c87455039f9b481d6067":{"kind":"REWORK","_number":3,"created":"2021-06-24 13:02:32.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/74/797874/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/74/797874/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/3"}}},"commit":{"parents":[{"commit":"f1b47b6ba8ee4888f6cba1efac0fbf747c5406e6","subject":"Merge \"Add Support For oslo.metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f1b47b6ba8ee4888f6cba1efac0fbf747c5406e6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-24 10:53:55.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-24 13:02:22.000000000","tz":120},"subject":"Add a new option to enforce the OpenSSL FIPS mode","message":"Add a new option to enforce the OpenSSL FIPS mode\n\nThis option ``ssl_enforce_fips_mode`` allow us to enforce the FIPS mode\nif supported by the version of python in use.\n\nhttps://en.wikipedia.org/wiki/Federal_Information_Processing_Standards\n\nChange-Id: I50c7de71bfd38137eb83d23e910298946507ce9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ff1914f5f5719127be63c87455039f9b481d6067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ff1914f5f5719127be63c87455039f9b481d6067"}]},"branch":"refs/heads/master"},"384738a92d1d840bec762086ec10523c52f7497b":{"kind":"REWORK","_number":4,"created":"2021-11-08 14:05:45.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/74/797874/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/74/797874/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/74/797874/4"}}},"commit":{"parents":[{"commit":"f1b47b6ba8ee4888f6cba1efac0fbf747c5406e6","subject":"Merge \"Add Support For oslo.metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f1b47b6ba8ee4888f6cba1efac0fbf747c5406e6"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-06-24 10:53:55.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-11-08 14:05:30.000000000","tz":60},"subject":"Add a new option to enforce the OpenSSL FIPS mode","message":"Add a new option to enforce the OpenSSL FIPS mode\n\nThis option ``ssl_enforce_fips_mode`` allow us to enforce the FIPS mode\nif supported by the version of python in use.\n\nhttps://en.wikipedia.org/wiki/Federal_Information_Processing_Standards\n\nChange-Id: I50c7de71bfd38137eb83d23e910298946507ce9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/384738a92d1d840bec762086ec10523c52f7497b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/384738a92d1d840bec762086ec10523c52f7497b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
