)]}'
{"id":"openstack%2Foslo.messaging~800564","triplet_id":"openstack%2Foslo.messaging~master~Iebc8a96e868d938a5d250bf9d66d20746c63d3d5","project":"openstack/oslo.messaging","branch":"master","hashtags":[],"change_id":"Iebc8a96e868d938a5d250bf9d66d20746c63d3d5","subject":"limit maximum timeout in the poll loop","status":"MERGED","created":"2021-07-12 21:15:55.000000000","updated":"2022-12-05 07:41:00.000000000","submitted":"2021-09-13 16:39:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800564","meta_rev_id":"500f34687c05088ffc442827daf0dc3b6b2cd1aa","_number":800564,"virtual_id_number":800564,"owner":{"_account_id":30834,"name":"Nikita Kalyanov","email":"nikitakalyanov@gmail.com","username":"nikitakalyanov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-13 16:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"recommended":{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},"all":[{"value":1,"date":"2021-07-13 11:58:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},{"value":2,"date":"2021-09-13 15:10:44.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2021-07-15 19:36:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},{"value":2,"date":"2021-09-09 14:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":0,"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},{"value":1,"date":"2021-09-13 15:10:44.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 22:50:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-13 11:58:39.000000000","updated_by":{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},"reviewer":{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},"state":"REVIEWER"},{"updated":"2021-07-15 19:36:31.000000000","updated_by":{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},"reviewer":{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},"state":"REVIEWER"},{"updated":"2021-09-09 14:04:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-10 14:08:31.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"CC"},{"updated":"2021-09-13 15:10:44.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2022-10-21 18:06:30.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"CC"},{"updated":"2022-12-05 07:40:40.000000000","updated_by":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"reviewer":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"state":"REVIEWER"},{"updated":"2022-12-05 07:41:00.000000000","updated_by":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"reviewer":{"_account_id":31662,"name":"Cong Nguyen Thanh","email":"ntcong1705@gmail.com","username":"congnt95"},"state":"CC"}],"messages":[{"id":"826f34a3a347a722976fcc2b044f9b855f56458f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30834,"name":"Nikita Kalyanov","email":"nikitakalyanov@gmail.com","username":"nikitakalyanov"},"date":"2021-07-12 21:15:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be37e78e08a1059b37a53bc2363534524ad10da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 22:50:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75adb568fb024bdc96a18e82d141b3b8 : SUCCESS in 1h 18m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a615eb88179e49be880027283e5b83cb : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8c7d09e6314a3c994297eb1e464da2 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3fb2e6d15f54584bee1524d4bfe71d2 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22e7c47e0a7a41559ac2e4fd295e5093 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05ae5ce0d3e0487cacdd64b534711a0c : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f4894fbcc645899a0403e7a47dca98 : SUCCESS in 4m 56s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/324539795fee47b3b522561e83f45ed0 : SUCCESS in 6m 29s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/27d9366c73744e3c8b33124ce9f62c73 : FAILURE in 7m 53s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/bf97c44df87145f5a415f303780b5b59 : FAILURE in 6m 38s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/7b695a07b3434a8ebb1952e3fb867cd1 : FAILURE in 6m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/daca3b60ba9d46a89c24b42ec99474f6 : SUCCESS in 1h 15m 20s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/8d72c2ad44634feea1e4abf1f0539403 : SUCCESS in 1h 12m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f7d97d81cd6a4719a145c897f69b124c : SUCCESS in 1h 13m 23s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/27a95022408c4d18887bd5f7fad6a1c9 : SUCCESS in 59m 53s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e354b1636b444f8b9be64880a5429649 : SUCCESS in 1h 31m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0f79764a5bdfec37745dc9c5dd70a12e30f2d41","author":{"_account_id":10224,"name":"Feodor Tersin","email":"ftersin@hotmail.com","username":"ftersin"},"date":"2021-07-13 11:58:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0681fcd0eed912a0d1785a94eaf50fe219571019","author":{"_account_id":10234,"name":"Andrey Pavlov","email":"andrey.mp@gmail.com","username":"andrey-mp"},"date":"2021-07-15 19:36:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2b9d52e04c883e90192d8a9682bf33694fe3fb80","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-09 14:04:22.000000000","message":"Patch Set 1: Code-Review+2\n\nHello,\n\nThat seems to make sense.\n\nHere are some thinking about that.\n\nThe original changes [1] define:\n```\nACK_REQUEUE_EVERY_SECONDS_MIN \u003d 0.001\nACK_REQUEUE_EVERY_SECONDS_MAX \u003d 1.0\n```\n\nAt first glance the current timeout is based on `ACK_REQUEUE_EVERY_SECONDS_MIN` so, it will give:\n\n```\n\u003e\u003e\u003e min(0.001*2, 1.0)\n0.002\n\u003e\u003e\u003e max(0.001*2, 1.0)\n1.0\n```\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/440206","accounts_in_message":[],"_revision_number":1},{"id":"d18eeec25147454c62f1b084c0bb3db24d79ef05","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-09-10 14:08:31.000000000","message":"Patch Set 1:\n\nthat is a really great fix!\nI can recall a tone of issues with RPC reply timeouts hitting logs with a weird 5-10 min delays.\nHopefully we\u0027ll face the notorious \"Timed out waiting for a reply\" now less often :)","accounts_in_message":[],"_revision_number":1},{"id":"b11fff45deed84acfb9a20b6cbd3774e65657e0a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-09-13 15:10:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nOof, good catch! The Xena branches have been cut, so we should be safe to merge this. It will need to be backported anyway.","accounts_in_message":[],"_revision_number":1},{"id":"2164684eeceb19b6813d35d884708471f93185f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 15:13:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e2fc6b6f84ae12d3819b2e736932e7bf1f47775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 16:39:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0764547974284b4faaf626ff374ae387 : SUCCESS in 1h 21m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dceacb803874f6dad391e4ff2c5dc52 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c604f618e05419d827f33853ec59ca1 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dcd9cd45dfe4061ad3ff60e8338e6c8 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e550c01be60b40b29ea2de70d1db4188 : SUCCESS in 4m 30s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/0fd0235c7c5a48999bf2bc13a3f28db5 : SUCCESS in 5m 59s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/53473e954ffa4046bb5155a3754f4fac : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9eb2bcb3fbf4313b34c7430618207cdc33bbcd01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 16:39:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9b5f48118749ea22c02cb3a8ae0a83649b8c93d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 16:43:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf210252d4004d0db19b1ad473a25819 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1},{"id":"08d3c712b01d4a071d214a2d31ea6232bf872294","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2022-10-21 18:06:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bdcf915e788bb368774e5462ccc15e6f5b7223d7","revisions":{"bdcf915e788bb368774e5462ccc15e6f5b7223d7":{"kind":"REWORK","_number":1,"created":"2021-07-12 21:15:55.000000000","uploader":{"_account_id":30834,"name":"Nikita Kalyanov","email":"nikitakalyanov@gmail.com","username":"nikitakalyanov"},"ref":"refs/changes/64/800564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/64/800564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/800564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/800564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/800564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/64/800564/1"}}},"commit":{"parents":[{"commit":"9ab3f4f30d4bf0be7dab5f0ca0f0a47177681fdf","subject":"Merge \"Remove the oslo_utils.fnmatch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9ab3f4f30d4bf0be7dab5f0ca0f0a47177681fdf"}]}],"author":{"name":"Nikita Kalyanov","email":"nikitakalyanov@gmail.com","date":"2021-07-12 21:03:31.000000000","tz":180},"committer":{"name":"Nikita Kalyanov","email":"nikitakalyanov@gmail.com","date":"2021-07-12 21:15:38.000000000","tz":180},"subject":"limit maximum timeout in the poll loop","message":"limit maximum timeout in the poll loop\n\nWe should properly limit the maximum timeout with a \u0027min\u0027 to avoid\nlong delays before message processing. Such delays may happen if\nthe connection to a RabbitMQ server is re-established at the same\ntime when the message arrives (see attached bug for more info).\n\nMoreover, this change is in line with the original intent to\nactually have an upper limit on maximum possible timeout (see\ncomments in code and in the original review).\n\nCloses-Bug: #1935864\nChange-Id: Iebc8a96e868d938a5d250bf9d66d20746c63d3d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bdcf915e788bb368774e5462ccc15e6f5b7223d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bdcf915e788bb368774e5462ccc15e6f5b7223d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
