)]}'
{"id":"openstack%2Foslo.messaging~800917","triplet_id":"openstack%2Foslo.messaging~stable%2Ftrain~I46d20ae85ddb5c76119ce90bf06aeb4584bd7607","project":"openstack/oslo.messaging","branch":"stable/train","topic":"missing-whitespace","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2021-11-03 12:45:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I46d20ae85ddb5c76119ce90bf06aeb4584bd7607","subject":"[trivial] fix typos (missing a white space)","status":"MERGED","created":"2021-07-15 11:33:00.000000000","updated":"2021-11-03 12:47:16.000000000","submitted":"2021-11-03 12:45:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800917-missing-whitespace","meta_rev_id":"21223286af3d4adffd66c2b24dd24aa91877949f","_number":800917,"virtual_id_number":800917,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-03 12:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-11-03 11:08:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-11-03 11:08:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 13:48:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-03 11:08:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"e1f570ab91a53e13fb5ca0cab3a71e37567c5a81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-07-15 11:33:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4657c8b4977990213f42adce0848cd332a5ec57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 13:48:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8f4f51124b6d40798077c42623a75cc3 : SUCCESS in 1h 27m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94455458c336491d91cc1ad9d91ab647 : SUCCESS in 1h 20m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfe4d03cf5a64ceb9b7f010734c345b8 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2282dd9abb6d4751a80767e6ae3a354f : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c69d1dad5ac7474f8cd575d05f018275 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2a3c896e31b49cfa3124e15368a13b8 : SUCCESS in 7m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/731cb3cd46a749319c10e689994726aa : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cf74ed2f2654da58cf561107194c669 : SUCCESS in 4m 29s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/3ce739f9ba5b436abf2b086ccea2018b : SUCCESS in 6m 08s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/419f5294ee034eb69ce02f3d0f03d347 : FAILURE in 14m 52s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/d453c9c498fc48dc933b76a57de1f1a5 : SUCCESS in 6m 14s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/067bdedf8e0a4656922d26d1adaeca3a : SUCCESS in 7m 02s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/1ce46efc00a14516bd80ed10c56293ba : FAILURE in 13m 42s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/417ffe4b5a3a46e8989d678f88f56931 : SUCCESS in 7m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d45b2ca2a4e34bffa5b5e415c9a22a68 : SUCCESS in 1h 09m 16s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0bce10013a9a4d03bc205178b8f151aa : SUCCESS in 2h 02m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/0ffe51b31ab64a0e902370e7f2c67f25 : FAILURE in 12m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e816c337b5e84adb90494d1a575692be : SUCCESS in 1h 30m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/bec3dacb641d4fe2abadee06ee854b8b : FAILURE in 12m 44s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f1267185e90d4e60987f344e1988fab0 : SUCCESS in 56m 13s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/d54042713f6647afb56fcdec7a79e10e : FAILURE in 16m 43s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/b73054a9ac094b4c84cbd53aece625c9 : FAILURE in 36m 00s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/8424042fe7aa4ab39c406c7ddce2be49 : FAILURE in 37m 00s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/9cd31c09fece4f4d86a96817590c6c00 : FAILURE in 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d19123b04808f27b8eda0143ae7994d3a609468f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-03 11:08:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f1810dd1ed9e3d1674e6ce9159735b557e4b5a87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 11:09:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d6e6cf6645d962542848f082c9efcf23d78955e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:45:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d2ddbc398e8e4aa9b84f93650d1f4352 : SUCCESS in 1h 27m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dea6a933c8bf4a558b9a167d994d79f1 : SUCCESS in 1h 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf041c81e6884597bd9d810188a7cbe1 : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f53dd1d2c664a6b83f052dda305a9b2 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb371ea19bb4448d9d6e785e10c86360 : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80fde4f7f17e413a99800188ab165f37 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8296b76446284ddb86d7ee43042fbb82 : SUCCESS in 6m 15s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/7cd8786e2d4d4bfb858e2c3c00c75b67 : SUCCESS in 7m 21s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8a00b65316684e9dba2379cd78d74769 : SUCCESS in 1h 35m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c704df19bd9ccb81bfcbeb7bb04d8ccb86b0c0ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:45:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"21223286af3d4adffd66c2b24dd24aa91877949f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 12:47:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba1e8ebe35684a9b803db05d1ec421a1 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4627f14ff8313c9793757984f535824a6e069d12","revisions":{"4627f14ff8313c9793757984f535824a6e069d12":{"kind":"REWORK","_number":1,"created":"2021-07-15 11:33:00.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/17/800917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/17/800917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/800917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/800917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/17/800917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/17/800917/1"}}},"commit":{"parents":[{"commit":"82281a0d4e6abffa93175f30b2cbd7acca3cf9fb","subject":"Correctly handle missing RabbitMQ queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/82281a0d4e6abffa93175f30b2cbd7acca3cf9fb"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-07-15 11:30:43.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-07-15 11:30:43.000000000","tz":120},"subject":"[trivial] fix typos (missing a white space)","message":"[trivial] fix typos (missing a white space)\n\nBad rendering in docs [1][2].\n\n[1] https://docs.openstack.org/oslo.messaging/train/configuration/opts.html#oslo_messaging_rabbit.heartbeat_in_pthread\n[2] https://access.redhat.com/documentation/en-us/red_hat_openstack_platform/16.1/html/configuration_reference/barbican#annotations:295aeb97-f9bd-40bb-8ab5-753b8bd76fab\n\nChange-Id: I46d20ae85ddb5c76119ce90bf06aeb4584bd7607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4627f14ff8313c9793757984f535824a6e069d12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4627f14ff8313c9793757984f535824a6e069d12"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
