)]}'
{"id":"openstack%2Foslo.messaging~803670","triplet_id":"openstack%2Foslo.messaging~master~I4f286cf3af0d578f472b84fe355c812910c7a121","project":"openstack/oslo.messaging","branch":"master","topic":"amqp1-fix-tls-failover","hashtags":[],"change_id":"I4f286cf3af0d578f472b84fe355c812910c7a121","subject":"Revert \"Disable AMQP 1.0 SSL unit tests\"","status":"MERGED","created":"2021-08-05 20:33:06.000000000","updated":"2021-08-20 17:43:35.000000000","submitted":"2021-08-20 17:41:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":464220,"submission_id":"803670-amqp1-fix-tls-failover","meta_rev_id":"a6109af049f77cc42c6f1f6bb1ff938409bd6751","_number":803670,"virtual_id_number":803670,"owner":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-20 17:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-08-20 16:21:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-08-20 17:43:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-08-20 16:21:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-05 20:33:06.000000000","updated_by":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2021-08-05 20:33:06.000000000","updated_by":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"reviewer":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"state":"REVIEWER"},{"updated":"2021-08-05 20:33:06.000000000","updated_by":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2021-08-05 20:33:06.000000000","updated_by":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"reviewer":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"state":"REVIEWER"},{"updated":"2021-08-05 22:11:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-06 09:09:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-20 16:21:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"49aa40dffc0156243a8ff06934448179d5b30845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-08-05 20:33:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb159676a0376e113dad7bf8ec4dac8a22418278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 22:11:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df97ae00c879480c9180414b08729a35 : SUCCESS in 1h 31m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2633246955c4cb3a7d8d6cdb3797401 : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd9d971ba286465b810489292209247d : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70531d8ae23c4067b446a5e07d94f05f : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf05fe3e39f84f23b405e866b3730fc2 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a194c188bdb541f0b9e969b3fd7fe3eb : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4778bd7e5784473b628b17e47c25130 : SUCCESS in 4m 15s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/2eedccd47b674e37804403fa5fadd14e : SUCCESS in 5m 33s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/495b74b489cb4174870cc90ad4d613ae : FAILURE in 6m 28s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/c3888365bf4b407e9f001f96296a9b6e : FAILURE in 5m 20s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/a6dfd321b0f94d4cac10d3bcec29a139 : FAILURE in 5m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fc517be014a94e11894528f4bf6bf44e : SUCCESS in 1h 21m 01s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d5b34299484441cf9b9ae2830ca3c5e5 : SUCCESS in 1h 26m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3d4c821cbdc546c988874610fb0430e7 : SUCCESS in 1h 27m 22s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9fd1b0584ad3422883546e73edd09151 : SUCCESS in 1h 12m 26s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/32bcc50193e84c6bae3688bfd34316dc : SUCCESS in 1h 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9365620bdcd68ce6dc454cda1ade698f5f4833d8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-06 09:09:02.000000000","message":"Patch Set 1: Code-Review+2\n\nThe gate seems happy with that. LGTM","accounts_in_message":[],"_revision_number":1},{"id":"8a779a553bf2da250099ee39af7adfde97fb53e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-08-10 21:04:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"89d8f6cd7d3c0135066043097b67228e94cd46f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 23:57:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d960601d1344d57903e3aaae60a16d7 : SUCCESS in 1h 13m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7082dc04442456f8d45f1ee494db990 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f022dd3b48477cab37968f715c608d : SUCCESS in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3da40524d9d3479fa7632c106c746806 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ae5c6b90a3444359682ef96117624d2 : SUCCESS in 10m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d226d9c9d012441cb83bd1da0ca50560 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d537614710e34ee59c0aa518c14ed9f5 : SUCCESS in 8m 05s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a7e8712b11cb403b94aca865c99f02f1 : SUCCESS in 9m 18s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/50739ff069124e009d9d57bf30e865df : FAILURE in 7m 37s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/00ae2ba2e8bf4782969b05a325ff2fd6 : FAILURE in 6m 24s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/0b671adc53cd410ab693647a979669db : FAILURE in 6m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b8c7eeb0327c4e21bfbac3485db9e0af : SUCCESS in 1h 29m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/489efe6a6c56490cbd5d0ae4943d9e7f : SUCCESS in 1h 25m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a3add53a9a5949c3906bfcd110680c99 : SUCCESS in 1h 31m 34s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/644d31205a24459c9c1ee4791a571979 : SUCCESS in 1h 15m 28s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/71e42a2d60cc49ebb7e93f324a8bfff8 : SUCCESS in 1h 25m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9fa056a5ea5de87f32936fb4bf3ee653956eb423","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-20 16:21:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0622d5c30f32792b2bbf0ecb66ecefbc9dd9075c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 16:22:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84eb263707389cafc6da0fecde7f953c752ea023","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:41:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e4b508f313b43f2b3eacdbaddbcc480 : SUCCESS in 1h 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada78e4cdc6e487c8a78045613b86d32 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10ca118ce02b401d84fde65b9c93f0e5 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f37cb57415a406c9ca5e5145e5a27e7 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14bf4d2ffb3249eab45dbe479cb99203 : SUCCESS in 5m 17s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/0a121f6d5b5e47aebd9db4f973ec6080 : SUCCESS in 6m 24s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a6539a75b6ec4f61af8a8f2d7a0c7ef6 : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a1b1f59cbacfbec364e92d5fc1b8b1f2d3c88835","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:41:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e4bcde3d184a0db41ad8c09a26c96d1add7463e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:43:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40926be76ec43569df55834d600fd86 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc49e04e6df4f3245ae8a313d8ba4a54d48c42a0","revisions":{"365bdb684240554dc2f98a51a3ddb7311c8b5122":{"kind":"REWORK","_number":1,"created":"2021-08-05 20:33:06.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/70/803670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/70/803670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/803670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/803670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/803670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/70/803670/1"}}},"commit":{"parents":[{"commit":"8f5cfda6642ea7f75206d3183c2507e2e83c5693","subject":"Disable AMQP 1.0 SSL unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8f5cfda6642ea7f75206d3183c2507e2e83c5693"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-08-05 20:33:06.000000000","tz":0},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-08-05 20:33:06.000000000","tz":0},"subject":"Revert \"Disable AMQP 1.0 SSL unit tests\"","message":"Revert \"Disable AMQP 1.0 SSL unit tests\"\n\nThis reverts commit 8f5cfda6642ea7f75206d3183c2507e2e83c5693.\n\nReason for revert: This was supposed to be temporary to unblock the gate.  Whatever broke SSL cert generation in the first place appears to be fixed because I can run SSL tests now.\n\nChange-Id: I4f286cf3af0d578f472b84fe355c812910c7a121\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/365bdb684240554dc2f98a51a3ddb7311c8b5122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/365bdb684240554dc2f98a51a3ddb7311c8b5122"}]},"branch":"refs/heads/master"},"fc49e04e6df4f3245ae8a313d8ba4a54d48c42a0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-08-10 21:04:38.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/70/803670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/70/803670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/803670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/803670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/803670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/70/803670/2"}}},"commit":{"parents":[{"commit":"ee5eed86549b695255d7a9f313a59290604e6212","subject":"Merge \"Changed minversion in tox to 3.18.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee5eed86549b695255d7a9f313a59290604e6212"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-08-05 20:33:06.000000000","tz":0},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-08-10 20:55:36.000000000","tz":-240},"subject":"Revert \"Disable AMQP 1.0 SSL unit tests\"","message":"Revert \"Disable AMQP 1.0 SSL unit tests\"\n\nThis reverts commit 8f5cfda6642ea7f75206d3183c2507e2e83c5693.\n\nReason for revert: This was supposed to be temporary to unblock the gate.  Whatever broke SSL cert generation in the first place appears to be fixed because I can run SSL tests now.\n\nChange-Id: I4f286cf3af0d578f472b84fe355c812910c7a121\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fc49e04e6df4f3245ae8a313d8ba4a54d48c42a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fc49e04e6df4f3245ae8a313d8ba4a54d48c42a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
