)]}'
{"id":"openstack%2Foslo.messaging~804141","triplet_id":"openstack%2Foslo.messaging~master~Ia2f977795abc2e81a996e299867e05d41057f33f","project":"openstack/oslo.messaging","branch":"master","topic":"amqp1-fix-tls-failover","hashtags":[],"change_id":"Ia2f977795abc2e81a996e299867e05d41057f33f","subject":"amqp1: re-organize TestFailover to be reused by TestSSL","status":"MERGED","created":"2021-08-10 21:04:38.000000000","updated":"2021-08-20 18:11:02.000000000","submitted":"2021-08-20 18:09:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804141-amqp1-fix-tls-failover","meta_rev_id":"8e7ae5157ee6201d29034a5b0f969ff9e81fef52","_number":804141,"virtual_id_number":804141,"owner":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-20 18:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-08-20 16:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-08-16 11:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-08-20 16:21:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-10 23:52:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 11:30:33.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-20 16:21:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"a215995d7d48bcb1676d6ad7d5936bc1976d169b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-08-10 21:04:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fabfb538cf1ddd16079ce88f890ed4e2d46a9e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 23:52:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37018a6a3a764064af3c0073d95b12fd : SUCCESS in 1h 25m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/973d81c6408445d2a0a7470703baaec7 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb415517d3e4929b918420100754852 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d72c863d4a4044c388e56bde22f30b69 : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df000000cc7745d0afa32d2f9c587a2a : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82811f810e054df49e52c979a4fb0d69 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5098b20bfc2746a1abe2d734ab773e6c : SUCCESS in 5m 32s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f1a21419f320442b9ca60933e4afb3a1 : SUCCESS in 6m 34s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/351a593906334be884c81a2c6a841f80 : FAILURE in 7m 33s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/b9019cef95214711a515ec1492934583 : FAILURE in 6m 26s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/81ed2b25ee2d40c89a7c9c96a1e86580 : FAILURE in 6m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/aa1b2d9986df4fe6a4e6261fe4209834 : SUCCESS in 1h 11m 37s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/77ffafcd2cc740c7806fd334bf73cd2c : SUCCESS in 1h 12m 21s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/040c620df6794850a47b7a1830f635bd : SUCCESS in 1h 17m 53s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/544fc0681233457da8ccad95e77d8c4b : SUCCESS in 53m 56s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/0bc42c9db1bf483c83eb73f51c7807a0 : SUCCESS in 1h 09m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93a43a93b2804c0a30be8cc3b442f4cc35d43cd8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-16 11:30:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67644a2a57a7d2ff6ac3e7f8132d463a89310bb9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-20 16:21:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9040fdcee271e9176e9fb1ce0a69ebb77b23572","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 16:22:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd2f2f2a4dd9b13d712c72d74987785e430d089d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 18:09:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ca86f6a9d044d39afa9fd8fe40a6c6a : SUCCESS in 1h 40m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20677cbbfc3a4a3eadfc618905b354c4 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed083dbaecf944dfb5b48c63a8705b36 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f048ff78c0143ab97ea31f0f071d1d5 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bbccbe964224d9dbd4c926773fd5be5 : SUCCESS in 4m 33s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/948b64b8582842af9d16aa5118f14100 : SUCCESS in 7m 40s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/136370a8e73d4cab87b3a90b44674aaf : SUCCESS in 1h 44m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f69bb0fe62e26d606fdcc9769952a4ba588911e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 18:09:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8e7ae5157ee6201d29034a5b0f969ff9e81fef52","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 18:11:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fa1545abd464dc0b4e12bf965647b62 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01f5b3787494f0402c22c9e1fe78d293154dac6a","revisions":{"01f5b3787494f0402c22c9e1fe78d293154dac6a":{"kind":"REWORK","_number":1,"created":"2021-08-10 21:04:38.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/41/804141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/41/804141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/41/804141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/41/804141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/41/804141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/41/804141/1"}}},"commit":{"parents":[{"commit":"fc49e04e6df4f3245ae8a313d8ba4a54d48c42a0","subject":"Revert \"Disable AMQP 1.0 SSL unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fc49e04e6df4f3245ae8a313d8ba4a54d48c42a0"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-08-10 20:40:52.000000000","tz":-240},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-08-10 20:55:44.000000000","tz":-240},"subject":"amqp1: re-organize TestFailover to be reused by TestSSL","message":"amqp1: re-organize TestFailover to be reused by TestSSL\n\nThis breaks out the generation of brokers and transport_url into\nseparate methods.  These methods are used in the next patch in this\nseries, where TestSSL is updated to inherit from TestFailover, and\nTestSSL overrides the _gen_brokers and _gen_transport_url methods to\nsupply the necessary SSL-aware options.\n\nChange-Id: Ia2f977795abc2e81a996e299867e05d41057f33f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/01f5b3787494f0402c22c9e1fe78d293154dac6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/01f5b3787494f0402c22c9e1fe78d293154dac6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
