)]}'
{"id":"openstack%2Foslo.messaging~807240","triplet_id":"openstack%2Foslo.messaging~stable%2Fstein~I5c468d10230f05438df780e045f1e92662edf070","project":"openstack/oslo.messaging","branch":"stable/stein","topic":"bug/1883038","hashtags":[],"change_id":"I5c468d10230f05438df780e045f1e92662edf070","subject":"Catch ConnectionForced Exception","status":"MERGED","created":"2021-09-03 07:09:58.000000000","updated":"2021-09-08 15:28:28.000000000","submitted":"2021-09-08 15:26:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"807240-bug/1883038","meta_rev_id":"348deb052fb369c2d6c3418fa22b0a1060e6123b","_number":807240,"virtual_id_number":807240,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-08 15:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-09-08 13:45:25.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2021-09-03 16:08:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-03 07:45:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2021-09-08 13:45:25.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-03 07:09:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"state":"REVIEWER"},{"updated":"2021-09-03 08:52:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-03 16:08:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-09-08 13:45:25.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"}],"messages":[{"id":"34fac6f1fedca955adb2c633739659e8bff4592c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 07:09:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4514082a640974edcb0602f1ce690bac9756b4f3","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-09-03 07:45:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f8514a215bba9bd10260109e2503d912db4fb725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 08:52:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/523fbfe8902d4046af59c9134fc76a18 : SUCCESS in 1h 25m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af1bd0b542304379a922b11329fb995c : SUCCESS in 1h 24m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32a67fc8f8f446c88f48b8dbb845bd92 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a181d34d78b4db3bb56b7c742e15747 : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51cf3916361043d3ac0dd3156a876f97 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/567c4a1ac2284fba8bb1f1f4729ba0b3 : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a081cbceb3434439bdf5275fb1b94d96 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5a9750eb4a64bd481d13b2eef824f6b : SUCCESS in 5m 57s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/206734d91cc547daa7b24efbd78a62e4 : SUCCESS in 6m 51s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/dd612a4fe87c44908ed604249891e3ea : SUCCESS in 6m 11s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/634469cd977247f68a664001c4f967c4 : FAILURE in 6m 30s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/f8f3c6e041514a56873aca600d957296 : FAILURE in 5m 34s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/45394e7a84d94dcda9df4114d3933aef : FAILURE in 6m 11s (non-voting)\n- oslo.messaging-tox-py35-func-kafka https://zuul.opendev.org/t/openstack/build/a6e92326457d46599afd86ee455ded5d : FAILURE in 7m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f109eccb97d04a6d9021341a2a64134a : SUCCESS in 1h 39m 46s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/de1217cc399b4228b0d22bae7d45e511 : SUCCESS in 1h 29m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/3b8015b5791543d5af8cc481911dc3a3 : FAILURE in 13m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e6c118fb810d4509a986307b542336d9 : SUCCESS in 1h 33m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/bc94619412264adb8a19b7bc44e80e7a : FAILURE in 20m 47s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3827b44c22454e85bba2c912ad3d9ea4 : SUCCESS in 51m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/31b83fd50b6e43a1b2ebaaf2953d3476 : FAILURE in 22m 02s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/76c1195bad14412c9fd1ed099276c842 : FAILURE in 35m 16s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/34b3fa7239ee4534a72eee51b916ab62 : FAILURE in 37m 47s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/1818fd5f3f9a42cbb86ddeeabf4153d1 : FAILURE in 36m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb2f837aa5032969e1c2d80f6d1f90b74f326ddd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-09-03 16:08:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fc9e5897c500a12a1b51ae30b6d696e29119cc5","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-09-08 13:45:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4590c84b1e82a2aed306022ddea2aab34f12be2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 13:45:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f78d3a9bfc72afd0d8443810a80b37f651600d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 15:26:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/b9ec9b59a94a48af8b59a286f19d1dbf : SUCCESS in 1h 26m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e4a86ab29a6431d97a8a84ab4232b0a : SUCCESS in 1h 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a12c3769f34b4367a19148c2c9c08de9 : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/817b83f09a0845fe9267b6be1ab30c24 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d388bc4851646dfba13e5646f96b509 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cf845bd73d341e9aad6c63fd6a87710 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b0cf19398242a1ace2360c2c887afc : SUCCESS in 4m 32s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/18ed7cd9f30e4c9b9df3590e37d53ed4 : SUCCESS in 6m 02s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1854b56d4b5346a496f305792375233b : SUCCESS in 1h 35m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6725d327052d4572bf5295d0a80025faad19e2e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 15:26:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"348deb052fb369c2d6c3418fa22b0a1060e6123b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 15:28:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46516abfbc5640a5aba674831db91c64 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e05bb37ae2b91df0a8391fa1f3c6d91de24aa9f1","revisions":{"e05bb37ae2b91df0a8391fa1f3c6d91de24aa9f1":{"kind":"REWORK","_number":1,"created":"2021-09-03 07:09:58.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/807240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/40/807240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/807240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/807240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/40/807240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/40/807240/1"}}},"commit":{"parents":[{"commit":"0c47ed4e1b03dd6999a3ae464eb311297a8ac4cd","subject":"Dropping lower constraints testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0c47ed4e1b03dd6999a3ae464eb311297a8ac4cd"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-06-30 00:56:26.000000000","tz":480},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-03 07:07:55.000000000","tz":120},"subject":"Catch ConnectionForced Exception","message":"Catch ConnectionForced Exception\n\nOccasionally I saw large number of connectionforced exceptions.\nsuch exceptions cannot be healed automatically and requires a reboot.\ncatch it and ensure connection established.\n\nChanges:\n  oslo_messaging/_drivers/impl_rabbit.py\n\nNOTE(stephenfin): We need to add the \u0027absolute_import\u0027 future to disable\nthe legacy import flow. Without this, the import of \u0027amqp.exceptions\u0027 in\n\u0027oslo_messaging._drivers.impl_rabbit\u0027 attempts to import \u0027exceptions\u0027\nfrom the sibling module, \u0027oslo_messaging._drivers.amqp\u0027, instead of the\ntop-level \u0027amqp\u0027 module. Thank God for Python 3.x, eh? More information\nis available at [1], for those curious enough.\n\nNOTE(hberaud): conflicts within the cherry-pick on stable/stein due to\nremoved log translation and i18n on stable/train [2]. I decided to ignore this\npatch [2] for stable/stein first to prepare cherry-picking on previous stable\nbranches (rocky, queens) and secondly because I don\u0027t think we want it\non older branches.\n\n[1] https://stackoverflow.com/q/33743880/\n[2]\nhttps://opendev.org/openstack/oslo.messaging/commit/4f385720d7d43aadfaff7c6f41b0534312b23ba8\n\nChange-Id: I5c468d10230f05438df780e045f1e92662edf070\nCloses-Bug: #1883038\n(cherry picked from commit 26e8bb4cc9097fddf3c2e34599b131e3f5d73867)\n(cherry picked from commit 360deb89a7e95d332267564b486219e389a793f7)\n(cherry picked from commit 54ced5361b49fb9b0cf53ff2e6f5825ba7ae513a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e05bb37ae2b91df0a8391fa1f3c6d91de24aa9f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e05bb37ae2b91df0a8391fa1f3c6d91de24aa9f1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
