)]}'
{"id":"openstack%2Foslo.messaging~807693","triplet_id":"openstack%2Foslo.messaging~stable%2Fqueens~I5c468d10230f05438df780e045f1e92662edf070","project":"openstack/oslo.messaging","branch":"stable/queens","topic":"1883038","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-11-30 16:26:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5c468d10230f05438df780e045f1e92662edf070","subject":"Catch ConnectionForced Exception","status":"ABANDONED","created":"2021-09-07 09:23:24.000000000","updated":"2022-11-30 16:26:38.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"42c9355581a33c17cd8c7d041af6580997975f7c","_number":807693,"virtual_id_number":807693,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-07 10:51:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-07 09:23:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"state":"REVIEWER"},{"updated":"2021-09-07 10:51:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"940cd4083f656f9db3aecdcd4d60d50e4d9ddef1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-07 09:23:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa0e9e97874c15cc55643fd0230146bcd4c03ec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 10:51:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a25fc588ed14352a3abee6a55e267ef : FAILURE in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c1add97e74c41ad8ee547f7065574c9 : FAILURE in 5m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a0af86b81ff4c07aea02c9d8c8cab1a : FAILURE in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a215219468440e0a325d077491077c8 : FAILURE in 4m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cfe88acb5c784f799abea03543006614 : SUCCESS in 1h 15m 25s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/f3f841f1a9504a24a434272ddc9ba7aa : FAILURE in 4m 53s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/144c3418865c46998c231790bf5461ef : FAILURE in 4m 53s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/ee4f13694fb947bf8799d6e47230abfc : FAILURE in 4m 33s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/cc646f7540e44c9cbe869564ad288f07 : FAILURE in 4m 45s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/e3bbb4018e704c0291f247db9164a466 : FAILURE in 4m 55s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/968abfc6572847a8b3936b77b83ada12 : FAILURE in 4m 55s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/122933508191417fadfaeac463df22c7 : FAILURE in 4m 19s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/8affa7594ebe42b9b0a8086fd1d59a39 : FAILURE in 4m 52s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/3b4bb963fc034faa8c880178fd13166a : FAILURE in 4m 50s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/a3b9601ec35241e589125215b471fc8d : FAILURE in 4m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/7838772b9a9c4384acd5218c4edd7684 : SUCCESS in 1h 07m 52s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5dccef7e03d84445a78e05b9c48c0b5c : SUCCESS in 1h 11m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/4db2a884fc3242f899147eed376fc32b : FAILURE in 22m 20s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/6c57c44a43b245048804fde8a2d62daa : FAILURE in 21m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/6d4ebd6b5b744dfd9aecb7daf5337484 : FAILURE in 26m 23s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d66be684af3d4e2295884e62c94a4a1b : SUCCESS in 51m 25s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/396001995a6c4b9b925333eca0e6bc7b : SUCCESS in 1h 16m 52s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/1fd76858f7a8478299d84b1bf1ae420b : FAILURE in 32m 22s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/365abda66402406bb69748376be56610 : FAILURE in 31m 45s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/c7d876549991459aa61ba74a49dfb3bd : FAILURE in 27m 05s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/071ecbc3a6054c748530005f41b958ee : SUCCESS in 1h 10m 32s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0e51c12f3e9b4ddea0c63df9ad993a96 : SUCCESS in 1h 22m 30s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/a8f40c15d77441b1badd8b0e5e89a24b : FAILURE in 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28bfec1d9578ba4d64d51d4fe19ca4c3d7ade22e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-14 08:06:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5be3f905ad49dbb5ededed5aa4c4b3236e4cafd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-14 11:23:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c87ae6e2414182a85ced2ad461bbde : FAILURE in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b53661a67cc249259429dd9216310ca3 : FAILURE in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f9c8502c055843d49407c424433fcd0a : FAILURE in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a1533aae4d4faf80b4988a06dc7b79 : FAILURE in 4m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/977d7ac5832241868a589f72f168f5d1 : RETRY_LIMIT in 4m 13s\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/3ad297b93ec34f04b629c7ebbb869c85 : FAILURE in 4m 55s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/efcfc4f578514a9fb01821d59c9fb0ab : FAILURE in 4m 58s (non-voting)\n- oslo.messaging-tox-py27-func-pika https://zuul.opendev.org/t/openstack/build/4af1d812d226484e9e11078b11f21915 : FAILURE in 4m 52s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/6ede90477cc9448c9e110d1948392ce7 : FAILURE in 4m 28s\n- oslo.messaging-tox-py27-func-zmq-proxy https://zuul.opendev.org/t/openstack/build/6e5c429235dd43ca83d1c26b7b814f6e : FAILURE in 4m 11s (non-voting)\n- oslo.messaging-tox-py27-func-zmq-pubsub https://zuul.opendev.org/t/openstack/build/7676aaf8128e499c8d9bd340e0201415 : FAILURE in 4m 02s (non-voting)\n- oslo.messaging-tox-py27-func-zmq https://zuul.opendev.org/t/openstack/build/4aae2ed8dba34a5fb27b8f8951413ab3 : FAILURE in 3m 58s (non-voting)\n- oslo.messaging-tox-py35-func-amqp1 https://zuul.opendev.org/t/openstack/build/f67870c2f4fe4fd19840b67536e5a35b : FAILURE in 4m 09s (non-voting)\n- oslo.messaging-tox-py35-func-rabbit https://zuul.opendev.org/t/openstack/build/876cc6e7a3924b3fb40eb8ff157128bf : FAILURE in 4m 31s (non-voting)\n- oslo.messaging-tox-py35-func-zmq https://zuul.opendev.org/t/openstack/build/61ee79078f404fc6adfa7e95e2d4927c : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit-default https://zuul.opendev.org/t/openstack/build/930b134ddc9a4e878ea8536330bdd720 : FAILURE in 8m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d499658e94384cc9bc6783932d41b67f : FAILURE in 8m 32s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-dual-centos-7 https://zuul.opendev.org/t/openstack/build/14a414b297a94ce99ca9a4f0dfda3b33 : FAILURE in 22m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default-centos-7 https://zuul.opendev.org/t/openstack/build/5ce4bab48c4848d696874449fb13fa15 : FAILURE in 22m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-default https://zuul.opendev.org/t/openstack/build/3610ea71b1714d82b1ba91e234d74d7d : FAILURE in 9m 18s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/32fb9a1eefcd45f0a01a65af51fad57e : FAILURE in 11m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/5de6500394cd4133bf0b41731fceebdb : FAILURE in 13m 38s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/9523463e81c44263a9bbaee337c891da : FAILURE in 9m 18s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/63fa89d40ce94aea814bead458b4ce51 : FAILURE in 10m 43s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/1b5855f26d7d44609502b3fb1d2ef4cf : FAILURE in 9m 31s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-rabbit-default https://zuul.opendev.org/t/openstack/build/de1902240da04fe394f544c46a55addd : FAILURE in 8m 20s\n- oslo.messaging-tempest-neutron-dsvm-src-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ce9ce4899c9b40629a39262b8a74e536 : FAILURE in 8m 14s (non-voting)\n- oslo.messaging-tempest-neutron-dsvm-src-kafka-default https://zuul.opendev.org/t/openstack/build/3ac3570396354a21ad010f2046ddee36 : FAILURE in 7m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42c9355581a33c17cd8c7d041af6580997975f7c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-11-30 16:26:38.000000000","message":"Abandoned\n\nThis branch of this project has transitioned to End of Life, open patches need to be abandoned to be able to delete the branch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0aecec9806a174aa76e190a871bb80eb94b4deea","revisions":{"0aecec9806a174aa76e190a871bb80eb94b4deea":{"kind":"REWORK","_number":1,"created":"2021-09-07 09:23:24.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/93/807693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/93/807693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/807693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/807693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/93/807693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/93/807693/1"}}},"commit":{"parents":[{"commit":"5d7ba77ca63f33d59b76930f284ec717d045c15d","subject":"Remove devstack-plugin-zmq and -pika usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5d7ba77ca63f33d59b76930f284ec717d045c15d"}]}],"author":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2020-06-30 00:56:26.000000000","tz":480},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2021-09-07 09:23:14.000000000","tz":120},"subject":"Catch ConnectionForced Exception","message":"Catch ConnectionForced Exception\n\nOccasionally I saw large number of connectionforced exceptions.\nsuch exceptions cannot be healed automatically and requires a reboot.\ncatch it and ensure connection established.\n\nChanges:\n  oslo_messaging/_drivers/impl_rabbit.py\n\nNOTE(stephenfin): We need to add the \u0027absolute_import\u0027 future to disable\nthe legacy import flow. Without this, the import of \u0027amqp.exceptions\u0027 in\n\u0027oslo_messaging._drivers.impl_rabbit\u0027 attempts to import \u0027exceptions\u0027\nfrom the sibling module, \u0027oslo_messaging._drivers.amqp\u0027, instead of the\ntop-level \u0027amqp\u0027 module. Thank God for Python 3.x, eh? More information\nis available at [1], for those curious enough.\n\nNOTE(hberaud): conflicts within the cherry-pick on stable/stein due to\nremoved log translation and i18n on stable/train [2]. I decided to\nignore this patch [2] for stable/stein first to prepare cherry-picking\non previous stable branches (rocky, queens) and secondly because I\ndon\u0027t think we want it on older branches.\n\n[1] https://stackoverflow.com/q/33743880/\n[2] https://opendev.org/openstack/oslo.messaging/commit/4f385720d7d43aadfaff7c6f41b0534312b23ba8\n\nChange-Id: I5c468d10230f05438df780e045f1e92662edf070\nCloses-Bug: #1883038\n(cherry picked from commit 26e8bb4cc9097fddf3c2e34599b131e3f5d73867)\n(cherry picked from commit 360deb89a7e95d332267564b486219e389a793f7)\n(cherry picked from commit 54ced5361b49fb9b0cf53ff2e6f5825ba7ae513a)\n(cherry picked from commit e05bb37ae2b91df0a8391fa1f3c6d91de24aa9f1)\n(cherry picked from commit 84ff0bbb977139ef557b79d1eb00c8e4eea0aa78)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0aecec9806a174aa76e190a871bb80eb94b4deea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0aecec9806a174aa76e190a871bb80eb94b4deea"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
