)]}'
{"id":"openstack%2Foslo.messaging~808550","triplet_id":"openstack%2Foslo.messaging~master~I556f3f9d615086f1b428c141a82a1c72d7df55b1","project":"openstack/oslo.messaging","branch":"master","topic":"kafka-driver-batch-size","attention_set":{},"removed_from_attention_set":{"33991":{"account":{"_account_id":33991,"name":"Haruki Okada","email":"ocadaruma@gmail.com","username":"ocadaruma"},"last_update":"2024-01-23 10:39:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I556f3f9d615086f1b428c141a82a1c72d7df55b1","subject":"batch_size should be mapped to batch.size librdkafka conf","status":"ABANDONED","created":"2021-09-11 15:36:57.000000000","updated":"2024-01-23 10:39:02.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0f5ff25f557f8e89158d37afdd5d526405ec4c2a","_number":808550,"virtual_id_number":808550,"owner":{"_account_id":33991,"name":"Haruki Okada","email":"ocadaruma@gmail.com","username":"ocadaruma"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-19 01:46:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":1,"date":"2022-02-08 13:48:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},{"value":-2,"date":"2021-10-04 16:59:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2021-10-25 12:40:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2022-02-08 13:48:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-11 17:05:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 13:01:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-15 10:52:32.000000000","updated_by":{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},"reviewer":{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},"state":"REVIEWER"},{"updated":"2021-10-04 16:59:05.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-02-08 13:48:14.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"8ebd1bebb24e4f843462658f51689bce0c105235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33991,"name":"Haruki Okada","email":"ocadaruma@gmail.com","username":"ocadaruma"},"date":"2021-09-11 15:36:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36e22d6bdbd13118cec9ab991ef72310541f3ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-11 17:05:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f93af6c5f254162939e28aa301f40d9 : SUCCESS in 1h 27m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae31e105d9cc44e09f08855c142bb6f7 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c520535c174d388e513f7417f8c3ca : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b4def633a5e4fb1a3a02cc12ebbddb2 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/260e8bec5ab14769a97fc4f11a229f5a : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ef0338c9c234150b8fa4bbe781806ae : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e516cc695d44f13aa172be40d3974cb : SUCCESS in 4m 56s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/3224a6b5795b463d9604345e40144772 : SUCCESS in 6m 23s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/e569e21f859f45f48c41d1f9b91418ad : FAILURE in 7m 36s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/836579e2b6bf4c78b7fb5841bf46074c : FAILURE in 5m 45s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/566bd2ff69a1438ca1f1b79b424e47f5 : FAILURE in 5m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/35bcdf62e82c499390a0ad250261a412 : SUCCESS in 1h 15m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/0f8608e95588412fb66256f20c10c59a : SUCCESS in 1h 17m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e60016408cdc45f39333ae60ea56f39e : SUCCESS in 1h 13m 31s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f7cb14f267134477ae500b5bcbd8ebab : SUCCESS in 1h 05m 22s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c75b2d920cf459da89fef42c1decdaa : SUCCESS in 1h 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4426e7caf56d5d25c1b821a333f805d428fa585","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-13 13:01:39.000000000","message":"Patch Set 1: Code-Review+2\n\nSeems legit","accounts_in_message":[],"_revision_number":1},{"id":"d1eb8573e71aaa1ea07646642eb4338ac9f1f5f0","author":{"_account_id":33991,"name":"Haruki Okada","email":"ocadaruma@gmail.com","username":"ocadaruma"},"date":"2021-09-15 10:39:38.000000000","message":"Patch Set 1:\n\nHi, I\u0027m new to contribute oslo.messaging, so if there\u0027s anything I should do for this patch to be merged, please let me know.","accounts_in_message":[],"_revision_number":1},{"id":"019bbc50cddcb861fe6106032945f85142814e83","author":{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},"date":"2021-09-15 10:52:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aff3c2fb831898f67fea6ee54ad777e66d41942e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33991,"name":"Haruki Okada","email":"ocadaruma@gmail.com","username":"ocadaruma"},"date":"2021-09-19 00:17:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8a8998d07609581e3c9ed4d4dca6d1d7beef8c85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-19 01:46:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e8a56ea28a143d9bff692529ed0e4bc : SUCCESS in 1h 24m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b07092eee124922a069aa811ea97140 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d31a0352e7a492b86e7b48665e51e34 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0aaafe4692f04a488017e0fbea17b438 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c075060aaf2343a0ba044a2a088b573f : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bb88dbe272542aa855b68cb6f4ae9cb : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79e66cef77bf4109912ccb787cf61aed : SUCCESS in 4m 13s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/cbb9ecf5e9264eff894389b4abd07369 : SUCCESS in 5m 30s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/b29e0262bbe34eaf8672896c7efc001e : FAILURE in 7m 11s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/3ac7481ee6854971976337eda18f666c : FAILURE in 5m 07s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/165374ce7600466882812d210a30f74a : FAILURE in 5m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a41270e1496b4de7ae6185ea003b6370 : SUCCESS in 1h 07m 32s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f91f21e060574898a078a96039c02649 : SUCCESS in 1h 11m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e7f6e9139f3e402897664c2a35bfe7cd : SUCCESS in 1h 15m 20s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/db7412c2ed8c42deaef216908ff5d4cb : SUCCESS in 55m 52s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/81967785cfda4ced9ed054d27be5ae34 : SUCCESS in 1h 04m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54b722808a99c59a40d988879c626b7997955d19","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-04 16:59:05.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)\n\nThere isn\u0027t anywhere near enough information about the bug in question here. I also think the fix is wrong. Could you please open a bug and expand the commit message to help everyone understand what you think is wrong and why the fix you\u0027ve suggested will resolve the issue.","accounts_in_message":[],"_revision_number":2},{"id":"043a0bb83245873f4a8a30e0189d1bb03361a0b9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-10-25 12:40:09.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"00533b2663d7876e1137ae7b059472cea3aad3b1","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-02-08 13:48:14.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0f5ff25f557f8e89158d37afdd5d526405ec4c2a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 10:39:02.000000000","message":"Abandoned\n\nI\u0027ll abandon this because of no updates from the author for ~ 2 years after the negative review feedback. Please feel free to reopen this in case you still consider working on this.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"00d132c04c166f93d2853836e2f1c211ffd6b8a7","revisions":{"4f5f9d7199fea2036e52d0429a01778d67bd69f4":{"kind":"REWORK","_number":1,"created":"2021-09-11 15:36:57.000000000","uploader":{"_account_id":33991,"name":"Haruki Okada","email":"ocadaruma@gmail.com","username":"ocadaruma"},"ref":"refs/changes/50/808550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/808550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/808550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/808550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/808550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/808550/1"}}},"commit":{"parents":[{"commit":"f9de265f392dca5386491120593d57b124342c5f","subject":"amqp1: Do not reuse _socket_connection on reconnect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f9de265f392dca5386491120593d57b124342c5f"}]}],"author":{"name":"ocadaruma","email":"ocadaruma@gmail.com","date":"2021-09-11 15:31:44.000000000","tz":540},"committer":{"name":"ocadaruma","email":"ocadaruma@gmail.com","date":"2021-09-11 15:31:44.000000000","tz":540},"subject":"batch_size should be mapped to batch.size librdkafka conf","message":"batch_size should be mapped to batch.size librdkafka conf\n\nChange-Id: I556f3f9d615086f1b428c141a82a1c72d7df55b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f5f9d7199fea2036e52d0429a01778d67bd69f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4f5f9d7199fea2036e52d0429a01778d67bd69f4"}]},"branch":"refs/heads/master"},"00d132c04c166f93d2853836e2f1c211ffd6b8a7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-19 00:17:28.000000000","uploader":{"_account_id":33991,"name":"Haruki Okada","email":"ocadaruma@gmail.com","username":"ocadaruma"},"ref":"refs/changes/50/808550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/50/808550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/808550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/808550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/50/808550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/50/808550/2"}}},"commit":{"parents":[{"commit":"d4f7ea21fc10ec00aaac5ca16bcb5903c3e81bac","subject":"Merge \"use message id cache for RPC listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d4f7ea21fc10ec00aaac5ca16bcb5903c3e81bac"}]}],"author":{"name":"ocadaruma","email":"ocadaruma@gmail.com","date":"2021-09-11 15:31:44.000000000","tz":540},"committer":{"name":"Haruki Okada","email":"ocadaruma@gmail.com","date":"2021-09-19 00:17:28.000000000","tz":0},"subject":"batch_size should be mapped to batch.size librdkafka conf","message":"batch_size should be mapped to batch.size librdkafka conf\n\nChange-Id: I556f3f9d615086f1b428c141a82a1c72d7df55b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/00d132c04c166f93d2853836e2f1c211ffd6b8a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/00d132c04c166f93d2853836e2f1c211ffd6b8a7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
