)]}'
{"id":"openstack%2Foslo.messaging~810509","triplet_id":"openstack%2Foslo.messaging~master~I2e9dbfa27887e26807f199c9d359bacd7c15c67a","project":"openstack/oslo.messaging","branch":"master","topic":"amqp1-heartbeat-pthread","hashtags":[],"change_id":"I2e9dbfa27887e26807f199c9d359bacd7c15c67a","subject":"rabbit: move stdlib_threading bits into _utils","status":"MERGED","created":"2021-09-22 19:05:38.000000000","updated":"2021-10-04 18:31:42.000000000","submitted":"2021-10-04 18:29:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810509-amqp1-heartbeat-pthread","meta_rev_id":"b1fc2141f5a1f43939cb3d6d16589fc724507882","_number":810509,"virtual_id_number":810509,"owner":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-04 18:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-10-04 17:02:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-09-27 08:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-10-04 17:02:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-22 20:34:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-27 08:37:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-10-04 17:02:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"02795812912a03166243331eae5503909b157216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-09-22 19:05:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76a9b748dfed27dc213c17ccc59173c262897102","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 20:34:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfde8680d7044088bf22c07a6b95d9db : SUCCESS in 1h 14m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dac0320ccb9949ab8db256abf1044868 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3057813fe4f54c729c14ae67cb39ab01 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/184fe6fe2930472993c0a8582cf46264 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08c37b68d6c44d3cafd53d6e5e67a602 : SUCCESS in 7m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a71ca7f6343e41859ecbf9ae8376bcaa : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfaf42f364904aa3b29cfc1848db102a : SUCCESS in 5m 33s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/942a2683c79848eb85ebff38e2f32c0c : SUCCESS in 6m 24s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/4cec9ed0c2a647d2affdc39823497248 : FAILURE in 7m 06s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/4ba5459acea640558a72b42ae1e42a5e : FAILURE in 6m 27s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/95ffddb8421840869856c8ef78771e42 : FAILURE in 6m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/607c6b6f57bf46ea9c022570db315911 : SUCCESS in 1h 18m 43s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/48d439b3bbc24d028cde9e46831be62c : SUCCESS in 1h 15m 22s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/83201193a35046e8bf61707bf951feb9 : SUCCESS in 1h 17m 42s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/97986fdac66e4d6294b0c2fb103086b5 : SUCCESS in 1h 26m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/587866f1af164353acda542667e13cfe : SUCCESS in 1h 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3613b6e04912f837747c71f2312d5d4f5283f310","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-27 08:37:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"732d9ae7f34f67b2b27cd3ffd359173bfd20a2a9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-04 17:02:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM and stands on its own feet, even if we don\u0027t merge the next patch in the series (currently WIP)","accounts_in_message":[],"_revision_number":1},{"id":"c8d7c5f4c8ab2c1660fb570967c745fa60eb065b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 17:08:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc64e746151d2f98c842fd5fbbd86d370520e45f","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-10-04 17:53:44.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+2 Workflow+1\n\u003e \n\u003e LGTM and stands on its own feet, even if we don\u0027t merge the next patch in the series (currently WIP)\n\nACK thx for the review, the other patch is working in my testing so I just need to clean it up and get it submitted once I get a break from other fires (hopefully later today or tomorrow)","accounts_in_message":[],"_revision_number":1},{"id":"c1684d2282f76655161bc6ac43c83559226ec078","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 18:29:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40b18984e50a49ed9b70c3f0c8e998ad : SUCCESS in 1h 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7771912cfc364e01987cfcc0f553bc45 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b93e65cde1d545e7b1ee2819007b96a7 : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2781f1d23a7743b5ba28fd873e8abdd9 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ceb4cd1a5d047ab931d784d4e0718b7 : SUCCESS in 5m 56s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/e8f4d57641ee4fc98b199a5726a8c6d0 : SUCCESS in 8m 33s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c99b7f91e36e46528635b735a7d3e030 : SUCCESS in 1h 15m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f0f660569969ddb29d984fdde92dc7dd2f193c38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 18:29:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b1fc2141f5a1f43939cb3d6d16589fc724507882","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 18:31:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ccdd609cc8841b2b6af11d1eded2ef8 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca939fc0e4683efce87b567a9a074063a9c75b4f","revisions":{"ca939fc0e4683efce87b567a9a074063a9c75b4f":{"kind":"REWORK","_number":1,"created":"2021-09-22 19:05:38.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/09/810509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/09/810509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/09/810509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/09/810509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/09/810509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/09/810509/1"}}},"commit":{"parents":[{"commit":"d4f7ea21fc10ec00aaac5ca16bcb5903c3e81bac","subject":"Merge \"use message id cache for RPC listener\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d4f7ea21fc10ec00aaac5ca16bcb5903c3e81bac"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-09-22 18:45:33.000000000","tz":-240},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-09-22 18:45:39.000000000","tz":-240},"subject":"rabbit: move stdlib_threading bits into _utils","message":"rabbit: move stdlib_threading bits into _utils\n\nThe amqp1 driver also needs this same logic, so split it out and share\nit.\n\nChange-Id: I2e9dbfa27887e26807f199c9d359bacd7c15c67a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ca939fc0e4683efce87b567a9a074063a9c75b4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ca939fc0e4683efce87b567a9a074063a9c75b4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
