)]}'
{"id":"openstack%2Foslo.messaging~810510","triplet_id":"openstack%2Foslo.messaging~master~I6df90254c75174e899fb5d3d74fa86fc04e59823","project":"openstack/oslo.messaging","branch":"master","topic":"amqp1-heartbeat-pthread","attention_set":{},"removed_from_attention_set":{"6928":{"account":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"last_update":"2024-02-20 15:00:28.000000000","reason":"Change was abandoned"},"9257":{"account":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"last_update":"2024-02-20 15:00:28.000000000","reason":"Change was abandoned"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-01-16 10:37:18.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}}},"hashtags":[],"change_id":"I6df90254c75174e899fb5d3d74fa86fc04e59823","subject":"amqp1: add option eventloop_in_pthread","status":"ABANDONED","created":"2021-09-22 19:05:38.000000000","updated":"2024-02-20 15:00:28.000000000","total_comment_count":10,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"e7c28ac770c8df650512b5e9866140bbb4d54bfc","_number":810510,"virtual_id_number":810510,"owner":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-26 22:15:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2021-10-28 14:35:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":-1,"date":"2024-02-20 09:43:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":2,"date":"2021-12-15 10:37:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":-1,"date":"2021-11-10 19:52:12.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-22 20:34:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-27 09:04:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-10-28 14:35:04.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2021-12-15 10:37:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-01-12 16:06:33.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"},{"updated":"2024-02-16 14:06:11.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2024-02-20 09:43:27.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"2e8d0bc96aebb16f0b4cb04306578a63a51fe6d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-09-22 19:05:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09cda01216f508adde8384112f65cc9cea052972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 20:34:58.000000000","message":"Patch Set 1: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34f42c0560df4da2963389bf2a24d2bf : SUCCESS in 1h 24m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60f6ddb677424cf89ad369525dfb5819 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc6fbf00717d46cc9887d931a6c8860a : FAILURE in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f17c72c4493a462a85e80edbb12c13df : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1ebe41edd8045f3a537cfd4636494e0 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10de957ea9654d88bfad7e0ddee17f79 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50d1944552164fedb5e725682d50f57b : SUCCESS in 4m 27s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a794416835b94b9484b4233ec8d0e526 : SUCCESS in 5m 48s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/9d10886b63434c47999b90289ef98ea5 : FAILURE in 7m 01s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/49f420537c3e4d82b054ffe0d1d61997 : FAILURE in 5m 22s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/de25dad72202481986e51f48e026be4b : FAILURE in 4m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/27a5738eac6648f4b8dd2afc70185962 : SUCCESS in 1h 11m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c4543c18a136455ab5231849fe83ce67 : SUCCESS in 1h 20m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/786dc28f59d240c4bc3b7bb8b7d64255 : SUCCESS in 1h 13m 08s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/da11eb52773f45bd9b2f8232f58b20bb : SUCCESS in 1h 02m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/0fd0493f168a4154a9ba0fa96fd2ce54 : SUCCESS in 1h 12m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb473092ac8d0ae91f9824081a724996d0491ed7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-09-22 20:53:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"add8fc393dcb30bea4999faf547a593027350261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 22:21:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4ba2196ebf349d4b323b3299e079957 : SUCCESS in 1h 18m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80d06b7dcf4c478eb88ac8943ade4b6a : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef75cb388cc4a3492e7daf020b1504b : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ac7ab209c494287b6076600dc90e597 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c182a496cda94e868cf902a46b981dba : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1064b8b919774962b8c5201a4f99c8dc : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bffc6b9cf9948829b11bc366ee0ccb3 : SUCCESS in 4m 44s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/dec29661df9340ae8aec05cd2633b714 : SUCCESS in 5m 38s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/74eb7e097e0348c9a1e161f95ea8d93b : FAILURE in 7m 30s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/68cd4676af584ccb91d3b49ab34d7fee : FAILURE in 5m 38s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/544d2d779584457a85b5ef3cea92f9d4 : FAILURE in 6m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1976ab7813a24a4099ea242231b9e344 : SUCCESS in 1h 21m 22s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ec5bd43bb3034172996036973ecf24e8 : SUCCESS in 1h 13m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/697800d3fd40441e9f01e063903a41ef : SUCCESS in 1h 22m 05s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b52fc66a76a045c59f8c219b08f7b6a1 : SUCCESS in 56m 07s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/974bd931e2d440ffbde8fbd33b66d9b9 : SUCCESS in 1h 04m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"10745b6fbcd00ae519d44fd2d36a95a053634412","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-27 09:04:18.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nHello John,\n\nThanks for these changes.\nLet me know when they are no longer WIP.","accounts_in_message":[],"_revision_number":2},{"id":"7179c8787931f4915f288d3e2e9a34f7acb39137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-10-26 20:18:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4e8e42e1f2bb41b4e88f58ef78707037ec8fbf19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 22:15:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d365c1bf43844b098b4d78adfe31b129 : SUCCESS in 1h 19m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c694f63f62654a84ba6cf0c3314b1607 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe6766855d44fe5854270845d921a38 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd32c532e87641169f7188752059d66a : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1b2f8dbec541999657a61d2447094e : SUCCESS in 7m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d94d8650978a4a23ae2d395ec8298b19 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36cb270574644dc7b3d7d37a8f6b058a : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a300722abcfc45c69b1e6310aff0fe0a : SUCCESS in 5m 39s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/caa48c6a4bf94cd88f25bdbb351686b8 : SUCCESS in 7m 22s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/25baa42b49fa4b04b8f8c40cc16d3e46 : FAILURE in 8m 05s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/3afcef507e7a4e50b89b3e2bbfb00959 : FAILURE in 6m 30s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/40c00e985c7b4e33bcd543969df6edc5 : FAILURE in 6m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5c3899029eff4fa59857340ba3a298ce : SUCCESS in 1h 24m 27s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1b728ec0315c47fcb39411329957deff : SUCCESS in 1h 18m 13s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/86855b475da140e68f39f88d33556470 : SUCCESS in 1h 20m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8d769701003b4c0294721253e25edea2 : SUCCESS in 59m 32s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/d22738dd311c457ea13d02c9e2c34601 : SUCCESS in 1h 07m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0132456641269d1aeea1b5a94422e4156724e343","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2021-10-28 14:35:04.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"099eadccfd15c5da6b8c0dee6280d3e41a59e79f","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-10-28 14:40:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b39dce25b08116ad398b11a65b1ee5a591ba3033","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-11-10 19:52:12.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0fe1fe67322454e1db19d88ae038f99b1d4806ae","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-12-15 10:37:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a28d4284f53131a412802a6cac50ed8b875c01db","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-12 16:06:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"319bae24e1c1096a56e42389a08ef649937cc511","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2024-01-15 20:38:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b0075ec1ee930939a2aed1d615e9bdcba6167b5","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-16 10:37:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f7ef5b8113432b0745e3e8f50e7edfcfa23228c2","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-02-16 14:06:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c7c3ce01b726ce3482cf83a60d1afb80db9f8e12","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-20 09:43:27.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e7c28ac770c8df650512b5e9866140bbb4d54bfc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2024-02-20 15:00:28.000000000","message":"Abandoned\n\namqp1 driver is deprecated, this patch doesn\u0027t work as-is, and I have no time/priority to fix it.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"53bc87a74c04a509b6624868c586078b50a06e66","revisions":{"c00c4938365d4c536697d70b0dbdf62f63596dea":{"kind":"REWORK","_number":1,"created":"2021-09-22 19:05:38.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/10/810510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/10/810510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/1"}}},"commit":{"parents":[{"commit":"ca939fc0e4683efce87b567a9a074063a9c75b4f","subject":"rabbit: move stdlib_threading bits into _utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ca939fc0e4683efce87b567a9a074063a9c75b4f"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-09-22 19:03:27.000000000","tz":-240},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-09-22 19:03:27.000000000","tz":-240},"subject":"WIP amqp1: add option eventloop_in_pthread","message":"WIP amqp1: add option eventloop_in_pthread\n\nChange-Id: I6df90254c75174e899fb5d3d74fa86fc04e59823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c00c4938365d4c536697d70b0dbdf62f63596dea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c00c4938365d4c536697d70b0dbdf62f63596dea"}]},"branch":"refs/heads/master"},"864b85e807d068e0d92cc9def4af70cf5641785a":{"kind":"REWORK","_number":2,"created":"2021-09-22 20:53:35.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/10/810510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/10/810510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/2"}}},"commit":{"parents":[{"commit":"ca939fc0e4683efce87b567a9a074063a9c75b4f","subject":"rabbit: move stdlib_threading bits into _utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ca939fc0e4683efce87b567a9a074063a9c75b4f"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-09-22 19:03:27.000000000","tz":-240},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-09-22 20:47:32.000000000","tz":-240},"subject":"WIP amqp1: add option eventloop_in_pthread","message":"WIP amqp1: add option eventloop_in_pthread\n\nChange-Id: I6df90254c75174e899fb5d3d74fa86fc04e59823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/864b85e807d068e0d92cc9def4af70cf5641785a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/864b85e807d068e0d92cc9def4af70cf5641785a"}]},"branch":"refs/heads/master"},"53bc87a74c04a509b6624868c586078b50a06e66":{"kind":"REWORK","_number":3,"created":"2021-10-26 20:18:14.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/10/810510/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/10/810510/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/10/810510/3"}}},"commit":{"parents":[{"commit":"ca939fc0e4683efce87b567a9a074063a9c75b4f","subject":"rabbit: move stdlib_threading bits into _utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ca939fc0e4683efce87b567a9a074063a9c75b4f"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-09-22 19:03:27.000000000","tz":-240},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-10-26 20:16:39.000000000","tz":-240},"subject":"amqp1: add option eventloop_in_pthread","message":"amqp1: add option eventloop_in_pthread\n\nAdd a new option `eventloop_in_pthread` to the amqp1 driver.  The\ndefault value is False.  When enabled, this forces the driver\neventloop thread to run in a real pthread, regardless of the\ncontaining execution model.\n\nThis option should be set to False in most circumstances.  At the time\nof this writing, the only common scenario to enable this option is\nwhen running nova-api under mod_wsgi/uWSGI.  This is because:\n\n* nova-api uses eventlet for its execution model\n* nova-api monkey-patches \"thread\"\n\nIn this circumstance, the eventloop runs in a greenthread and can only\nbe scheduled for execution during the processing of an active wsgi\nrequest.  Because heartbeat packets are scheduled and sent by the\neventloop, idle workers will not be able to heartbeat and will thus\nrepeatedly disconnect and reconnect their AMQP connection needlessly.\n\nWith the option enabled, it forces the eventloop into a real pthread\nwhich is scheduled normally by the operating system and is not tied to\nthe lifecycle of in-flight wsgi requests.  This allows heartbeats to\ncontinue processing, even when the wsgi workers are idle.\n\nCloses-Bug: #1941652\nChange-Id: I6df90254c75174e899fb5d3d74fa86fc04e59823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/53bc87a74c04a509b6624868c586078b50a06e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/53bc87a74c04a509b6624868c586078b50a06e66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
