)]}'
{"id":"openstack%2Foslo.messaging~817080","triplet_id":"openstack%2Foslo.messaging~master~Ife62a7d76022908f0dc6a77f1ad607cb2fbd3e8f","project":"openstack/oslo.messaging","branch":"master","topic":"amqp1-reset-race","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2021-12-13 12:00:15.000000000","reason":"stephenfin replied on the change"},"9257":{"account":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"last_update":"2021-12-13 13:25:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ife62a7d76022908f0dc6a77f1ad607cb2fbd3e8f","subject":"amqp1: fix race when reconnecting","status":"MERGED","created":"2021-11-08 20:44:53.000000000","updated":"2021-12-13 13:27:19.000000000","submitted":"2021-12-13 13:25:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"817080-amqp1-reset-race","meta_rev_id":"ebd64d5cbe34fd47e9ad80f92700292935623dd7","_number":817080,"virtual_id_number":817080,"owner":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-13 13:25:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-12-13 12:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-11-15 12:47:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-12-13 12:00:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-12-13 12:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-08 22:11:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-09 13:08:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-11-15 12:47:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-12-10 14:32:19.000000000","updated_by":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"9082ae23d957ebd2a1e6f485f44f68d9f29a2e2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-11-08 20:44:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7730c128957950b4d13bde99a0565bb1d5c1fb77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 22:11:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee6fa42848244038ac39686fd10d64d7 : SUCCESS in 1h 20m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59ce4560fcff4866a33246af1bf6ecf5 : TIMED_OUT in 51m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5513a459540416b8c7a3858e4001717 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07ce8a6e5ae74dd5bf9136d2cbc55fa5 : TIMED_OUT in 41m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e020f0201694d75847d42631b8b13bc : SUCCESS in 7m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/824dc1c2f694451db9926cfe0e023ad4 : TIMED_OUT in 41m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083d0fc17cb7491581676de00cb33757 : SUCCESS in 5m 10s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/b9dfd2e0f3c34ce68ae9fcf5ab25dbca : SUCCESS in 6m 12s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/02e8987fd5b5480cb4543f3f13be6421 : FAILURE in 7m 19s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/4a73381ec43a4d9d8679b0653a95f272 : FAILURE in 5m 57s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/8c4d33a9e98144b3ab887817d98c26f8 : FAILURE in 6m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c5c767bf4b0348739f323616a33e76c4 : SUCCESS in 1h 13m 23s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/90419858760349f0b9af4a74cfbc2442 : SUCCESS in 1h 11m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f1c56ad81a684dcfa4bc92ace84c1721 : SUCCESS in 1h 18m 40s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/557267ea94894c1791c7cbe5d0b7faca : SUCCESS in 52m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f16df406c8744bf82bd68877486a0ff : SUCCESS in 1h 03m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4f795705003211808d6379a91525b28d9d9c57a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-09 13:08:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c99e6a87ce287b501622a7029b21876f09c4a9f6","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-11-09 14:21:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1060eaee0e423e8b25e5850cd99383a83ecf41d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 14:44:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d41288da71a43d79a505f462b8dc6ed : SUCCESS in 1h 30m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4298bb95a2744831a9acd7c09b3cb577 : TIMED_OUT in 51m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dced182c7f440e981d3d99f801074f9 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95a36fb50c954dd68111a0c6e5a0c4d6 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/126ed621982a4bd8af47f18988455d4d : TIMED_OUT in 41m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b311bd466a334777955d880eca46615d : TIMED_OUT in 41m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2b63a6c01e1444b9919823a48a49b5b : SUCCESS in 5m 00s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a279e9172c3e489189a4606018b5fe37 : SUCCESS in 6m 41s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/13e50be1772046d5a358a25d4f35c1e9 : FAILURE in 8m 18s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/64cecd35771e4848a02cb11917b6b785 : FAILURE in 6m 45s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/1c78ba57c9964a79bd2a6d10e64b0d99 : FAILURE in 6m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/663ae3af547d4f3e82bb531e82d20ce3 : SUCCESS in 1h 10m 31s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2aacd53b066040c0a46f69c4d98c6442 : SUCCESS in 1h 22m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/97570dfd70634fe290ff625e80dd13e6 : SUCCESS in 1h 21m 49s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5369ea44aa44491cb69ad7dffc0856cd : SUCCESS in 1h 15m 06s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/699338c9cd2340efbed30261a3204533 : SUCCESS in 1h 02m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad027fd7326b4e1b2737bb4db7ece089e769ee17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-11-09 21:00:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c1ba65fc175ecea7ec8256d87f2eceb07faef6d","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-11-09 21:04:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"301c0f022abba05dc4a3244da0f55bfb108c9773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 22:53:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a09e20d0f0e04debbf3a822827e2b7ab : FAILURE in 1h 50m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f2c51df06404815a8171b7596cb876c : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0abeb178f79040bb8ae655a92dd096aa : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/828b6597d48049ceae6514a963104a0d : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82a6346bb2bd4e35a1e86790ba512286 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ec060d79e164a7d9ebf1d0c04b51fea : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36ce5663c79c4954ad126aac68456000 : SUCCESS in 4m 34s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/74b7731418ce4193943f24bec02e92df : SUCCESS in 6m 09s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/bbabcfc519854f0e8793e43fea793745 : FAILURE in 6m 41s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/cca061f6a9e74baeb9e9e0d4a7d3af13 : FAILURE in 5m 51s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/f29bcb439b5c493499f5844c8b79bb85 : FAILURE in 5m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/12077e183c474b1cb13d8c7f2d989888 : SUCCESS in 1h 23m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9a43e7538e214b2db42bbf7eb42d805f : SUCCESS in 1h 20m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/77c86b301644412bb57d623bc3e7abee : SUCCESS in 1h 19m 49s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d295aee4bab7499aafbe322ae7519511 : SUCCESS in 1h 05m 36s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/44520784bd084dd9af89eeb7b0512120 : SUCCESS in 1h 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e81e8413641a1b3c786b3b69fa484a3e6e86534","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-11-10 14:53:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cc99eba2f0b434df12a33821bdb4f0cc890b7792","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-11-10 15:44:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b6713805a01f67c3d8536b8b8fd07109a184e2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 17:38:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f759eb8a96eb487e8d9d89f8ae51539f : SUCCESS in 1h 23m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cca33ea3d1da452099c83c29284fad45 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e095ea22fa649ecaa4f2ba1ca239bc6 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f57ca7d7a6748838b626e750825e040 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9a6fbece9694169985cbfee4a53f857 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d15916e78a3642e0adb8621478563107 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3184ae34fa49d1b7287e69866b29fc : SUCCESS in 6m 03s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/1cb1ee3c1c5248d1995134927ca84fd7 : SUCCESS in 7m 56s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/d7eb8c2e5e8b4865a94a47d0be0c0074 : FAILURE in 9m 45s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/1098724de64643e2960a595a2143b5b6 : FAILURE in 8m 22s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/2ecf9ad5334e462ab43df4c1f413b6bc : FAILURE in 6m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3cabdf1ff76d42a69176ea84fd5651ed : SUCCESS in 1h 29m 31s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4fe244f222cf4f7c9a2269e2ef182f0a : SUCCESS in 1h 29m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/cac670e767ae43258c368a34d0d71dc5 : SUCCESS in 1h 41m 48s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/8a3cad176a374a838e4d1c1583085959 : SUCCESS in 1h 04m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e34427bcd3e465cbd97b72b1d32a90a : SUCCESS in 1h 20m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"272f48f79b6ab6fa7dc9523a88f981118138208f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-15 12:47:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"969f9853156c8fbf789f3d057c413b1cb59b6a57","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-12-10 14:32:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21ad73604ad411bb063e5dae64aab945e0748b32","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-13 12:00:15.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a748cd4bb2ce9264278c1ed42b4b50960af9cbdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 12:00:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"131058949e0cd7db8dccc951e69f9054c224bc82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 13:25:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6aec2ff43444fb1b0866604b25b4c20 : SUCCESS in 1h 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e583378e6a7947e98cce5cf75deefc20 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eccca5ddf7134c81884ed3fbb5cd4b24 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5eccd4774044b2cbf9602bf789fe21a : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09d8096cb5f4c1483a4a9153bf8e97a : SUCCESS in 3m 56s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/3dfa7a19e1be46c8a1705a92102ed1cf : SUCCESS in 5m 22s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/77ebcad6fdf24b36b8419f65a1b0ecf3 : SUCCESS in 1h 20m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6b97a56d1a21c23fa1a36cb3bfac0c7b9bee4aa3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 13:25:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ebd64d5cbe34fd47e9ad80f92700292935623dd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 13:27:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1cd2fee1bb41fbae7a06f32f435981 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02a38f507d8f0c377a2ef468e3497b5e897f1b09","revisions":{"3128a5b2de62299664939bba8f59ca3b484e8c1f":{"kind":"REWORK","_number":1,"created":"2021-11-08 20:44:53.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/80/817080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/817080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/817080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/817080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/817080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/817080/1"}}},"commit":{"parents":[{"commit":"feb72de7b81e3919dedc697f9fb5484a92f85ad8","subject":"Merge \"Remove deprecation of heartbeat_in_pthread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/feb72de7b81e3919dedc697f9fb5484a92f85ad8"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-11-08 20:19:45.000000000","tz":-300},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-11-08 20:19:45.000000000","tz":-300},"subject":"amqp1: fix race when reconnecting","message":"amqp1: fix race when reconnecting\n\nCurrently this is how reconnect works:\n\n- pyngus detects failure and invokes callback\n  Controller.connection_failed() which in turn calls\n  Controller._handle_connection_loss()\n\n- The first thing that _handle_connection_loss does is to set\n  self.addresser to None (important later)\n\n- Then it defers _do_reconnect after a delay (normally 1 second)\n\n- (1 second passes)\n\n- _do_reconnect calls _hard_reset which resets the controller state\n\nHowever, there is a race here.  This can happen:\n\n- The above, up until it defers and waits for 1 second\n\n- Controller.send() is invoked on a task\n\n- A new Sender is created, and critically because self.reply_link\n  still exists and is active, we call sender.attach and pass in\n  self.addresser.  Remember _handle_connection_loss sets\n  self.addresser to None.\n\n- Eventually Sender.attach throws an AttributeError because it\n  attempts to call addresser.resolve() but addresser is None\n\nThe reason this happens is because although the connection is dead,\nthe controller state is still half-alive because _hard_reset hasn\u0027t\nbeen called yet since it\u0027s deferred one second in _do_reconnect.\n\nThe fix here is to move _hard_reset out of _do_reconnect and directly\ninto _handle_connection_loss.  The eventloop is woken up immediately\nto process _hard_reset but _do_reconnect is still deferred as before\nso as to retain the desired reconnect backoff behavior.\n\nCloses-Bug: #1941652\nChange-Id: Ife62a7d76022908f0dc6a77f1ad607cb2fbd3e8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3128a5b2de62299664939bba8f59ca3b484e8c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3128a5b2de62299664939bba8f59ca3b484e8c1f"}]},"branch":"refs/heads/master"},"02a38f507d8f0c377a2ef468e3497b5e897f1b09":{"kind":"REWORK","_number":2,"created":"2021-11-09 21:00:26.000000000","uploader":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"ref":"refs/changes/80/817080/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/80/817080/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/817080/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/817080/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/80/817080/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/80/817080/2"}}},"commit":{"parents":[{"commit":"feb72de7b81e3919dedc697f9fb5484a92f85ad8","subject":"Merge \"Remove deprecation of heartbeat_in_pthread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/feb72de7b81e3919dedc697f9fb5484a92f85ad8"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-11-08 20:19:45.000000000","tz":-300},"committer":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-11-09 20:59:54.000000000","tz":-300},"subject":"amqp1: fix race when reconnecting","message":"amqp1: fix race when reconnecting\n\nCurrently this is how reconnect works:\n\n- pyngus detects failure and invokes callback\n  Controller.connection_failed() which in turn calls\n  Controller._handle_connection_loss()\n\n- The first thing that _handle_connection_loss does is to set\n  self.addresser to None (important later)\n\n- Then it defers _do_reconnect after a delay (normally 1 second)\n\n- (1 second passes)\n\n- _do_reconnect calls _hard_reset which resets the controller state\n\nHowever, there is a race here.  This can happen:\n\n- The above, up until it defers and waits for 1 second\n\n- Controller.send() is invoked on a task\n\n- A new Sender is created, and critically because self.reply_link\n  still exists and is active, we call sender.attach and pass in\n  self.addresser.  Remember _handle_connection_loss sets\n  self.addresser to None.\n\n- Eventually Sender.attach throws an AttributeError because it\n  attempts to call addresser.resolve() but addresser is None\n\nThe reason this happens is because although the connection is dead,\nthe controller state is still half-alive because _hard_reset hasn\u0027t\nbeen called yet since it\u0027s deferred one second in _do_reconnect.\n\nThe fix here is to move _hard_reset out of _do_reconnect and directly\ninto _handle_connection_loss.  The eventloop is woken up immediately\nto process _hard_reset but _do_reconnect is still deferred as before\nso as to retain the desired reconnect backoff behavior.\n\nCloses-Bug: #1941652\nChange-Id: Ife62a7d76022908f0dc6a77f1ad607cb2fbd3e8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/02a38f507d8f0c377a2ef468e3497b5e897f1b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/02a38f507d8f0c377a2ef468e3497b5e897f1b09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
