)]}'
{"id":"openstack%2Foslo.messaging~821764","triplet_id":"openstack%2Foslo.messaging~stable%2Fxena~Ife62a7d76022908f0dc6a77f1ad607cb2fbd3e8f","project":"openstack/oslo.messaging","branch":"stable/xena","topic":"amqp1-reset-race-stable/xena","attention_set":{},"removed_from_attention_set":{"9257":{"account":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"last_update":"2021-12-15 13:33:39.000000000","reason":"John Eckersberg replied on the change"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-04-08 14:44:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ife62a7d76022908f0dc6a77f1ad607cb2fbd3e8f","subject":"amqp1: fix race when reconnecting","status":"MERGED","created":"2021-12-15 10:25:42.000000000","updated":"2022-04-08 14:46:06.000000000","submitted":"2022-04-08 14:44:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821764-amqp1-reset-race-stable/xena","cherry_pick_of_change":817080,"cherry_pick_of_patch_set":2,"meta_rev_id":"fe1cf224b3f327425da5a4989bf9320a9cf14eff","_number":821764,"virtual_id_number":821764,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-08 14:44:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"recommended":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"all":[{"value":1,"date":"2021-12-15 13:33:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-08 13:46:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-08 13:26:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-15 10:25:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"state":"REVIEWER"},{"updated":"2021-12-15 11:57:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-08 13:46:31.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"08c066f74f6547f9dc49adebd7f4899fd5a40301","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-12-15 10:25:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d3345877e272d69b9f57dacc1729f7e7b6249d5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 11:57:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af925f2ac42640259b396d7ebd21acdb : SUCCESS in 1h 24m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09ec3090a31e404ea69697a50b409bb0 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e406803c56e481bb4ab2af5b9e79bd4 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a68f0d11a4334ecbb47473c2ceec0e16 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2164c4afe879472d8aac6b73c021b782 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fad32a0a71848bd9b171acd67e05290 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa0fc14d0ef4d92818333ec725f49e0 : SUCCESS in 6m 46s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/e63a91ae97164dfd8159bdb83bef24da : SUCCESS in 6m 31s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/4338a3c5d4614733b9c1c8d7eb9042f2 : FAILURE in 11m 03s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/ad8a381a256544de8df555f021336dd1 : FAILURE in 5m 47s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/5c1979abfd584d27815932a8e02db63d : FAILURE in 10m 12s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/93ebc80721614fa88fdf7238410de6d3 : SUCCESS in 1h 20m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e3145850381140bcb9cb42d02d4bfb14 : SUCCESS in 1h 21m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9057e9bb916146f3a7f5a6a2200009b2 : SUCCESS in 1h 25m 31s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/61f09f9ea691459ebe3ec53d6873ad62 : FAILURE in 1h 06m 01s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a2d93626b1b440783f5162a1c47d8e4 : SUCCESS in 1h 13m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6df31247464b1e250f8bcaaf2ff3963ce2f3f50","author":{"_account_id":9257,"name":"John Eckersberg","email":"jeckersb@redhat.com","username":"jeckersb"},"date":"2021-12-15 13:33:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6bf4f6eb8ea7164b1b4713936e9069e1c57654c9","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-02-08 13:46:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08d6dd7b93a653e21505014ac51c0b40673cd3f7","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-08 13:26:34.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc60c353d46106b71432b829a8546a6e7cf89f6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:27:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"276320f7e010d9aeeb2ac863f1725ecc6df3602a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 14:44:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb03914ab8474b3c98212b15a0891c94 : SUCCESS in 1h 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c567afc55cd94fc982dd1e24a1393b2b : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01d2005d5332423b869c5cffbcc363f1 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41a551d5ccec44108701327a2b7696bb : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d03b006de22d4664b1d017c4e93902c0 : SUCCESS in 4m 23s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/003ccad42cde4a5b84e564d39db28236 : SUCCESS in 4m 52s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/99544eef8acb498781ef43475f170780 : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d5adb3f6a013b2e70deb67eb6765bd0bf8fe18cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 14:44:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fe1cf224b3f327425da5a4989bf9320a9cf14eff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 14:46:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285f0cf5375b48dc8074e92206c775dd : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d57eccd862c1a008f9b1f868c0832e938a75415e","revisions":{"d57eccd862c1a008f9b1f868c0832e938a75415e":{"kind":"REWORK","_number":1,"created":"2021-12-15 10:25:42.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/64/821764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/64/821764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/821764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/821764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/821764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/64/821764/1"}}},"commit":{"parents":[{"commit":"b657365568ff84d4bb454fc7958ccb8ff3239365","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b657365568ff84d4bb454fc7958ccb8ff3239365"}]}],"author":{"name":"John Eckersberg","email":"jeckersb@redhat.com","date":"2021-11-08 20:19:45.000000000","tz":-300},"committer":{"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","date":"2021-12-15 10:25:42.000000000","tz":0},"subject":"amqp1: fix race when reconnecting","message":"amqp1: fix race when reconnecting\n\nCurrently this is how reconnect works:\n\n- pyngus detects failure and invokes callback\n  Controller.connection_failed() which in turn calls\n  Controller._handle_connection_loss()\n\n- The first thing that _handle_connection_loss does is to set\n  self.addresser to None (important later)\n\n- Then it defers _do_reconnect after a delay (normally 1 second)\n\n- (1 second passes)\n\n- _do_reconnect calls _hard_reset which resets the controller state\n\nHowever, there is a race here.  This can happen:\n\n- The above, up until it defers and waits for 1 second\n\n- Controller.send() is invoked on a task\n\n- A new Sender is created, and critically because self.reply_link\n  still exists and is active, we call sender.attach and pass in\n  self.addresser.  Remember _handle_connection_loss sets\n  self.addresser to None.\n\n- Eventually Sender.attach throws an AttributeError because it\n  attempts to call addresser.resolve() but addresser is None\n\nThe reason this happens is because although the connection is dead,\nthe controller state is still half-alive because _hard_reset hasn\u0027t\nbeen called yet since it\u0027s deferred one second in _do_reconnect.\n\nThe fix here is to move _hard_reset out of _do_reconnect and directly\ninto _handle_connection_loss.  The eventloop is woken up immediately\nto process _hard_reset but _do_reconnect is still deferred as before\nso as to retain the desired reconnect backoff behavior.\n\nCloses-Bug: #1941652\nChange-Id: Ife62a7d76022908f0dc6a77f1ad607cb2fbd3e8f\n(cherry picked from commit 02a38f507d8f0c377a2ef468e3497b5e897f1b09)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d57eccd862c1a008f9b1f868c0832e938a75415e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d57eccd862c1a008f9b1f868c0832e938a75415e"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
