)]}'
{"id":"openstack%2Foslo.messaging~822407","triplet_id":"openstack%2Foslo.messaging~master~I26743858a0ca7a6e46bda821c8f29b6dff34ea15","project":"openstack/oslo.messaging","branch":"master","attention_set":{},"removed_from_attention_set":{"31827":{"account":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"last_update":"2021-12-21 12:25:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I26743858a0ca7a6e46bda821c8f29b6dff34ea15","subject":"Update python testing classifier","status":"MERGED","created":"2021-12-21 08:55:32.000000000","updated":"2021-12-21 12:27:13.000000000","submitted":"2021-12-21 12:25:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822407","meta_rev_id":"e2e28310d141f96579a698605dc49f7c45b43142","_number":822407,"virtual_id_number":822407,"owner":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-21 12:25:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-12-21 10:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-12-21 10:59:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-21 10:30:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-21 10:59:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"a44ced2bb615a33eed1afe5cf59a8e81d5dd4a80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"date":"2021-12-21 08:55:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c32728785b50c03f0800a101a2dcadd3e31faa70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 10:30:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/17754b62d67d467f8518b6db4bd8b87b : SUCCESS in 3m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48ef798c745e44399cd593e8a938f86b : SUCCESS in 1h 09m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20e307c871584186b73f33f71ff9804f : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6634ee15ba74fee816c1b7040a6ad5a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de3710706ea547c08ff963bf226c7c6a : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce85f6b2143f4be4a55fd6ec3775c5bf : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a020be99c4b0433a9fb0826e09c0dc69 : SUCCESS in 4m 37s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a0c27d51b0c44842a807aa09dbb6d3bb : SUCCESS in 7m 57s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/be643143920f4679a2d69f41655e777c : FAILURE in 6m 00s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/8fadba004c1845bda27b4f76c9ef0892 : FAILURE in 5m 10s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/52f0f0d6ee2842afbdeee4f029796e99 : FAILURE in 4m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3d80fe1ee83c4b1fa61c869409d4ca2a : SUCCESS in 1h 29m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a5edb3d1d75246928f2741435f773124 : SUCCESS in 1h 20m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/45eba090f468483b8bd34bff9fb6bd70 : SUCCESS in 1h 17m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1c4603ac4deff8d3715d41cdd414e2185c2dc3e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-21 10:59:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfc28c9fb87c7dd3a5c55ed5d1bcb954659029ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 11:00:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"decef4f30bd353b8877442eeb3f43353ded20a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:25:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/85248db036d6466fbb04b7b485ab2720 : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6268d0f8a464a2eb40d410603b8f602 : SUCCESS in 1h 23m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6734199522e34ecf9ec57d71104337f7 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/180d10f559304de79acd33e83042f688 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10514dbf18cc4b2c8a1563d7f690c064 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01eec31c704b4faa89835ec66de93a2a : SUCCESS in 7m 15s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f81dff946ebd49e39c4f195ddd089a1f : SUCCESS in 6m 34s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/31cfcd4cdc57461ab3a7df14d10c6129 : SUCCESS in 1h 14m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5ddd493619e6cc517713b8b84b2e73ab0c88f48b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:25:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e2e28310d141f96579a698605dc49f7c45b43142","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:27:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52fefc48b63f44dab88d4ec98e429a72 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76182ec7e7fc430112806a4e13b786ce6a677bc9","revisions":{"76182ec7e7fc430112806a4e13b786ce6a677bc9":{"kind":"REWORK","_number":1,"created":"2021-12-21 08:55:32.000000000","uploader":{"_account_id":31827,"name":"Jorhson Deng","email":"jorhson_deng@163.com","username":"dengzhaosen"},"ref":"refs/changes/07/822407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/07/822407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/822407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/822407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/07/822407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/07/822407/1"}}},"commit":{"parents":[{"commit":"02a38f507d8f0c377a2ef468e3497b5e897f1b09","subject":"amqp1: fix race when reconnecting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/02a38f507d8f0c377a2ef468e3497b5e897f1b09"}]}],"author":{"name":"dengzhaosen","email":"dengzhaosen@inspur.com","date":"2021-12-21 08:53:13.000000000","tz":480},"committer":{"name":"dengzhaosen","email":"dengzhaosen@inspur.com","date":"2021-12-21 08:53:31.000000000","tz":480},"subject":"Update python testing classifier","message":"Update python testing classifier\n\nYoga testing runtime[1] has been updated to add py39\ntesting as voting. Unit tests update are handled by the\njob template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/820286\n\nthis commit updates the classifier in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: I26743858a0ca7a6e46bda821c8f29b6dff34ea15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/76182ec7e7fc430112806a4e13b786ce6a677bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/76182ec7e7fc430112806a4e13b786ce6a677bc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
