)]}'
{"id":"openstack%2Foslo.messaging~836679","triplet_id":"openstack%2Foslo.messaging~master~I41538fd85956780250e3b7440130381151991b9c","project":"openstack/oslo.messaging","branch":"master","topic":"fix-gate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-01-23 13:23:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I41538fd85956780250e3b7440130381151991b9c","subject":"Bump bandit","status":"MERGED","created":"2022-04-05 16:26:27.000000000","updated":"2024-01-23 13:24:18.000000000","submitted":"2024-01-23 13:23:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836679-fix-gate","meta_rev_id":"2222790a0d5fcbddd6621d77b823f5bf551b9e54","_number":836679,"virtual_id_number":836679,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-23 13:23:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-01-23 11:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 11:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-01-23 11:54:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 18:11:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 07:19:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-04-27 13:35:18.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"},{"updated":"2024-01-23 11:50:49.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-01-23 11:54:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"973a9a64044ee2c0fbd1998a5c2a4ce5d37f940a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-04-05 16:26:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fc51fc4fe849d87a97ba42105f87d7d43d46319","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 18:11:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/916aa9030a534a6c95401151df9c180d : SUCCESS in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e3801fd8a614affa9d2453a8db0f2c1 : SUCCESS in 1h 22m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4523faad18514b499d8f3802888fe453 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3911c23b9f784dc297ec7e23dcbebb52 : FAILURE in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0169a6465883473ab3a6f28de26388ca : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67f35d8395804f33a6598d68f353e3f4 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb35b37adb4843a6a03c0563e8b4ce0a : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f140fab86bd45b38973f7a619dd77a7 : SUCCESS in 3m 57s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f35dc78fb255470aa3f31a6405b16ad3 : SUCCESS in 5m 48s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/b77099c1fdbb47eb9262fbdf43b0837f : FAILURE in 7m 04s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/5ee76ac2cf4649deb7eecf8ff6c7db8e : FAILURE in 5m 30s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/90c9dfa1138441978a060a316294613f : FAILURE in 5m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a18077da34394b358e12ebeb3e19d7c3 : SUCCESS in 1h 27m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ce0a5424b4f0436d99aff3738f2aa1b6 : SUCCESS in 1h 30m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/da7dd2ffb8f74018932d9d85111828ac : SUCCESS in 1h 36m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"491b617606137757dc70b163e332bb3f8063a23b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-06 07:19:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42c7ac7112a934bff595ed709dea7843636e3100","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-27 13:35:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7f7ed33439e541ee069f465e260c7db3f985398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:03:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1bd78ff3892c46fe8fa1ed152aa2d420 : SUCCESS in 4m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53adc8a95eaf469dad96b6624e98f2be : SUCCESS in 1h 22m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcaf06609d27405e98ac7b39b8968b06 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562e4b21dbed48b0a3d489a51bc6a595 : FAILURE in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c031d5ea32724a14ae6fc376151d4bd9 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50013729b5b64558b41b745a3e6e0b97 : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77f0b5f8f0f04ea0a8df8ab0e0a3a14b : FAILURE in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/149edde28dc2457da99f4a8fe48cd26f : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26b4ff6974ad474ebd0fa142639833f3 : SUCCESS in 4m 37s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/bb852596c4024b61a55f95a8954a096e : SUCCESS in 6m 54s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/bfe6dbf6cee14b839d3d1b27de2341af : FAILURE in 9m 12s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/24064d0882bb4aba9c276c1d9c1d9667 : FAILURE in 6m 48s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/b4bc2a1f7d7248bf90b52edf7478f4b7 : FAILURE in 6m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6d28a0b7071945b7b40ff9141db96202 : SUCCESS in 1h 21m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3ebdc9f222054f48878ec3ad520579f3 : SUCCESS in 1h 21m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9fd8bab3667f42ddb48ce88f6b982a6d : SUCCESS in 1h 19m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7714f8b74e2bfde105a186746283549425e2ec49","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-08-25 07:11:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e5c1f7842141aed4c92bdf76f064a8016b608f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 08:57:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6242bf9b75d24b5c99c2e329272a6ef8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0581fe3c2edb46dc8e5315c103004a76 : SUCCESS in 3m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1acf0fb1844b4296a84d7292fd3b7043 : SUCCESS in 1h 25m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5caefb8ea3204d1e8e20d6510744e638 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/968a11bcad9b464d8b1375256c1662c9 : FAILURE in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f66b0fed24a49d0bab1cec271f492e4 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62d8e0c5f2804d3f96856bd0cf1c1e96 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/912e32a8b6464d5eb4f5da8b9e6160f8 : RETRY_LIMIT in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f20bd20eac45d9abb2c01d4c4dd19c : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7f9e3b965264a66b193648fe2916d0b : SUCCESS in 4m 04s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/29512884a34c4172a18554c76910850a : SUCCESS in 5m 30s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/aea473a017bb4d34b49f6069f04ae01b : FAILURE in 6m 35s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/96dd9f1bebed4460a4057d4b7f6433c6 : FAILURE in 5m 08s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/accf0c917d8241aa8edec93fdba55a4f : FAILURE in 5m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/cb092a660a234c43a8c37b61f09a2e77 : SUCCESS in 1h 41m 36s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c9ef06cc059d4702bc2f6bea559b50c8 : SUCCESS in 1h 34m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2ab5602619db4f1b929afa46b33edb2d : SUCCESS in 1h 29m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9fa1ecbc681bc4943d5f02f236fb71a6c20f2fb","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-23 09:26:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e0f593e50534edb17b1fb2c43fbbbbbb52c7abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 11:29:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7d2ff0d0e7f494c9f2b86bd2f211089\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b81063b4418f46439d3e032034a6f2b8 : SUCCESS in 2m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c8f5040423c4b9fabca0e4785a652ff : SUCCESS in 1h 42m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89dc6d205bb8483ab1476593f5d6197e : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c34edc7a27db46f68ca5bb300bcff703 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e87f3e10ac249ecb2c666b4c38a775b : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a5511f9a0ed4e91a21cc2cefbaa9eb2 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df809d8427d465a86d7d089f5802e74 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c7416616b954dab91f6b7082b1bcb79 : SUCCESS in 3m 16s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/f58a2eee738746d29559e623852c22d2 : SUCCESS in 4m 47s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/d92bc466ab0d4883a1b27729c896477a : FAILURE in 6m 55s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/e42c3198e99648808a988619eb9f08ae : FAILURE in 2m 59s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/da5ae3e5d0cc4def94e79258ca643eb4 : FAILURE in 2m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/db63cf5246cd434c9c54af4000c942e7 : SUCCESS in 1h 27m 39s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/93896d7291464ad9bcca530c2d0ed032 : FAILURE in 17m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/69abdf96115443ecb527de1c14491293 : SUCCESS in 1h 38m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9874d99c128feea9beb7fa14c6f1c1f090e2bcb","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-23 11:50:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7755420cdc190652e080e3fcd8fb67f236d8aad","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 11:54:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"13a54bd3e44c8f15ecb0e5bb358ca829d8ab73b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 11:55:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3afc37a2399938c5972799d53a4514f7596fd62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 13:23:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0fbe25f4d79458485fdb66a2ef42f2a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b36237754824df7a921478005be1084 : SUCCESS in 2m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98f32ac6b88441e0bb24d9e272a479bb : SUCCESS in 1h 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff3b3947b5c416696f67d71b7955f4d : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/944035e1c53d4907acd37f0ea268e031 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f4b48447d204cdcb23c52a930608eab : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f2450ac07341c4bef553a38057b2e1 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/403aa0a6b20349fc9ecb6cf2f562893b : SUCCESS in 3m 12s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/3b51dc06683341118cf85d7b5ff89a91 : SUCCESS in 4m 44s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/f08014bc40824f72a75be7cf70adefc8 : SUCCESS in 1h 24m 19s","accounts_in_message":[],"_revision_number":1},{"id":"798374c0c57eefa9102a171be0a1ace74dd63e1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 13:23:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2222790a0d5fcbddd6621d77b823f5bf551b9e54","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 13:24:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4e048e97834200990db8e122e45dcc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b960201f3b942508d2e7d2c4857742a : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ac74ec1b57247e6ac8e1bdac1063370 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03dcc3c79a9dd9939915a55b42738813d6f1fca5","revisions":{"03dcc3c79a9dd9939915a55b42738813d6f1fca5":{"kind":"REWORK","_number":1,"created":"2022-04-05 16:26:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/79/836679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/79/836679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/79/836679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/79/836679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/79/836679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/79/836679/1"}}},"commit":{"parents":[{"commit":"723513a9d95b6d7678cc851afdfafba41120693c","subject":"tests: Fix test failures with kombu \u003e\u003d 5.2.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/723513a9d95b6d7678cc851afdfafba41120693c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-05 16:23:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-04-05 16:23:07.000000000","tz":60},"subject":"Bump bandit","message":"Bump bandit\n\nUse a more recent version because why not.\n\nChange-Id: I41538fd85956780250e3b7440130381151991b9c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/03dcc3c79a9dd9939915a55b42738813d6f1fca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/03dcc3c79a9dd9939915a55b42738813d6f1fca5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
