)]}'
{"id":"openstack%2Foslo.messaging~839264","triplet_id":"openstack%2Foslo.messaging~master~I5c38d3a3cafd49f8abc031e36bc595f32a8631d2","project":"openstack/oslo.messaging","branch":"master","topic":"bug-1970276","attention_set":{},"removed_from_attention_set":{"31039":{"account":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"last_update":"2022-04-27 15:16:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c38d3a3cafd49f8abc031e36bc595f32a8631d2","subject":"Add EXTERNAL as rabbit login method","status":"MERGED","created":"2022-04-25 22:03:28.000000000","updated":"2022-05-04 19:13:52.000000000","submitted":"2022-04-27 15:16:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839264-bug-1970276","meta_rev_id":"afbf7b083db4f44b599299be7872e7b5e7508cb4","_number":839264,"virtual_id_number":839264,"owner":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-27 15:16:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-04-27 08:24:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-27 13:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-27 13:33:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-26 00:18:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 08:24:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-04-27 13:33:57.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"bd728e60973c1fc77f31180938bc74192b8c5ad0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2022-04-25 22:03:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"121bab6684e3f4e0372dee0a7d58a686a183ee66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2022-04-25 22:34:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"68e5d1dfe47ded00c84336615f66399f370b94cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 00:18:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92eb2f6a710041489011a1c6e2dbf583 : SUCCESS in 1h 41m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96133bd49216424aaf7c2a94e99bbc60 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8987a3fb5f7840d0863d51447b8981f4 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f905db3baabd4c85975bb140918b6b0c : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a44184aacedd48c0b1152fbc78734f5b : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b65df01f734d369ece08c2f4dd6e5e : SUCCESS in 5m 13s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/841149e5802e49be9e7413a32f09e48b : SUCCESS in 6m 15s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/b4ec3da77b26463f86e338eeab398036 : FAILURE in 7m 10s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/cff25cae3e2b4436a2b111581138cd41 : FAILURE in 5m 15s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/dd434ffd2f514cdbbfae1e0fc566515a : FAILURE in 5m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9a98f6f3912444649e1ca05010830213 : SUCCESS in 1h 27m 40s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/903484436f1448ee85b5922f987080fb : SUCCESS in 1h 35m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d08fc56f28d04bcdb1d01d4db7fccd15 : SUCCESS in 1h 23m 52s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/312f465f69a24157a49424c78da59cfe : SUCCESS in 1h 00m 47s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b75da1bf014496ba77b30ba3ea72583 : SUCCESS in 1h 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"93018421a0e69609f61651526a2825a63d56f8ad","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-27 08:24:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca2b038c8a200cf958783a492568da5c679b315d","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-04-27 13:33:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"06f53d507a824b2088b8f4fe4cc378a05845a67e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 13:34:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e8ba78ad955c1f9970e34b5f813f1e4ede0dd200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:16:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/423c6d446f52468aaffbf38cfac70114 : SUCCESS in 1h 41m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/784d31ba96bd4b9c845b732d06296fd9 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bc7d386d711405b8075ac064b80c5ea : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/358e06f63e754b95b5e3c149547fe325 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e910b59a2a8a465cb8afe2b8efba996a : SUCCESS in 4m 38s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a68fb1db444046949befeea2803c671b : SUCCESS in 6m 28s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ea904b282e9546a19503044607288b3d : SUCCESS in 1h 31m 32s","accounts_in_message":[],"_revision_number":2},{"id":"7f320e7e62c3f934159f19cb4d18f97dc85d8069","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:16:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"31fef720c90a6326e29bddcb38da32a4c66fcd74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:18:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95a97c991804a7a85d859b35672fd30 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2},{"id":"afbf7b083db4f44b599299be7872e7b5e7508cb4","author":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"date":"2022-05-04 19:13:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"821197b947393e4f6051e4b45818cc343abf06cc","revisions":{"d104d71f46221d0298cd1f5c5d078058ee899c58":{"kind":"REWORK","_number":1,"created":"2022-04-25 22:03:28.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/64/839264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/64/839264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/839264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/839264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/839264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/64/839264/1"}}},"commit":{"parents":[{"commit":"723513a9d95b6d7678cc851afdfafba41120693c","subject":"tests: Fix test failures with kombu \u003e\u003d 5.2.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/723513a9d95b6d7678cc851afdfafba41120693c"}]}],"author":{"name":"hamza alqtaishat","email":"hamalq@verizonmedia.com","date":"2022-04-25 22:02:30.000000000","tz":0},"committer":{"name":"hamza alqtaishat","email":"hamalq@verizonmedia.com","date":"2022-04-25 22:04:06.000000000","tz":0},"subject":"Add EXTERNAL as rabbit login method","message":"Add EXTERNAL as rabbit login method\n\nAs explained in the link below kombu has login method called external\nhttps://docs.celeryq.dev/projects/kombu/en/latest/_modules/kombu/connection.html\n\nThe login method external is not listed as a choice in the Rabbit driver\n\nCloses-Bug: #1970276\nChange-Id: I5c38d3a3cafd49f8abc031e36bc595f32a8631d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d104d71f46221d0298cd1f5c5d078058ee899c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d104d71f46221d0298cd1f5c5d078058ee899c58"}]},"branch":"refs/heads/master"},"821197b947393e4f6051e4b45818cc343abf06cc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-25 22:34:25.000000000","uploader":{"_account_id":31039,"name":"hamza","email":"hamalq@verizonmedia.com","username":"hamalq"},"ref":"refs/changes/64/839264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/64/839264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/839264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/839264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/64/839264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/64/839264/2"}}},"commit":{"parents":[{"commit":"723513a9d95b6d7678cc851afdfafba41120693c","subject":"tests: Fix test failures with kombu \u003e\u003d 5.2.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/723513a9d95b6d7678cc851afdfafba41120693c"}]}],"author":{"name":"hamza alqtaishat","email":"hamalq@verizonmedia.com","date":"2022-04-25 22:02:30.000000000","tz":0},"committer":{"name":"hamza alqtaishat","email":"hamalq@verizonmedia.com","date":"2022-04-25 22:33:12.000000000","tz":0},"subject":"Add EXTERNAL as rabbit login method","message":"Add EXTERNAL as rabbit login method\n\nAs explained in the link below kombu has login method called external\nhttps://docs.celeryq.dev/projects/kombu/en/latest/_modules/kombu/connection.html\n\nThe login method external is not listed as a choice in the Rabbit driver\n\nAs explained in RabbitMQ documention\nhttps://www.rabbitmq.com/access-control.html\nfor Authentication using Client TLS (x.509) Certificate Data\nclients must be configured to use the EXTERNAL mechanism.\n\nCloses-Bug: #1970276\nChange-Id: I5c38d3a3cafd49f8abc031e36bc595f32a8631d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/821197b947393e4f6051e4b45818cc343abf06cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/821197b947393e4f6051e4b45818cc343abf06cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
