)]}'
{"id":"openstack%2Foslo.messaging~842984","triplet_id":"openstack%2Foslo.messaging~master~I3a0242208f411b430db0e7429e2c773f45b3d301","project":"openstack/oslo.messaging","branch":"master","topic":"hacking","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-08-30 19:45:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a0242208f411b430db0e7429e2c773f45b3d301","subject":"update hacking pin to support flake8 3.8.3","status":"MERGED","created":"2022-05-23 14:39:57.000000000","updated":"2022-08-30 19:47:24.000000000","submitted":"2022-08-30 19:45:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842984-hacking","meta_rev_id":"336a3b7373f2d9932e2e218e4a93a4e3d884db86","_number":842984,"virtual_id_number":842984,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-30 19:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-05-25 14:07:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-13 15:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-13 15:22:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-23 16:06:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-25 14:07:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-06-13 15:22:26.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"31bb455eb7c63c32e4af16a5ff6c12232c16d28e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-05-23 14:39:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"417c9a80143830b54ecdaf7948249fb7f685a8ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 16:06:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9520807a0e6747afa486177c2f4d09ff : SUCCESS in 4m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8bb4216f83148cebfb7846852193460 : SUCCESS in 1h 14m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c55f028634b4402a544987ea4579a7c : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef82d9e9a6254db388504a3a80e79a4b : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b92e4b59fa804a2b89b8e6c62b9e1ad6 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44248640cecf44f9ad52ac4dcd40ae1d : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f82bc16c53464ac1a0222bdb776dc568 : RETRY_LIMIT in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e08c9a1227346d7a0d60ec83eaf20f8 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17c508389b484d8c83d346f74efb9bb8 : SUCCESS in 5m 17s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/5132b711d03148db884c60cfeacccfe0 : SUCCESS in 5m 19s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/bec609a46f0a4a30afe42f0f201d27f7 : FAILURE in 7m 09s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/427b860e5479499e925738a0b07fd925 : FAILURE in 6m 34s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/22f856d171ae40bca629ed47120dc33b : FAILURE in 6m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/be637e07a7de4fe5bd5836cc7d4264b2 : SUCCESS in 1h 23m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a27e002d02304fe58033cf52d9958cb1 : SUCCESS in 1h 02m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/8352e0b299e64c12b198c23c9aa9c6eb : SUCCESS in 1h 15m 47s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2afd3ba066da4ea695f88c484df0d299 : SUCCESS in 52m 59s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9e9e27f72e4490ca01aecf7f88a1976 : SUCCESS in 58m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f171e6a5721fbdd877084a1d693a93d7aeb19cb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-25 14:07:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a6ebe0f4d0b2f3f4e84fc2de929400353025130e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-06-13 15:22:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bff7d34f71058f6b93deb1bc67db33a2467f334d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 15:23:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ff5e9d180483793aa853b634c16abd35dbc4d66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 17:34:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b73e98857496489e838c86a61f49f4bb : SUCCESS in 3m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a6e3891e7714bf18f262c44c29a56d8 : FAILURE in 1h 38m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b836c8aa3546d9818081eb8de5f648 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe43ea3c2a044781ac20866bb221cbe4 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f54844bd7f34eec9454e704c3803362 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b01082bf1ed466193ad700cbefc05db : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f18ba15790e64976995626dd5f642a58 : SUCCESS in 3m 19s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/882972818e254a188ec7c5d4f8f07ea8 : SUCCESS in 5m 29s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4d41c774782d40db9b66bf81fab65f69 : SUCCESS in 1h 36m 31s","accounts_in_message":[],"_revision_number":1},{"id":"48aaa583add6d7d771004de37f9eee4945746871","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-08-30 16:20:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"516fd7e96ba54697a06ae36b8eceb9db034bedd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 18:04:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf7e9d78a3084adc8eb203e413c404bf\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7fe2b2773fa44739bb410f573bd24a1c : SUCCESS in 4m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce35d0220cd54aa794bf02e957155315 : SUCCESS in 1h 40m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2320f145b9804d90b24b453bd03f712f : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c07fabc3c171415f88a9ee7367fbd840 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4aa0200e89f473492cd8ddb26e1524f : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a41b0329e104143a2e19ac7ad76fb44 : SUCCESS in 7m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa3b1c92c4524a40898b36148e45e5a4 : RETRY_LIMIT in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1203a8deb2c4108a75cc4798d737737 : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/094690ab43874492ab614c8cd3155b81 : SUCCESS in 3m 26s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/3216b593a66d4d3ea14048865ad4d0f7 : SUCCESS in 5m 03s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/384fdd605cf94275b81181979aec1306 : FAILURE in 8m 12s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/41c889be59e047f1875beb0a080101d5 : FAILURE in 4m 34s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/3d3c3fd61da24478b06b41e6bdff4571 : FAILURE in 4m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/cc0a23fdf77d4d42ada3092e6e006d5e : SUCCESS in 1h 42m 36s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/210bb8243cf94ea587410b6f009a92ec : SUCCESS in 1h 38m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1d397911f68a4b6cb7d9bca3b12dbf67 : SUCCESS in 1h 40m 38s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fb93f8c2ec2947e3a045560cfa8810d5 : SUCCESS in 1h 06m 51s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ff530f051164f8e9f816329a540e551 : SUCCESS in 1h 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38f2be3151d952fc7e7d0a5fb09fa1c484b5af1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 18:05:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5e3cf8ebbb8330b0b4079fe5cf9105fea7409e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 19:45:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2577d850396f408da18a8cca8115df3d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/00807b824f67438aa7b8978fe9c858ad : SUCCESS in 3m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/690bee4ea84349a4b0f79addb20de033 : SUCCESS in 1h 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc6cf32fb4d74cb7baa7c994a126a290 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/134f3936edc545eea04d77a899798ff9 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea597585d6d04c2c845ec3d4a359459c : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abbdef73913a44bea46422d672c5aab8 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1a6fc75435f42958bc77bcbb75b3a94 : SUCCESS in 4m 54s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/0dc1db35a147495b9acafd7205973375 : SUCCESS in 5m 32s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/60b052a60fcf43ed9e776451ac32982e : SUCCESS in 1h 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b8255cf5fcf60c4d849597a7910e3c9f956a037d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 19:45:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"336a3b7373f2d9932e2e218e4a93a4e3d884db86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 19:47:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/672848f7fa8c4273b9c2076b5925d782\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed03a9f2ba534a16a90419266ae7138e : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4209d88b123e4a5fbd1f76dfb904f60e : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cde68026eb32ff25c45de895e5dc3608096ed873","revisions":{"cde68026eb32ff25c45de895e5dc3608096ed873":{"kind":"REWORK","_number":1,"created":"2022-05-23 14:39:57.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/84/842984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/84/842984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/842984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/842984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/842984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/84/842984/1"}}},"commit":{"parents":[{"commit":"f611370555c994dd685b680a4ef09dbd57b697b2","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f611370555c994dd685b680a4ef09dbd57b697b2"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-05-23 14:37:10.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-05-23 14:39:56.000000000","tz":0},"subject":"update hacking pin to support flake8 3.8.3","message":"update hacking pin to support flake8 3.8.3\n\nthis change updates the max version of hacking\nto 4.1.0 to allow pre-commit to work with the\nflake 3.8.3 release and correct one new error that was\nraised as a result.\n\nChange-Id: I3a0242208f411b430db0e7429e2c773f45b3d301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cde68026eb32ff25c45de895e5dc3608096ed873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cde68026eb32ff25c45de895e5dc3608096ed873"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
