)]}'
{"id":"openstack%2Foslo.messaging~848338","triplet_id":"openstack%2Foslo.messaging~master~I0e1453a983c1c24aef74c6f6c2502871759d71ed","project":"openstack/oslo.messaging","branch":"master","topic":"asyncio-nats","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-04-11 21:11:30.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-04-11 21:11:30.000000000","reason":"Change was abandoned"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-09-06 08:59:26.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"I0e1453a983c1c24aef74c6f6c2502871759d71ed","subject":"[WIP] Add NATS driver","status":"ABANDONED","created":"2022-06-30 22:02:11.000000000","updated":"2024-07-29 10:06:39.000000000","total_comment_count":20,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"3c2e862543a606d3d2b3c4c666e7059e2dc1d1a6","_number":848338,"virtual_id_number":848338,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-02 23:16:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-01 00:51:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-31 13:16:16.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-09-02 07:26:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-10-31 16:02:08.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-03-25 20:18:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"}],"messages":[{"id":"42452d75cf1610d05b5c25d62b994e3c51f4391a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-30 22:02:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbd4189ad35c240e79816cc659645a4866f16807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-30 22:27:51.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0d58d0f871f15ff2ef7462fc611b7a93ebd9ed25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-30 22:41:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a76617772d695be342483f59298bb74c7b60c874","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-30 23:23:22.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f937df55dcb14209833dcd4d1997af1565416c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 00:51:18.000000000","message":"Patch Set 4: Verified-1\n\n(22 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/24c1a5b4c83245dfb14d730983d611f5 : SUCCESS in 1m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1921c01045604e37abd5ba6dbe11ae78 : FAILURE in 3m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e448de65dec44943950e9ee33d14ec93 : SUCCESS in 1h 26m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c4100ea78554009a4ef02a8fa200d1d : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e139669c5b44266ba7d00f242a4ff5e : FAILURE in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5907db70934d4c608905f73054754228 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ce469cf3fd644c384ff5767e3452328 : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24d74c2caa6a41ffa23a9dbcc128c7dc : RETRY_LIMIT in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20388a02ce604fef94e08e3117e1d29b : SUCCESS in 5m 23s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/8566df6a52474379893ca9085c2f3c18 : SUCCESS in 5m 01s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/38a4c9a3a76340f1939fb9192b6d6a77 : FAILURE in 7m 27s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/eb431e6e91784a8ba72f2d26e1f71489 : FAILURE in 5m 09s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/e893b3b9d4e846d4aef8318a3ca03e34 : FAILURE in 5m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2479ab62928b4089a5df67f5888629b5 : FAILURE in 21m 15s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/eb5652a6de424bad9f376fb0f6f577e7 : FAILURE in 15m 11s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/90b40c4a47214ddf9b2500feb0f50dbd : SUCCESS in 1h 16m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0142f641c7674eb483b9957495458f00 : SUCCESS in 1h 21m 15s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b5731294f28c4ef6a9945b8f01654782 : SUCCESS in 52m 50s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/176584f97d3f4bebb8735a84aebcb326 : SUCCESS in 1h 01m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e43e701caf95e83ec47b328725600d11a7fedb46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-01 07:23:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9d31b0e777c1dbf52927b53c9eddf31b8b703d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 07:54:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/12859dd587aa4273ad61cc8e0c563dcd : SUCCESS in 2m 59s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/026e75b81d3342b89c280a6c35632783 : FAILURE in 29m 55s","accounts_in_message":[],"_revision_number":5},{"id":"0354294ce498f606f477d91a8360518ba12ba1f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-02 13:26:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c0a13e16c05a806a713f49175236904a5679e311","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 13:43:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb0bc3e9deaa4d289e6c7e10c6c21804\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56bf7f4cafe247e9a2fabf0e4adc602a : SUCCESS in 2m 37s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/7e96a3c80de7426394f1bafec553aa1a : FAILURE in 15m 10s","accounts_in_message":[],"_revision_number":6},{"id":"ec28724b4b063a73d8e8efcf5ee47bb3fb48a756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-02 13:45:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dbaaced59a7c8acc40530535313f897da52c7c7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 14:10:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87ed5fc7fe0646f8a719f8381f923021\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6824ce05f05944649cd0311098059b09 : SUCCESS in 1m 47s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/08551fe8335f4229b7b1294a5be78086 : FAILURE in 24m 06s","accounts_in_message":[],"_revision_number":7},{"id":"6009fe745f5954b6d9b84cefd5fbf6504451c452","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-02 14:29:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e2c967e3de4c961bf1d5ff975b2d60844e88dee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 16:00:25.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b78bfdd1c254797a8f9559b08627131\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9aedf86498864c42a157192602fe4a4b : SUCCESS in 1m 57s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/78d6f490ac504f689c4484e6c9cdce30 : FAILURE in 1h 25m 44s","accounts_in_message":[],"_revision_number":7},{"id":"5cf85aa3eb2d7bdb218e644a8999580020410196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-04 19:16:12.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7aa738182515505f66a7986fef9e11904a33a711","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 19:46:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9eaae6ac5c8d40f2bab6ff8b1c36f25d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6b1d5d61ba2344ce9232f4683582a8d6 : SUCCESS in 1m 52s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/53687b7c585f4a8fb92251fd7bd5abc3 : FAILURE in 25m 34s","accounts_in_message":[],"_revision_number":8},{"id":"a9381d8ebc160a674b08990d34d5f8b95906d9a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-04 20:17:53.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"d135560947bc44b401cf01caee53ffd3c8954888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 20:43:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08d33ee8f6c64cc293d49096af7bd399\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/35e6600115cc4466bb22703873031219 : SUCCESS in 2m 05s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/b1ad245460fd48c9aedeb0eb739884cd : FAILURE in 24m 07s","accounts_in_message":[],"_revision_number":9},{"id":"4ae5e047545c4a9c1f2d122738dcf3ac9306e021","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 18:22:32.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"384a021ef4121f1d20e362a9714350c15a376ceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:02:39.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da9673c9e8d742849506480e9b518f5c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9a017fde29c245a0992bc47c4500c1a6 : SUCCESS in 2m 17s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/fbe4e78b286a49a4971d84a0add96e52 : FAILURE in 1h 39m 22s","accounts_in_message":[],"_revision_number":9},{"id":"77a85c92807c05b672ff91a95e4f77bb48d80359","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 21:08:51.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"63e46cf6f2e058cb2af32841bce7fc97aa3f3c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:46:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d08bdb677acd41b3abec2859d95293a8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/096a81c80d2a44e7833ed571cd92ef3f : SUCCESS in 4m 16s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/581714d0a7ab4ea3b007c6a8049d0110 : SUCCESS in 1h 36m 49s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/1f14d274e0ce4406abfa432e61a319e1 : FAILURE in 37m 10s","accounts_in_message":[],"_revision_number":10},{"id":"13b37bc2b2cb8a72a253583214949dabe02acd80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 23:27:09.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c63e0724f7b69915249abcd4f08753b0512c0cb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-16 00:55:31.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5dee4eb47324c4b9a6cb73a886a81ee\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/529c38fe8bfb4cb5aede583fea7723d3 : SUCCESS in 2m 09s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c334ab8a356145e8b50e847fcb631503 : SUCCESS in 1h 27m 05s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/6a4c320c577148a9ab39c8d681ee6f9c : FAILURE in 39m 42s","accounts_in_message":[],"_revision_number":11},{"id":"ac582874b77d32866e61c8d2568b6531f530a41b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-15 08:02:51.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ee9f6a0a2fef0eafea1826d23842e6dee59e44ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 09:49:35.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3118b971b3a849ff8260a4e806ee7e5a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7627a1493e9540ef8a191b0be59f1199 : SUCCESS in 4m 04s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fd56daa1c7084f9ab07aec96fce2ced3 : SUCCESS in 1h 45m 13s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/b66a273c50ff449e9120f1ebc09a9625 : FAILURE in 1h 16m 49s","accounts_in_message":[],"_revision_number":12},{"id":"49c2725391865daf57d15824658bdd56e1f2d4fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-15 21:35:56.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8a3842c26e4f93f899d24f38835f659c348e9c01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-15 21:40:34.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"11d23adb8b7d7eac7bb15dd1ee753fb590fad66d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 23:14:55.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6c90250165548bb98ca60f4e82df109\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b9d88164583d400b8f6c68e52be1a8f5 : SUCCESS in 2m 13s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/be7c272e2c554b90843884ddb8b91568 : SUCCESS in 1h 08m 07s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/11a9bec389fb4b8bbdf0460ffcdef475 : FAILURE in 1h 29m 09s","accounts_in_message":[],"_revision_number":14},{"id":"f52f2b4c09134e07583d7f7013bc35e91b0b6444","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-17 07:05:03.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"e3c52be26d0e8fae5c98e0a767b8d9aacdedac7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 09:02:10.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/819d33309a954081b0581e0a3b8f2f25\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3585f5e6d77d480a826019c5e69605ae : SUCCESS in 2m 35s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/09697e58029d4193b6cb3bc8dd265edd : SUCCESS in 1h 36m 34s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/99ec794299cd47de835f271d6bbd11ae : FAILURE in 1h 52m 20s","accounts_in_message":[],"_revision_number":15},{"id":"4e8d675259eb6524a524e94ca80e72e5f17e8ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-17 12:22:06.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"ef9a3669796e045e13089fdef43548970596ad53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 14:07:39.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c197257b305403197b89cdc8bf2f429\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8543985af94244ab87bc9017edbd7171 : SUCCESS in 2m 30s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3249d2a922a242fe89b09de5b635edab : SUCCESS in 1h 42m 59s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/390203588de94b12b2604180ce64f484 : FAILURE in 24m 14s","accounts_in_message":[],"_revision_number":16},{"id":"d1c8816714eefbed0539e1bfd1b31d1963f9a5f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-17 17:57:10.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"8eec8e645de09d91ab1c1bd2bfcd8f1b363d6403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 19:24:36.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12da14b146294f0bbc012cd110ee2165\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d3a11477521c43d88490e5244af718ee : SUCCESS in 4m 07s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fb9180cff150490899233ed416d9dd34 : SUCCESS in 1h 21m 18s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/c6c21c0df57947a085f77310e7cd508c : SUCCESS in 1h 25m 57s","accounts_in_message":[],"_revision_number":17},{"id":"fe689cde89bca168bc486039f81e4478d9c1a21e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-31 13:17:56.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"930fbf3c1bf6fa8b2073c16dfa9573a8aa0aa69e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-31 13:19:15.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"13394cf9414b3621ac3d235e1dcca73188c6b097","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-31 13:35:46.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"b28698be24ce7ec72e8fdcf6d635824c51a28ad3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-02 07:26:44.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"360dff19ef69899b04b50daf6b3d2ef04ba98c72","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-02 08:16:32.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"cad5af50644744ae539bc11fcab1b604cb45b103","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-02 08:49:00.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"0d8a5e811a0f54e24081f244e8287122bf9dba76","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-02 11:21:38.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"3cf5c58ae552dfc247d2639dd1fa0285bffba255","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-02 12:42:27.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e1636952adcba6d7fe868cc2f7b210a20141747b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-02 14:18:01.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"ff462f272f5662d76bff536f80a2ee5c398f0edc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-02 14:29:14.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8bfd03746588aa0486b0a0a8b19b1f89609de2e4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-02 14:56:25.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"93b954e24ee587dafd90b676237547d3eeb8a137","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-05 13:12:29.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"dd3b6468064eb91b146f941e0e9942a74c643800","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-05 14:58:40.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"4e6dc2116cb08972abb040b1ef391095fee3991b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-06 08:57:01.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"0839e0ba8c1840376f8689a03f3a49a79d64b399","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-06 08:59:26.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"5ac55038e896e45bac2b66a019a8253c752c9839","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-06 09:00:15.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e01f26a758bb20977ea66edbe53e8d460499c9f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-22 14:03:38.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"f0e148131becb346a2cc8e658ad9351e6f8a76ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 15:09:43.000000000","message":"Patch Set 18: Verified-1\n\n(19 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b755754006040f693eab363bce2da41\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6d96e46495fd4ca894ed4053e31faa91 : SUCCESS in 4m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6089c9aac804d189126f51a686c5cd8 : FAILURE in 3m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3296aacbd17462b818d5e693bee40fd : FAILURE in 37m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b04d684953d45258fd4781a8ee6c2b2 : TIMED_OUT in 51m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eb3467ed5d24ea58016873795954b34 : FAILURE in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c66662adf53844e4a5772d749bfc26bf : TIMED_OUT in 41m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebb617efd38346869dd2d5831e7074a0 : TIMED_OUT in 41m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18b456ddaba645b5b81b25e5bbefe692 : RETRY_LIMIT in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4339bca91f48668b1ed05a9cedf192 : SUCCESS in 4m 35s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a96f6d5a7c1248ee95f37643d21ea5a0 : TIMED_OUT in 41m 02s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/276379c517cf435a98c6fbe1be13b828 : TIMED_OUT in 41m 00s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/5dbd43b8475348c0b349f1d51ac35ae0 : FAILURE in 4m 31s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/58d1d3f82ee249059fe6a90df0fa4c62 : FAILURE in 6m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e257942102f8437f91056831bdb46a5e : FAILURE in 34m 01s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6e2ec0f6514540f1ae6fdc54da98969e : FAILURE in 34m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5de62529fd6d4c08bc7501463feff0e4 : FAILURE in 30m 32s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/6d76d7571c2b4cb78cba412d0b8fadc6 : FAILURE in 51m 16s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b8f38a0911b472b96b43a1178725b6f : FAILURE in 1h 02m 52s (non-voting)\n\nWarning:\n  Comments left for invalid file oslo_messaging/rpc/client.py\n  Comments left for invalid file oslo_messaging/server.py\n  Comments left for invalid file oslo_messaging/transport.py","accounts_in_message":[],"_revision_number":18},{"id":"fa7c7823a79f900cf520bfda93a857555fff977c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-22 19:07:04.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"bd8268dda96f5be887f88254318cf84a74e3d720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 20:56:59.000000000","message":"Patch Set 19: Verified-1\n\n(19 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b962eb6130c84bb4aa6e71b555c9d21d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1df2ad1c50014dc6a6368e28f955b5e1 : SUCCESS in 1m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/67318da5da0e42acb07ca4b783f0e8ef : FAILURE in 2m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4c1f218c2b84a1893212120f7987780 : SUCCESS in 1h 44m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9ba7129231446bc8a794517fd9e804f : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f0ca3d8f0f64cdbad85b7a6f9578a7b : FAILURE in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3572152d9184eab9b1d24c88140cf5e : FAILURE in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a04c28fae9ae4f11ae77fb8581f7eb57 : FAILURE in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ae3c2c28f3f407ba7ef220a361d1e0d : RETRY_LIMIT in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53609ed19874bbe82ad2ae681a9decd : SUCCESS in 4m 14s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/d449be9d3b4c41a588100123c1481cbb : SUCCESS in 7m 42s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/78d0ac1085684dddbbb2501ddcba7a20 : FAILURE in 6m 36s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/05bbda8cab91450fa35c730b751ce18f : FAILURE in 6m 21s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/126c463e5e3b46e69a442987878e15e8 : FAILURE in 5m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d1b64eb50f3a41e1bb4d56f67ecbe0da : FAILURE in 1h 33m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c50d59f03950499bac53143b14b3e336 : SUCCESS in 1h 31m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/43bbdf69959c4e50ad06a4d64a878f58 : SUCCESS in 1h 16m 38s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/eecbca95f20441d7a0566d04c007f89e : SUCCESS in 57m 46s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e1d7a9383634b48932feea92378f4d0 : SUCCESS in 1h 08m 43s (non-voting)\n\nWarning:\n  Comments left for invalid file oslo_messaging/rpc/client.py\n  Comments left for invalid file oslo_messaging/transport.py","accounts_in_message":[],"_revision_number":19},{"id":"86f7308756dbbf30e5bd24c6bed567cc8b00dee5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-23 06:02:47.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"cd9925529eb61a4812c853c55b3df01d85c21b59","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-23 06:05:26.000000000","message":"Topic async removed","accounts_in_message":[],"_revision_number":19},{"id":"7563fa534ed07fe44ea523f86368cd772e28239c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-23 06:05:30.000000000","message":"Topic set to asyncio-nats","accounts_in_message":[],"_revision_number":19},{"id":"3ca377c38b521f0a4cccd88083461a0c8461ef12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 07:40:58.000000000","message":"Patch Set 19:\n\n(19 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5667586d86d74eb18e948be49b458c16\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5918e7383e7d4c878b71f5f049ae70a8 : SUCCESS in 3m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/64004c0f1abd4e15aa648f1dbf150c62 : FAILURE in 4m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f6b8aeac9c649e3aac38b05c744ecd7 : SUCCESS in 1h 36m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fea77045bff47deaab7336674f5f912 : FAILURE in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c5c7f9a4e274019a554013e451e2e59 : FAILURE in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bb9d7362936409c93a195f4b1609e95 : FAILURE in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b93e32009c4d4fb7b2fec22fda46a817 : FAILURE in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51c4c3a8acca45f3ba24c651bcbf910e : RETRY_LIMIT in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b6b2b74d6ae466faff363c20263b479 : SUCCESS in 4m 38s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/4352921604264f7597853b8b1b2b6aff : SUCCESS in 5m 37s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/e7a9daaca556433088776f5ffe85abd5 : FAILURE in 7m 43s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/1006a9097b914e91bb400d0eb7681cef : FAILURE in 5m 23s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/0b1c37675430474bab3d0821fcd38b91 : FAILURE in 5m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/aabe4a1afda644299e6791089f2c3f06 : SUCCESS in 1h 30m 54s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/1aec8f5cc85c4ac4a9576e13a6cd1fa4 : SUCCESS in 1h 17m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e884ae1a632d4ec2b3e7537582a41acb : SUCCESS in 1h 29m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/5df6009c803c45309b63eed9c8689768 : SUCCESS in 54m 31s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/08ffa08aa0554d649ca125f86d4248cc : SUCCESS in 1h 08m 03s (non-voting)\n\nWarning:\n  Comments left for invalid file oslo_messaging/rpc/client.py\n  Comments left for invalid file oslo_messaging/transport.py","accounts_in_message":[],"_revision_number":19},{"id":"ac3ff35a39a496f409f6069cdd0b34e5a6b4e50c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-02 21:41:02.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"f17babd97b04e0eb00af210e4794128f973f4f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-02 23:16:57.000000000","message":"Patch Set 20: Verified-1\n\n(19 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95f8e221743e46e99bc4b5789a621ef1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af14cdd96d5942748f835ccca48e86b2 : SUCCESS in 2m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c53c495043da48a092bdf4430ff018c8 : FAILURE in 4m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/164a51f66f61480a8c727052137e7309 : SUCCESS in 1h 09m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35bec0d8411e4261b7aca17ef5c31879 : FAILURE in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528040dd295749978bba796042c9b212 : FAILURE in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/868c286a80954573939e8b26429c1139 : FAILURE in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16fed1fb97304878bd33eb5b8655c96e : FAILURE in 6m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0f403160c214d7499549aaf85b554d5 : RETRY_LIMIT in 3m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d9815e17104296a764ac15fe2bd233 : SUCCESS in 4m 40s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/c066fc2751aa4638a84e183df4c48310 : SUCCESS in 5m 13s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/508b66f1ab024dafa679972713ac0dcb : FAILURE in 7m 31s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/c3a4c91ab3d14917951380f6cb806774 : FAILURE in 5m 04s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/db23fe746a844c439fb087078f01c88d : FAILURE in 6m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1a5b83d482014314a5051bffaa044d38 : SUCCESS in 1h 05m 19s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/d9ce71c602a94b8690812c9786a66886 : SUCCESS in 1h 34m 04s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/aafe7b171c404a8db1636aa2d72d8c4e : SUCCESS in 1h 30m 44s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4527c4e646de44fdaf367c599a17ee6f : SUCCESS in 54m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a87e6afdd51463389d3336a333e198e : SUCCESS in 51m 26s (non-voting)\n\nWarning:\n  Comments left for invalid file oslo_messaging/rpc/client.py\n  Comments left for invalid file oslo_messaging/transport.py","accounts_in_message":[],"_revision_number":20},{"id":"664252595f85426645b1444345a27858ced983b0","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-03-25 20:18:40.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_30491\u003e.","accounts_in_message":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"_revision_number":20},{"id":"b78a8ecf1eb159615b83b0b458209a3c721f3e57","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-11 21:11:30.000000000","message":"Abandoned\n\nWe will not work on the async driver, trying sync driver in another change","accounts_in_message":[],"_revision_number":20},{"id":"3c2e862543a606d3d2b3c4c666e7059e2dc1d1a6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-07-29 10:06:39.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"6f57130ac1c340ba4b930b2832ab5b94f1e1a338","revisions":{"407df91186a3cc4a98b15af0bf7741e205b8f6d7":{"kind":"REWORK","_number":1,"created":"2022-06-30 22:02:11.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/1"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-30 22:02:09.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/407df91186a3cc4a98b15af0bf7741e205b8f6d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/407df91186a3cc4a98b15af0bf7741e205b8f6d7"}]},"branch":"refs/heads/master"},"4699378a4eebfdabba41ff9b300d1ae2f178120c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-30 22:27:51.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/2"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-30 22:27:37.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4699378a4eebfdabba41ff9b300d1ae2f178120c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4699378a4eebfdabba41ff9b300d1ae2f178120c"}]},"branch":"refs/heads/master"},"59a8e7aa3cfc59b18c374ef37977386441686b79":{"kind":"REWORK","_number":3,"created":"2022-06-30 22:41:19.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/3"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-30 22:41:05.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/59a8e7aa3cfc59b18c374ef37977386441686b79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/59a8e7aa3cfc59b18c374ef37977386441686b79"}]},"branch":"refs/heads/master"},"dd5186c91f0fbf61d9be9e6c58789f8d47fd5897":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-06-30 23:23:22.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/4"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-30 23:23:10.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it again.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dd5186c91f0fbf61d9be9e6c58789f8d47fd5897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/dd5186c91f0fbf61d9be9e6c58789f8d47fd5897"}]},"branch":"refs/heads/master"},"209cd246f6060684e0066bfa5dbfaa63f2ed847e":{"kind":"REWORK","_number":5,"created":"2022-07-01 07:23:56.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/5"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-07-01 07:23:48.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it again.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/209cd246f6060684e0066bfa5dbfaa63f2ed847e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/209cd246f6060684e0066bfa5dbfaa63f2ed847e"}]},"branch":"refs/heads/master"},"c65d6b21a6af569c79a764a557bbbdc19e670c60":{"kind":"REWORK","_number":6,"created":"2022-07-02 13:26:43.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/6"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-07-02 13:26:36.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it again.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c65d6b21a6af569c79a764a557bbbdc19e670c60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/c65d6b21a6af569c79a764a557bbbdc19e670c60"}]},"branch":"refs/heads/master"},"d184a3e0c8c361c3231df713e7f7bfc5096f97cb":{"kind":"REWORK","_number":7,"created":"2022-07-02 13:45:19.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/7"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-07-02 13:45:13.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it again.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d184a3e0c8c361c3231df713e7f7bfc5096f97cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d184a3e0c8c361c3231df713e7f7bfc5096f97cb"}]},"branch":"refs/heads/master"},"79758384c4d45b1a67f1612ebfba5df476bb1510":{"kind":"REWORK","_number":8,"created":"2022-07-04 19:16:12.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/8"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-07-04 19:16:05.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it again.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/79758384c4d45b1a67f1612ebfba5df476bb1510"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/79758384c4d45b1a67f1612ebfba5df476bb1510"}]},"branch":"refs/heads/master"},"24e6340b2e7cd188d8dca5d19172662a306637f9":{"kind":"REWORK","_number":9,"created":"2022-07-04 20:17:53.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/9"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-07-04 20:17:44.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it again.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/24e6340b2e7cd188d8dca5d19172662a306637f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/24e6340b2e7cd188d8dca5d19172662a306637f9"}]},"branch":"refs/heads/master"},"1ce9f2f14907d677d398d6177028f2b2e189b0ae":{"kind":"REWORK","_number":10,"created":"2022-07-15 21:08:51.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/10"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-07-15 21:08:41.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it again.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1ce9f2f14907d677d398d6177028f2b2e189b0ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1ce9f2f14907d677d398d6177028f2b2e189b0ae"}]},"branch":"refs/heads/master"},"8745a6b6904097062ebc7a8aedd6e4f75ec5cec1":{"kind":"REWORK","_number":11,"created":"2022-07-15 23:27:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/11"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-07-15 23:27:03.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it again.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8745a6b6904097062ebc7a8aedd6e4f75ec5cec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8745a6b6904097062ebc7a8aedd6e4f75ec5cec1"}]},"branch":"refs/heads/master"},"cf2492b4ffcf22bb164e6f58628de4afcc4c8303":{"kind":"REWORK","_number":12,"created":"2022-08-15 08:02:51.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/12"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-08-15 08:02:42.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it again.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cf2492b4ffcf22bb164e6f58628de4afcc4c8303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cf2492b4ffcf22bb164e6f58628de4afcc4c8303"}]},"branch":"refs/heads/master"},"0a760840429c1f3c3fb6c89cde44198e4cbb125d":{"kind":"REWORK","_number":13,"created":"2022-08-15 21:35:56.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/13"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-08-15 21:35:54.000000000","tz":120},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nFor the fun of it, just testing around with it again.\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0a760840429c1f3c3fb6c89cde44198e4cbb125d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0a760840429c1f3c3fb6c89cde44198e4cbb125d"}]},"branch":"refs/heads/master"},"d7ec31e77b7de115b449f8681724bbb9613292e5":{"kind":"REWORK","_number":14,"created":"2022-08-15 21:40:34.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/14"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-08-15 21:40:25.000000000","tz":120},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d7ec31e77b7de115b449f8681724bbb9613292e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d7ec31e77b7de115b449f8681724bbb9613292e5"}]},"branch":"refs/heads/master"},"488063a0030c0650cac6a265b1454e2f9c224db9":{"kind":"REWORK","_number":15,"created":"2022-08-17 07:05:03.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/15"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-08-17 07:04:55.000000000","tz":120},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/488063a0030c0650cac6a265b1454e2f9c224db9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/488063a0030c0650cac6a265b1454e2f9c224db9"}]},"branch":"refs/heads/master"},"a0d1ddbba13ec6b64dffe1344476d4d630f35ea8":{"kind":"REWORK","_number":16,"created":"2022-08-17 12:22:06.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/16"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-08-17 12:22:01.000000000","tz":120},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0d1ddbba13ec6b64dffe1344476d4d630f35ea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a0d1ddbba13ec6b64dffe1344476d4d630f35ea8"}]},"branch":"refs/heads/master"},"55d633d1e0c5af0efff47aec9a53c7e400682c2f":{"kind":"REWORK","_number":17,"created":"2022-08-17 17:57:10.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/17"}}},"commit":{"parents":[{"commit":"41863867489c8a84823f2b3d73cab5585f1fb4ae","subject":"Merge \"Add quorum queue control configurations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/41863867489c8a84823f2b3d73cab5585f1fb4ae"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-08-17 17:57:04.000000000","tz":120},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/55d633d1e0c5af0efff47aec9a53c7e400682c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/55d633d1e0c5af0efff47aec9a53c7e400682c2f"}]},"branch":"refs/heads/master"},"2e4d82d70db2ab22276f8255b05e06e97afa7a3e":{"kind":"REWORK","_number":18,"created":"2022-09-22 14:03:38.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/18"}}},"commit":{"parents":[{"commit":"cd776de7798476509a946718518455e025cb17f2","subject":"[WIP] Add AsyncRPCClient and get_rpc_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/cd776de7798476509a946718518455e025cb17f2"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-09-22 13:33:13.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2e4d82d70db2ab22276f8255b05e06e97afa7a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2e4d82d70db2ab22276f8255b05e06e97afa7a3e"}]},"branch":"refs/heads/master"},"423772bfce8beb39f44b2810ad44ee56a2279cf3":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-09-22 19:07:04.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/19"}}},"commit":{"parents":[{"commit":"588f2da1d31ce560ce52bd39e26d580a71edaa7f","subject":"[WIP] Add AsyncRPCClient and get_rpc_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/588f2da1d31ce560ce52bd39e26d580a71edaa7f"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-09-22 19:07:04.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/423772bfce8beb39f44b2810ad44ee56a2279cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/423772bfce8beb39f44b2810ad44ee56a2279cf3"}]},"branch":"refs/heads/master","description":"Rebase"},"6f57130ac1c340ba4b930b2832ab5b94f1e1a338":{"kind":"REWORK","_number":20,"created":"2022-10-02 21:41:02.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/38/848338/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/848338/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/848338/20"}}},"commit":{"parents":[{"commit":"588f2da1d31ce560ce52bd39e26d580a71edaa7f","subject":"[WIP] Add AsyncRPCClient and get_rpc_client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/588f2da1d31ce560ce52bd39e26d580a71edaa7f"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-10-02 21:40:52.000000000","tz":0},"subject":"[WIP] Add NATS driver","message":"[WIP] Add NATS driver\n\nChange-Id: I0e1453a983c1c24aef74c6f6c2502871759d71ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6f57130ac1c340ba4b930b2832ab5b94f1e1a338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6f57130ac1c340ba4b930b2832ab5b94f1e1a338"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
