)]}'
{"id":"openstack%2Foslo.messaging~849099","triplet_id":"openstack%2Foslo.messaging~stable%2Ftrain~I406d379271084988a8cf06230058d69e2fd42c95","project":"openstack/oslo.messaging","branch":"stable/train","topic":"fix-gate","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-12-21 14:14:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I406d379271084988a8cf06230058d69e2fd42c95","subject":"Remove oslo.messaging-tox-py27-func-rabbit from gate queue","status":"ABANDONED","created":"2022-07-08 10:28:32.000000000","updated":"2023-12-21 14:14:47.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"27658ca239c855670d5b23eb23572c051c644681","_number":849099,"virtual_id_number":849099,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-08 11:59:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2022-07-08 14:07:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-08 11:59:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-08 14:07:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"24959ad4d0430f3741cbf885ff32537946f804e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-08 10:28:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fd13a7ef71fcea3d4354ea77a742e8be01cb9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 11:59:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eac492263f5848fb95e75ebf69490b60\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c569beada03b4edea0a645875394370a : SUCCESS in 1h 21m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/821601ad41c34f06b9dc95bad12fb757 : SUCCESS in 1h 19m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfb0968c950f493aa57ed4b0e76301bd : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baba0a3ed613444ea70d592ed209a65f : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0a5a9f59a3343f0b48ce5a474751bf0 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7016fe2e1da14c20825ddb5d60ad2109 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f54b628a74784b57811d550c2a646c76 : FAILURE in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4c551ac079465bb094c135a8b85c16 : SUCCESS in 6m 26s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/cf38db6020ce47198e0c0b9419165df2 : SUCCESS in 4m 13s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/959f51a9c0d5407a9b8787e748df9514 : FAILURE in 11m 00s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/74c52ee6d6f8418986f5bc505c83533f : SUCCESS in 4m 21s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/5755174382b84466acc482d45ade4779 : SUCCESS in 6m 14s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/85f8f1ba17f24cc9ad1e31b0394fa3a1 : FAILURE in 11m 38s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/8c593ff5653e4dbdbd9277f75845925f : SUCCESS in 4m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7a522db48cc04b6ea68724e16bef179c : SUCCESS in 1h 23m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/de50fdf741334403aeadc4469a684cfb : SUCCESS in 1h 19m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/188cfd94e5e04c8aa8ea35a8d9e9e462 : FAILURE in 17m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e7d6c63dfdcb455e8ca56c78e908ea22 : SUCCESS in 1h 25m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/5c8c3db90175450caa57173713155aa6 : FAILURE in 16m 41s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/2a2f34920c144993bf7aa2add32d88db : SUCCESS in 55m 00s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/9a8de8b71e1947b1908b7ab79e2b4998 : FAILURE in 17m 56s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/f414524536c44af7b7a8422e21b090d8 : FAILURE in 43m 03s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/678ceb9aae824340aff1b67a0427512e : FAILURE in 43m 01s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/d3fa31229c62452ba7c197769052baa9 : FAILURE in 42m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2a447d7f0426145c6f4ad5d2b9f5c02a4ba7f21","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-08 14:07:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27658ca239c855670d5b23eb23572c051c644681","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-12-21 14:14:47.000000000","message":"Abandoned\n\nTrain is about to transition to End of Life. Open patches needs to be abandoned before branch deletion.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee67e6cccbb2dc174674327e7244b8f61820f9b7","revisions":{"ee67e6cccbb2dc174674327e7244b8f61820f9b7":{"kind":"REWORK","_number":1,"created":"2022-07-08 10:28:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/99/849099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/99/849099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/849099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/849099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/99/849099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/99/849099/1"}}},"commit":{"parents":[{"commit":"1df75911900187c81c23185976e7fab8522bf558","subject":"Merge \"[trivial] fix typos (missing a white space)\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1df75911900187c81c23185976e7fab8522bf558"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-07-08 10:26:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2022-07-08 10:28:10.000000000","tz":60},"subject":"Remove oslo.messaging-tox-py27-func-rabbit from gate queue","message":"Remove oslo.messaging-tox-py27-func-rabbit from gate queue\n\nIn change I23844887722f77e827f97390fc8986f322fc7a78 we made the\noslo.messaging-tox-py27-func-rabbit job non-voting in the check queue,\nhowever, it was left in the gate queue. Non-voting gate jobs don\u0027t make\nmuch sense. Remove it entirely.\n\nChange-Id: I406d379271084988a8cf06230058d69e2fd42c95\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nStable-Only\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee67e6cccbb2dc174674327e7244b8f61820f9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee67e6cccbb2dc174674327e7244b8f61820f9b7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
