)]}'
{"id":"openstack%2Foslo.messaging~858747","triplet_id":"openstack%2Foslo.messaging~stable%2Ftrain~I29beb88dd241af72397bf0dd1fc46c98a229425c","project":"openstack/oslo.messaging","branch":"stable/train","topic":"cap-virtualenv-py37","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-02-27 12:09:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I29beb88dd241af72397bf0dd1fc46c98a229425c","subject":"[stable-only] Cap virtualenv for py37","status":"ABANDONED","created":"2022-09-21 14:36:28.000000000","updated":"2023-02-27 12:09:17.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0501e4b717c5cafee70b225e10db0a32de19167e","_number":858747,"virtual_id_number":858747,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-21 17:42:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-21 17:42:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-26 14:48:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"}],"messages":[{"id":"e37a09aea878eb863d89020d09201e5e8021aed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-21 14:36:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"358c43b0329f25df147c377a7136c94dffb8ab73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 17:42:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ecdf4e6b5c84326a1116a96944d3c8c\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/7fa1c46266fc4d659263fabd1028ac7b : FAILURE in 1h 24m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db7afc61dab04b728d98d4bc260e5642 : SUCCESS in 1h 26m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72405d98cb8d4b64accdc7ef120865e6 : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28efd7c53d364afeb7a81d3486192a7a : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b8883824bc94b5ea87f418b7ebc9a04 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1e777bb62ed47258886ee170f983bb8 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f6b4e17a1b44a02aac343cb0773f1be : FAILURE in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e54bf25f394635b1fc979440b7b3df : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3977301fd1734626b0817f4dcabdf76c : SUCCESS in 4m 49s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/d661eac8e5da448fa15d865084e288d8 : SUCCESS in 7m 20s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/a13af1ac6acb442a93318901c8e3be1d : FAILURE in 14m 15s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/c915e3bde3dd4f57974be759bd14eda3 : FAILURE in 8m 15s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/c6833b4ec21e420db8246af93fabe37d : FAILURE in 7m 36s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/fb81254708a64a7bb66f355da65c828a : FAILURE in 14m 03s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/e4dc7412d04d4f0aad46a5755bd9a7c4 : SUCCESS in 8m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bc363f15287a4c7b9324ae6cfb112057 : SUCCESS in 1h 30m 54s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e48fafd7a83147479c0bc42954bc13a1 : SUCCESS in 1h 28m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/911373eb7af64a85bf8a81fd08a15c61 : FAILURE in 30m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c16df78ab23c4b75ac13cdb08ae05384 : SUCCESS in 1h 24m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/e31f45588e6a4e1a88b3e72ff7a5cedb : FAILURE in 23m 13s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a1e5c11d0701409989dd8aef71cac247 : FAILURE in 2h 56m 09s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/2383267c8f9a4b8aa16b50343ff2fdff : FAILURE in 2h 57m 44s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/7812f949ed4f4a38a63a09aca01d26ef : FAILURE in 2h 57m 31s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/9b7813f7cccc4fce922efd40ee9d4a03 : FAILURE in 2h 56m 06s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/434309019f544cc7b3ee0eebf2af5389 : FAILURE in 2h 57m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f73257b3bca796f37536dd7bcff5e1afa2f45686","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-26 14:48:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24aaea44f3aa3ebb24ed0fe5ad51cefef254c40c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 17:56:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7438643a4cb4253a45e60e8524bd8d6\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9699beb8e5c04e11ad2a47091db9e10c : SUCCESS in 1h 26m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebb500ec331c4339bdaa31e544bf029e : SUCCESS in 1h 28m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b072e8494cb44f8bb1b9e8c757d02a86 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d25053cb6e1749f6ac4595d62097fe30 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1396a0c1e7a54b1daad3edd8e3c62aae : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a4657e771eb44f6b8cc66b1b1fb2c06 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76116cc1e0ed431fb83066c6d4764b66 : FAILURE in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8841eea57f14171997e6327f0cb13ab : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d11d60fa5c6d424b888be5fcb9b6f449 : SUCCESS in 3m 21s\n- oslo.messaging-tox-py27-func-rabbit https://zuul.opendev.org/t/openstack/build/4646bbe89c8942b3a7eec5c759f5f7e2 : SUCCESS in 5m 40s (non-voting)\n- oslo.messaging-tox-py27-func-amqp1 https://zuul.opendev.org/t/openstack/build/45e357afad3d48389b99c8b2c3d015c7 : FAILURE in 11m 49s (non-voting)\n- oslo.messaging-tox-py27-func-kafka https://zuul.opendev.org/t/openstack/build/33ef7c459f6f4b408f98a13035ce47e3 : SUCCESS in 6m 55s (non-voting)\n- oslo.messaging-tox-py36-func-rabbit https://zuul.opendev.org/t/openstack/build/4459717256a24428be909c18449989fb : SUCCESS in 5m 23s (non-voting)\n- oslo.messaging-tox-py36-func-amqp1 https://zuul.opendev.org/t/openstack/build/2efafb4775f24c2084d88e113dfb3fbf : FAILURE in 11m 57s (non-voting)\n- oslo.messaging-tox-py36-func-kafka https://zuul.opendev.org/t/openstack/build/3cfa68df9821435ea942dd2032e2ddac : SUCCESS in 7m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/91646855ccc94d3785ffd91f5be5607e : SUCCESS in 1h 33m 58s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6838f30aee084653814aa48b947e20b5 : SUCCESS in 1h 30m 07s (non-voting)\n- oslo.messaging-src-dsvm-full-amqp1-centos-7 https://zuul.opendev.org/t/openstack/build/e0364aef99134ac9968654692732effa : FAILURE in 33m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/45ffb77cceec484294c94f794ab66631 : SUCCESS in 1h 29m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-centos-7 https://zuul.opendev.org/t/openstack/build/d6987290059a4bf9b5617627bfa701db : FAILURE in 25m 10s (non-voting)\n- oslo.messaging-src-grenade-dsvm https://zuul.opendev.org/t/openstack/build/5b57aded1afd4d27baecab7eb838af3e : FAILURE in 2h 54m 57s (non-voting)\n- oslo.messaging-src-grenade-dsvm-multinode https://zuul.opendev.org/t/openstack/build/0cd2a6ae67514facb6bd8081e19e9564 : FAILURE in 2h 55m 51s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-rabbit https://zuul.opendev.org/t/openstack/build/42cac19a3e1f451d8e43f72b6739fccf : FAILURE in 2h 56m 39s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-amqp1 https://zuul.opendev.org/t/openstack/build/6e8364fb9a6746e0a701510dc36fcfb8 : FAILURE in 2h 56m 48s (non-voting)\n- oslo.messaging-telemetry-dsvm-integration-kafka https://zuul.opendev.org/t/openstack/build/e7738b0983574f55bbba4a248c03f584 : FAILURE in 2h 56m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0501e4b717c5cafee70b225e10db0a32de19167e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-27 12:09:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21388d351d75e82e0e62d6dd7fd5780561383303","revisions":{"21388d351d75e82e0e62d6dd7fd5780561383303":{"kind":"REWORK","_number":1,"created":"2022-09-21 14:36:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/47/858747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/47/858747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/858747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/858747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/47/858747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/47/858747/1"}}},"commit":{"parents":[{"commit":"1df75911900187c81c23185976e7fab8522bf558","subject":"Merge \"[trivial] fix typos (missing a white space)\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1df75911900187c81c23185976e7fab8522bf558"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-05 15:06:09.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-21 14:35:52.000000000","tz":120},"subject":"[stable-only] Cap virtualenv for py37","message":"[stable-only] Cap virtualenv for py37\n\npy37 test started to fail with latest virtualenv (20.16.4) as it\nbundles setuptools 65.3.0, which causes installation problems under\npy37. (Note: with py38 the problem does not occur).\n\nThis patch sets virtualenv\u003c20.16.4 to unblock the gate as\nvirtualenv 20.16.3 bundles setuptools 63.4.1, with which py37 gate\nworks well.\n\nConflicts:\n  tox.ini\n\nNOTE(elod.illes): conflict is caused by tox.ini cleanup patch, that was\n  added in ussuri: I13e72ec6cbd006ca6eccfe21f2a0584e86a52669\n\nChange-Id: I29beb88dd241af72397bf0dd1fc46c98a229425c\n(cherry picked from commit 766715d310fb5b9d074b2e34693fa04f585008c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/21388d351d75e82e0e62d6dd7fd5780561383303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/21388d351d75e82e0e62d6dd7fd5780561383303"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
