)]}'
{"id":"openstack%2Foslo.messaging~858934","triplet_id":"openstack%2Foslo.messaging~master~I859e4b3ba21abf236a588219986bfe2d3d4ff242","project":"openstack/oslo.messaging","branch":"master","topic":"asyncio-nats","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-04-11 21:11:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I859e4b3ba21abf236a588219986bfe2d3d4ff242","subject":"[WIP] Add AsyncMessageHandlingServer","status":"ABANDONED","created":"2022-09-22 14:03:38.000000000","updated":"2023-04-11 21:11:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41becd1f05010e8e4c0d20031ef5c0b676663a64","_number":858934,"virtual_id_number":858934,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-22 17:53:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-22 15:08:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5a569df7a919f1c881e22f27ce4a00ae01ad122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-22 14:03:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c23dafe19591ef30c0e23cd7b5b36919ff5b8055","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 15:08:48.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7229c59af814fc8aa2bb3e2fef244f5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a23e4751b1024dbd9bb8742a533c2a51 : FAILURE in 29m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c3c874703d24162b908c2b1fb66f194 : TIMED_OUT in 51m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca51056b4f764650a81f6ce8b1980828 : FAILURE in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47da16876a9c47eeba63060b9e941785 : TIMED_OUT in 41m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03526d10153e494487eadd3dcbed7709 : TIMED_OUT in 41m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90281abca5ab48a6b5897bdc8807ba67 : RETRY_LIMIT in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e88e0fabcd0b45a88e95129936049acd : SUCCESS in 5m 18s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/0e3a24bca4c2471c8f27fd8cf37d6260 : TIMED_OUT in 40m 54s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/390a5c5e1b324d7090587056f4b99fb3 : TIMED_OUT in 40m 57s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/bda5aa0ebbcf4d32b8bccc08652b3658 : FAILURE in 4m 16s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/3ea93c5729694b0e98fee999c6d0e4d2 : FAILURE in 6m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1379cfe317724b58a990f55368855264 : FAILURE in 30m 58s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/855ad6ee06ba45e18966a8a5f196b13a : FAILURE in 37m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5ae8787fd7274f17bf90405f293de864 : FAILURE in 34m 58s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b517610a666d4d9a8f30d23a5ac334e0 : FAILURE in 51m 40s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/78bb15c690db449985a6f64555737fbe : FAILURE in 1h 01m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e67b288244042c3223deb4bbfda3db5f9bf1838a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-22 16:10:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c62f4c280db34455ce3e900ab0ea4960678d4ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 17:53:59.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e89715c9928940f1ac0f25db4551b43b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbdb7cb09e3242d99d7ff0527c163e7f : SUCCESS in 1h 32m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab45b2b4ee854202b992bc19765e81c0 : FAILURE in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ac1d79ddc2244389f7050a917fd16aa : FAILURE in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59f40ab2c50448689201ac0b4c8d9334 : FAILURE in 8m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43b61e5b72ec4bc982b110122b99ba21 : FAILURE in 8m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f57ac504ed944ff8fdeb04de6d29b61 : RETRY_LIMIT in 3m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db2355e48b14d9e9d203e899e5c00d4 : SUCCESS in 7m 11s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/ab953ab121984bc3864ff8f2e0718974 : SUCCESS in 9m 05s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/b7f245eae9834ba7834e7eb2a844bfa0 : FAILURE in 10m 23s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/57349cbcf9664f089606e462c48d5109 : FAILURE in 7m 24s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/958fe78ba4024cc79392e47ff5dd9e97 : FAILURE in 7m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a5f062edb18a444eaa1e469751de23ba : SUCCESS in 1h 41m 01s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/fe36525887f2412faf3f6b6c110c4548 : SUCCESS in 1h 22m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/73cb9a915046467f8f44f218fbf8978c : SUCCESS in 1h 18m 45s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/90f45269dd4e40d8b7d9def3034a2607 : SUCCESS in 49m 48s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9539cfe95b8469aa9385c18b3f1f49d : SUCCESS in 1h 04m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e35fae9b3537f1a9bdc4eaebd6a0c5140250d202","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-22 19:08:08.000000000","message":"Topic async removed","accounts_in_message":[],"_revision_number":2},{"id":"e1db0ddf273a2bebe508d75711cc2a302f4e867a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-22 19:08:10.000000000","message":"Topic set to asyncio-nats","accounts_in_message":[],"_revision_number":2},{"id":"41becd1f05010e8e4c0d20031ef5c0b676663a64","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-11 21:11:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4e971f117ce013c4d444d0f9237748ce2e1d1145","revisions":{"35c7c7a0d9765d3b7eae0e90dcfebf9126551b5c":{"kind":"REWORK","_number":1,"created":"2022-09-22 14:03:38.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/34/858934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/34/858934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/858934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/858934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/858934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/34/858934/1"}}},"commit":{"parents":[{"commit":"e5e70a5d894b576b3de705c5f55f9704bb8e593e","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5e70a5d894b576b3de705c5f55f9704bb8e593e"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-09-22 13:17:50.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-09-22 13:17:50.000000000","tz":0},"subject":"[WIP] Add AsyncMessageHandlingServer","message":"[WIP] Add AsyncMessageHandlingServer\n\nChange-Id: I859e4b3ba21abf236a588219986bfe2d3d4ff242\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/35c7c7a0d9765d3b7eae0e90dcfebf9126551b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/35c7c7a0d9765d3b7eae0e90dcfebf9126551b5c"}]},"branch":"refs/heads/master"},"4e971f117ce013c4d444d0f9237748ce2e1d1145":{"kind":"REWORK","_number":2,"created":"2022-09-22 16:10:25.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/34/858934/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/34/858934/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/858934/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/858934/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/34/858934/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/34/858934/2"}}},"commit":{"parents":[{"commit":"e5e70a5d894b576b3de705c5f55f9704bb8e593e","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5e70a5d894b576b3de705c5f55f9704bb8e593e"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-09-22 13:17:50.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-09-22 16:10:19.000000000","tz":0},"subject":"[WIP] Add AsyncMessageHandlingServer","message":"[WIP] Add AsyncMessageHandlingServer\n\nChange-Id: I859e4b3ba21abf236a588219986bfe2d3d4ff242\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4e971f117ce013c4d444d0f9237748ce2e1d1145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4e971f117ce013c4d444d0f9237748ce2e1d1145"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
