)]}'
{"id":"openstack%2Foslo.messaging~861070","triplet_id":"openstack%2Foslo.messaging~master~Id2ca9cd9ee8b8dbdd14dcd00ebd8188d20ea18dc","project":"openstack/oslo.messaging","branch":"master","topic":"deprecate-qpid","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-08-13 10:36:35.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2023-03-25 20:02:01.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"Id2ca9cd9ee8b8dbdd14dcd00ebd8188d20ea18dc","subject":"Deprecate the amqp1 driver and Remove qpid functional tests","status":"MERGED","created":"2022-10-12 09:35:06.000000000","updated":"2023-08-15 15:33:11.000000000","submitted":"2023-08-13 10:36:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861070-deprecate-qpid","meta_rev_id":"4c67a9e8b8263b929a28516da07a52c1315b4202","_number":861070,"virtual_id_number":861070,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-13 10:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-10-25 07:35:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-11-08 10:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-11-08 10:16:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-12 09:41:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-10-12 10:44:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-10-12 11:09:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-25 07:35:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-11-08 10:16:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-12-21 06:13:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-03-25 20:02:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"}],"messages":[{"id":"4a319316d42f6a23dbcb4ad9c318b5c3f1c59ab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-12 09:35:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c208fae492fcb7a1867417dd8cf261bde40f89e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-12 09:36:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ee5c3b5d217ff8feecb7dfa3df40676520e6c39e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-12 09:41:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f94c8c9273bca01d131c9bf4abd01dd55281e296","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-12 09:45:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a48216ba48c290370b2bd6910017002d968fabf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-12 09:49:32.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf149a2a2358e2fa9049e869e08885a9d22e3552","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-12 10:44:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00cd7d8b04f9164a3bb46ec27f84ef0e28f152b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 11:09:22.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fc37f0160434189a22e5aa788e97e3f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05fc0a4a7680474dbb9f89dd90f41a4f : SUCCESS in 1h 31m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc8c4fd324e142d7aacc04cb0abea735 : FAILURE in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/450a63b64ea6460a8ee09ae3a9470207 : FAILURE in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/382af71dd7824a468f4aef588bf67725 : FAILURE in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9baf7095539942749e3d9c8b78eb92ac : FAILURE in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cd3fc54da6046cea465b8b77da0f509 : RETRY_LIMIT in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9902a021fb384e55ac432ec019083adc : SUCCESS in 4m 11s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/bdb01a97265343feb8426c5fa66b7b55 : SUCCESS in 4m 18s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/953a2d2d50ad47c884ab66a892a0edf4 : FAILURE in 6m 38s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/fd13621fa2bd41aab28ebe995773bb89 : FAILURE in 4m 43s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/b106425c57b246b3a24bc7b34147b6a8 : FAILURE in 3m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/773f9f756f4c462688cb772e18ca0bac : SUCCESS in 1h 25m 59s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5f9c4203f65340d9abd2d860b7cabe96 : SUCCESS in 1h 15m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/443fc50ef85b473ca81c569d6ee8123a : SUCCESS in 1h 17m 45s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/987c1ede4b1d47c5ae6b598556dc47f0 : SUCCESS in 50m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/50e4e0e78f1042edb7630912422c6a28 : SUCCESS in 51m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"142ac0a45d43c8883b10132a152275d41fc8663f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-14 10:13:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d52cadd519b77bfe791f77bd85052225642c1c0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-18 09:23:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d708c610f38c6e68916290e6a03e083e237dfe6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-18 09:27:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2e0608d050cfdf18da4b650ea034d0839da0699f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-18 09:28:26.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f4c8ad003cdbeec3e857532e25da5120e5472374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 11:29:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c852d289f3e546a69959545d6b5d7d6c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/baba9620cffd4bfe85e23a73d1c1641c : SUCCESS in 1h 53m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d7b35f0a9d1414fb885de9de41790a9 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd9d8d50dcba4c2685992c090e320b2c : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a3667912c92416fbd05c9d45b361b2f : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ef96df72c254abd897d4a30164c757d : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d0467006c824280a4b5a98eddef2550 : RETRY_LIMIT in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394f762b99c448cea4bc304efccb1bf8 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bafba16f558b40ea88aa4c05f9de3adc : SUCCESS in 3m 07s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/87a17d38413b4227a8dfd7554112c9ad : SUCCESS in 4m 45s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/1df11e2685a74ac9b7067fc4d58f8965 : FAILURE in 7m 02s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/1f99a105fa2f4a20892153d9cb21a601 : FAILURE in 3m 52s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/3f89e111245b4145a9de2c5a41865e57 : FAILURE in 3m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e88409f571f248f7bbc679eb19534510 : SUCCESS in 1h 23m 33s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/b5e23e86159e426bb43ab37729387d9c : SUCCESS in 1h 29m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2c0b07edae4a42d184d60ef2b82944a6 : SUCCESS in 1h 42m 17s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/06e08123ab514bd9b00c4a8741739bcd : SUCCESS in 57m 28s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e09a6324d460478a849987a4668ad7c9 : SUCCESS in 1h 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d9ba962c3b706841eef1045649bd9695eb4fea5a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-25 07:35:34.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"34763be994bfe50b6de37bf421fad81bbde24c82","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-25 07:41:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c0d702fd512ef96e1effda6f270d50ba3e54e424","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-10-25 07:42:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b64993a71e8ea23fc5777e5d928ee531825c83ca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-25 08:49:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c8fde0b54221d020f6437958e9bbad36faa0ba06","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-11-08 10:16:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"287f7880624d6132c8e1c45544d77a51c0ecc997","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 10:16:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f9203bef730df62efe273f9d225c6636e4f1164d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 11:55:18.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40dfce17a1834f1c96e4c40cc3ad52b3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97e382d61c314e64bd9ac3bd14a7270f : FAILURE in 1h 37m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a973761fe4014fb3b216d9b595b1a55d : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcbcf5d724ab4a35a445e55a1d694b7c : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2578fe3502f14000bca9917b1cc750fa : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b84bf23b4104f95ae92ab26462daeef : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2678f3af09349b1b9189acf14b0858e : SUCCESS in 3m 17s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f7994d327a9b47fc9aa4ad94ea004f28 : SUCCESS in 4m 43s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7eadd1e472a2414894481341dd3c4f50 : SUCCESS in 1h 07m 29s","accounts_in_message":[],"_revision_number":5},{"id":"b2ac1b320247781890aaffce9e51c10152dec435","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-21 06:13:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5fe4967f636d67d00fc5c4477676da28cc8ef739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 08:08:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/79964daae12941ceb5e6497503c1ffe0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ccc91bae73a0450db6fea42c5f6b2bf6 : SUCCESS in 1h 48m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6502bf441ca54f44a9359bddca85ec8e : RETRY_LIMIT in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa385b7af11e4d07a059cedeff66f9b1 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66aebc6ebce24c72a3957e8a29f71818 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa26071255844ec5a26988bece59c89a : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03d6807d38fc47dbacbec8d896028679 : RETRY_LIMIT in 2m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4241563b13724a799b469b3924d51b95 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ad4dc5472744c858c3c66f3e54d3afe : SUCCESS in 3m 06s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/41571a4ad49f4b1bb388e87d6f00dfb8 : SUCCESS in 5m 47s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/f74fe89be6074c658a0c85fbd0138326 : FAILURE in 6m 26s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/06991d2df054455d92e93940f3340aea : FAILURE in 4m 03s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/9f067b8b061e4a6bbc42582626275ebf : FAILURE in 3m 37s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9996029002874e6c92020faedacc51a9 : SUCCESS in 1h 39m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/abbe8bf9c2d44f5e8c14401bbe8e368e : FAILURE in 27m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/928a87eb00cb4770bb1f9c6fd014e7f5 : SUCCESS in 1h 38m 08s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/492edf6498be49b2a3d637a206a7773d : FAILURE in 1h 08m 48s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f44b6c990e541b2abd87c0023f28bc9 : SUCCESS in 1h 22m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"50a5c46d9d5b7298f80fee5d69e074f644860338","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-03-25 20:02:01.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_30491\u003e.","accounts_in_message":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"_revision_number":5},{"id":"a329fa1c5d9108dcfda6830da2eb7a0d7bf1bcef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-08-13 07:27:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ba6ad6e254ea6eb041318612f24981d9aef53f55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-13 08:50:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a3b775d93e04994a6cdddfda2fe0f38\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2eff297161a4104b92e37fc0ae29164 : SUCCESS in 1h 15m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d1596fcea824317aefc78cffe108b8d : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a33aebbbe674e38ab09cc76c1717c1d : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b0c7de90db54f338990cb205ff88bbe : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/183d298aa6e44e6b8b0023fc3b5966cb : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0821d8830c8437690cea6442bea5c7d : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25c932390214e08ad06d2f7f5d8aa07 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f16aefdffb2949d0b6a7dc6e1e7c8bf1 : SUCCESS in 3m 16s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/eae41ebb06454034b56faa7df9ccd5d9 : SUCCESS in 5m 36s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/8fab995f05034a2d9a32b161f6075fd9 : FAILURE in 4m 42s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/e9a347a1cf2d4f718efddde897d57955 : FAILURE in 3m 39s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/198183b6f11648e587608fe14942ad71 : FAILURE in 4m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/84f5b9e138194b35a98c3302c1d7cc29 : SUCCESS in 1h 09m 44s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a74983c5558346dd8d22b0ff5d61abfc : FAILURE in 18m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ad85ac12e53448259fb81eb40889f3b4 : SUCCESS in 1h 09m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/01e5e88171634f7bbdc484ac361ac960 : SUCCESS in 46m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/93b212627a30414e9a6e2043c57aa4e6 : SUCCESS in 1h 17m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7bb7b6c10398dbe380abdb5516cf250d2fc94af5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-13 08:51:18.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ef4d5ad8365ea65fef5f570017836dc735ec72a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-13 10:36:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e612421091d4118959c5c8d70863d6d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12ef98c19889478dad4bb4c61971385b : SUCCESS in 1h 40m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b37fc0a38f3f4c0d86399abebb57e5c6 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd70a3fa9994cb59a9cd46103a3deb8 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51e25c27656d4fdfb5d1a00abed6d1c8 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00491016a87c4e009dc9212ba7cb0d92 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bfda36733d649e5a306bfe56d7b798d : SUCCESS in 3m 19s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/ea0c4a6309534d8b8f19c2f702264dda : SUCCESS in 5m 06s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/0bf3acd8bed44d7fb95877563fb7a5be : SUCCESS in 1h 12m 08s","accounts_in_message":[],"_revision_number":5},{"id":"ee58bc90046c5ba83c408b9493a2312b9e8d73d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-13 10:36:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"bef08c1883b86158bf479c5d4a0ab2713ba7b43c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-13 10:37:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05ef99e48ef34d3d96ddc7282e23997a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8954970ca3354b71a4f84ac233a3083a : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/531cf8fc42c944d8aa62f04af80249d0 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":5},{"id":"5c914e38c2a807150fa69a0327a41f9bb5e4b63c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-15 02:03:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4c67a9e8b8263b929a28516da07a52c1315b4202","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-15 15:33:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0f63c227f5425995ae8c61f1d40ec85e7728528a","revisions":{"e8a9c827183e6fb3b0e67863fdb5915549137c56":{"kind":"REWORK","_number":1,"created":"2022-10-12 09:35:06.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/861070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/70/861070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/1"}}},"commit":{"parents":[{"commit":"e5e70a5d894b576b3de705c5f55f9704bb8e593e","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5e70a5d894b576b3de705c5f55f9704bb8e593e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-12 09:21:00.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-12 09:35:04.000000000","tz":120},"subject":"Remove qpid scenarios and tests","message":"Remove qpid scenarios and tests\n\nA recent oslo.messaging patch [1], not yet merged, who aim to update the\ntest runtime for antelope lead us to the following error:\n\n```\nqdrouterd: Python: ModuleNotFoundError: No module named \u0027qpid_dispatch\u0027\n```\n\nThe qpid driver has been deprecated and removed since a while.\n\nFunctional tests remains. However, note that neither debian nor ubuntu\nin the latest releases have any binary built for it, not even 3rd party.\nOnly qpid proton, the client lib, is available.\n\nThese changes propose to remove these functional tests to unlock our\ngates for antelope.\n\n[1] https://opendev.org/openstack/oslo.messaging/commit/10c39d5df4d71e06962db1d2563c3915519fc9a9\n[2] https://review.opendev.org/c/openstack/oslo.messaging/+/856643\n\nChange-Id: Id2ca9cd9ee8b8dbdd14dcd00ebd8188d20ea18dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e8a9c827183e6fb3b0e67863fdb5915549137c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e8a9c827183e6fb3b0e67863fdb5915549137c56"}]},"branch":"refs/heads/master"},"715a918fc5c4e24f32dae48348c86859e8dc54eb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-12 09:36:17.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/861070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/70/861070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/2"}}},"commit":{"parents":[{"commit":"e5e70a5d894b576b3de705c5f55f9704bb8e593e","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5e70a5d894b576b3de705c5f55f9704bb8e593e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-12 09:21:00.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-12 09:35:30.000000000","tz":120},"subject":"Remove qpid scenarios and tests","message":"Remove qpid scenarios and tests\n\nA recent oslo.messaging patch [1], not yet merged, who aim to update the\ntest runtime for antelope lead us to the following error:\n\n```\nqdrouterd: Python: ModuleNotFoundError: No module named \u0027qpid_dispatch\u0027\n```\n\nThe qpid driver has been deprecated and removed since a while.\n\nFunctional tests remains. However, note that neither debian nor ubuntu\nin the latest releases have any binary built for it, not even 3rd party.\nOnly qpid proton, the client lib, is available.\n\nThese changes propose to remove these functional tests to unlock our\ngates for antelope.\n\n[1] https://opendev.org/openstack/oslo.messaging/commit/10c39d5df4d71e06962db1d2563c3915519fc9a9\n[2] https://review.opendev.org/c/openstack/oslo.messaging/+/856643\n\nCloses-Bug: 1992587\nChange-Id: Id2ca9cd9ee8b8dbdd14dcd00ebd8188d20ea18dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/715a918fc5c4e24f32dae48348c86859e8dc54eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/715a918fc5c4e24f32dae48348c86859e8dc54eb"}]},"branch":"refs/heads/master"},"21916972e11fa6e445e457b7b2d1ef7f7ee9bf48":{"kind":"REWORK","_number":3,"created":"2022-10-18 09:23:28.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/861070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/70/861070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/3"}}},"commit":{"parents":[{"commit":"e5e70a5d894b576b3de705c5f55f9704bb8e593e","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5e70a5d894b576b3de705c5f55f9704bb8e593e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-12 09:21:00.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-18 09:18:23.000000000","tz":120},"subject":"Deprecate the amqp1 driver and Remove qpid functional tests","message":"Deprecate the amqp1 driver and Remove qpid functional tests\n\nA recent oslo.messaging patch [1], not yet merged, who aim to update the\ntest runtime for antelope lead us to the following error:\n\n```\nqdrouterd: Python: ModuleNotFoundError: No module named \u0027qpid_dispatch\u0027\n```\n\nNeither debian nor ubuntu in the latest releases have any binary built for it,\nnot even 3rd party. Only qpid proton, the client lib, is available.\n\nTo solve this issue, these changes propose to deprecate the AMQP1 driver\nwho is the one based on qpid and proton, and propose to remove the\nrelated functional tests.\n\nThe AMQP1 driver doesn\u0027t seems to be widely used.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/856643\n\nCloses-Bug: 1992587\nChange-Id: Id2ca9cd9ee8b8dbdd14dcd00ebd8188d20ea18dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/21916972e11fa6e445e457b7b2d1ef7f7ee9bf48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/21916972e11fa6e445e457b7b2d1ef7f7ee9bf48"}]},"branch":"refs/heads/master"},"bd6f0cfdb69db6c920cf48d1d0087b295149810c":{"kind":"REWORK","_number":4,"created":"2022-10-18 09:27:04.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/861070/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/70/861070/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/4"}}},"commit":{"parents":[{"commit":"e5e70a5d894b576b3de705c5f55f9704bb8e593e","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5e70a5d894b576b3de705c5f55f9704bb8e593e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-12 09:21:00.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-18 09:26:55.000000000","tz":120},"subject":"Deprecate the amqp1 driver and Remove qpid functional tests","message":"Deprecate the amqp1 driver and Remove qpid functional tests\n\nA recent oslo.messaging patch [1], not yet merged, who aim to update the\ntest runtime for antelope lead us to the following error:\n\n```\nqdrouterd: Python: ModuleNotFoundError: No module named \u0027qpid_dispatch\u0027\n```\n\nNeither debian nor ubuntu in the latest releases have any binary built for it,\nnot even 3rd party. Only qpid proton, the client lib, is available.\n\nTo solve this issue, these changes propose to deprecate the AMQP1 driver\nwho is the one based on qpid and proton, and propose to remove the\nrelated functional tests.\n\nThe AMQP1 driver doesn\u0027t seems to be widely used.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/856643\n\nCloses-Bug: 1992587\nChange-Id: Id2ca9cd9ee8b8dbdd14dcd00ebd8188d20ea18dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd6f0cfdb69db6c920cf48d1d0087b295149810c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bd6f0cfdb69db6c920cf48d1d0087b295149810c"}]},"branch":"refs/heads/master"},"0f63c227f5425995ae8c61f1d40ec85e7728528a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-10-18 09:28:26.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/70/861070/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/70/861070/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/70/861070/5"}}},"commit":{"parents":[{"commit":"e5e70a5d894b576b3de705c5f55f9704bb8e593e","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/e5e70a5d894b576b3de705c5f55f9704bb8e593e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-12 09:21:00.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-18 09:27:46.000000000","tz":120},"subject":"Deprecate the amqp1 driver and Remove qpid functional tests","message":"Deprecate the amqp1 driver and Remove qpid functional tests\n\nA recent oslo.messaging patch [1], not yet merged, who aim to update the\ntest runtime for antelope lead us to the following error:\n\n```\nqdrouterd: Python: ModuleNotFoundError: No module named \u0027qpid_dispatch\u0027\n```\n\nNeither debian nor ubuntu in the latest releases have any binary\nbuilt for the qpid backend, not even 3rd party. Only qpid proton,\nthe client lib, is available.\n\nTo solve this issue, these changes propose to deprecate the AMQP1 driver\nwho is the one based on qpid and proton, and propose to remove the\nrelated functional tests.\n\nThe AMQP1 driver doesn\u0027t seems to be widely used.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/856643\n\nCloses-Bug: 1992587\nChange-Id: Id2ca9cd9ee8b8dbdd14dcd00ebd8188d20ea18dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
