)]}'
{"id":"openstack%2Foslo.messaging~862563","triplet_id":"openstack%2Foslo.messaging~master~I9e042732b1d512bb4bebff7a72ea2063a25d2926","project":"openstack/oslo.messaging","branch":"master","topic":"deprecate-qpid","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-01-18 16:34:27.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-18 12:45:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-01-18 16:34:27.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-01-16 13:31:31.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}}},"hashtags":[],"change_id":"I9e042732b1d512bb4bebff7a72ea2063a25d2926","subject":"cleanup amqp1 scenarios remnants setups","status":"MERGED","created":"2022-10-25 08:46:34.000000000","updated":"2024-01-18 16:35:32.000000000","submitted":"2024-01-18 16:34:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862563-deprecate-qpid","meta_rev_id":"2b306f44b3850eeeecfbb12a4a0dd7237a9743b7","_number":862563,"virtual_id_number":862563,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 16:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2024-01-18 03:07:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-16 13:31:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2024-01-18 03:26:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2024-01-18 03:07:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-25 10:39:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-08 10:16:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-12-21 06:14:20.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-08-15 15:37:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-01-15 16:08:39.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-01-18 03:26:27.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"e12b8c0e88bec95327afc53be87d9f3969abaad5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-25 08:46:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf350fa8d7e633f260052a5d5512013f064671cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-25 08:47:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dc803e38d78c9bedae3e851faa4901ed71c6fe42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 10:39:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c404d601e1a14667ac61afe3b47f057d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae3d8fb0d2964f7198af7140422a20a0 : SUCCESS in 1h 35m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6cc44013353401a90fdc55c5108fa4d : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6bbc3c996e44820965ddb822979a2ba : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2337d5fa461f4abfa6bf642fc2a04f14 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3f6aa82e00d4531817796bd364a4740 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23bef68b22a34655bbc70145aee5043e : RETRY_LIMIT in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e059cdc07e94398b28bbf464619c9c0 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58a0647a5fce406fbe3c61bb3401033c : SUCCESS in 3m 09s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/5969307a0ac742fa836136fce461e906 : SUCCESS in 4m 32s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/c42dcfd4e1904d22bb33473ab428b978 : FAILURE in 9m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/c014cf8cfd49492eb92831c77b22c865 : SUCCESS in 1h 41m 11s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/49844c92d77548598d4e1fa9dcedd7b3 : SUCCESS in 1h 35m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a5d60de58049410c873ee72c69dc1627 : SUCCESS in 1h 46m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/cc60fe86c54f458e81c9397c48af987d : SUCCESS in 57m 56s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/21d3cb08c257430f906aa3adaf006af1 : SUCCESS in 1h 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a12a52749cf88feb9527fbe538af8c3b3e170d59","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-11-08 10:16:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"652b5a8061b4320d7d27907ea6ccb15e5625f733","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 10:17:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a794802005d9bed8515b3949e85b1999e3c4a535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 11:55:20.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/861070 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"ec8aa1cbc3dc5c30b53eb386fda23c14ebeb3adc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-21 06:14:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7995837e1d216327774e79828ed283246e818a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 07:59:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/559e213944a24ea190bb3b990664ae71\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/196515ecf5e74fb489b19b16aa150b27 : SUCCESS in 1h 41m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe10ac236a64458fb0e7dc02aa998bed : RETRY_LIMIT in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa643972ad81493597f279f81bba560a : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/947962fcf414458195fbed0ea6e2c9e2 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f76458721f6a4b01bc4859d2a726f7d3 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/884ef098fda7408291a4e4866841e7ef : RETRY_LIMIT in 2m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9692f39a56b4b1e8221acdf78a098be : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/229f7eeb594c4c098e459661a4173221 : SUCCESS in 3m 45s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/615dbaa632dd4044aaa4d778ffa530ae : SUCCESS in 4m 58s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/5f947d4d86934220a77d2d52d1ddc616 : FAILURE in 6m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/6a536ae35db747a9a060daa118b6ddb1 : SUCCESS in 1h 11m 42s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ed81eaadac55493e887c956ef37b7273 : FAILURE in 16m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/af462902f200448c82f5b8350160c784 : SUCCESS in 1h 43m 44s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/14117c16e99f4a94956737434ffe02a2 : SUCCESS in 1h 09m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea424835e0034e7885720171aed847ae : SUCCESS in 1h 18m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1da53c1c24b74e9b15351ddb62b9a5554b2c459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-15 15:36:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6829890ccc88588d0c33352d01d71f96873b8840","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-15 15:37:04.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fac01c023dc1ff4698cac8f7d66d4350218f52a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 19:28:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d17b5f6cc5d841d59eb5e62d3f05a3f2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45718c21f273478d9fcc3d37cb387ee4 : FAILURE in 59m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df0906a9ea47470bb3b9f1bac1aab7e5 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94cb3e98ba8a43c7a669d71ce2a57cb6 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af182603e42c4d11a0184877972b6f1b : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/894f0d997f3e4231b5d484d7ea680721 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2375bbe979544b2da0bcb9a9ad12d1e4 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e867a5e250be4ad298dd195751059b57 : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21ba3ec3138a4b1b9dd6da19c7c20414 : SUCCESS in 3m 29s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/4d91879eaca84918a927c8e97fe08f55 : SUCCESS in 4m 23s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/c5ca39e5d2244d0196909bc22203b2d7 : FAILURE in 4m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fa59038478734801a09fcc8693a2acd8 : SUCCESS in 1h 45m 29s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4f0165aa9e6d48309b06d39090f42cb9 : FAILURE in 19m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0b9eaa32e6dc4b1eab50e5674f7b71c5 : SUCCESS in 1h 34m 12s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9e62b580af504d7cb0d216f70d3e5701 : SUCCESS in 1h 01m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4a3f96e12e7468fb2fa4f2e776a5756 : SUCCESS in 1h 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d1b4607d49b97cade181336f12065a092a9d479d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-15 09:46:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8039e27e91be69e631dc04a1de44d7362ca000c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 11:42:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8a0ada3a30d48e09808b0042caac2f8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/810fcf2dddfa45dc8a2625cc5f7e3f8e : SUCCESS in 1h 39m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b92545dd2c80480baa1c9e6ee3f4ff45 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7426888478f4f0383a494356742a3db : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ebafaebebb34ded85299c72a1d84b97 : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/46372620fd5b4ea8b4f15eaca711decf : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ff71d2120948f6a63440c09285652a : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27d381b5539e4cc186db4738d2bedd3f : SUCCESS in 3m 27s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/179d1c3c630f4717a58fb9d8c7376e1e : SUCCESS in 4m 52s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/de790e725a04475d9b8f238fcecdb395 : FAILURE in 8m 28s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/c8ceebacca29416388b0fc8baa318d31 : FAILURE in 4m 13s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/6d0e2c2452f846cc896c54a5c2b0ddc7 : FAILURE in 3m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b3d4cb5d5d6443d786a8c73847a99231 : FAILURE in 1h 37m 52s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/cef34173d35340d0a9791e5b9d9514a3 : FAILURE in 29m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/cd4dbd306d3e408cb3fe8f785ff75c37 : SUCCESS in 1h 39m 10s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d5e7aadc4d214ae4846dd47c44cced5a : SUCCESS in 50m 39s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c318dfd5a9954abd92bca54fe292dd2f : SUCCESS in 1h 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"20e6b6a0aa7ac0a95cad24decbeee7c79f944416","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-15 13:10:40.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"73183b43c42a7e5b7e20f0a083e393146bf34437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 15:04:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cb80501fdbf4125ac94742fd3a54f9e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0ea239329dc4303b51285d6b804618a : SUCCESS in 1h 52m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af58b2224d1f41cdb4479bc4377e302c : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c1519c97804869a0c9ad1983ad283a : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92ea39c725304bbf81d2d1253e6fb73b : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f5e4a6626964d0fbbcb8d7f64cd1ed2 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a28d5f5aaafc44e99c8bb1a0b1212b8f : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e259b4f9f6074f03849382841f136bd8 : SUCCESS in 3m 09s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ca4c6aaa37b24a15acb76e1f6840548e : SUCCESS in 4m 46s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f9bddaaca315455583ae8b01ac834437 : FAILURE in 7m 45s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/27fa424247c94d6b95fdc556c235f7a5 : FAILURE in 4m 03s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/b9c75ac107f5482d997499f567adc7fe : FAILURE in 3m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1c51c1e5a5fe450dbf0c9eb6f5d29558 : SUCCESS in 1h 40m 13s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6fd6ee1ce0114079962093282bf34066 : FAILURE in 27m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/1acf4e4eb03d4923b06a3ddc15155957 : SUCCESS in 1h 41m 00s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c0ece685c2aa431cbf0f9906e79e787f : SUCCESS in 53m 54s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/16635447db94414f8f35cda7882f6e76 : SUCCESS in 1h 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cf9e34b7f5597457382e13d5ab61d1fd8f55c1ef","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-15 16:08:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9046a946988d680b81aa9a163d8fc57ec5f02d60","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-16 13:14:34.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"062247ee5048c06f4eed5faca85e6a7a33194213","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 13:15:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c86eee0b8d9a9ed672a588431552ba86953dec15","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-16 13:16:10.000000000","message":"Patch Set 4: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"909717d962eb2b8606479bd83ceecafa5ff7a26f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 13:16:22.000000000","message":"Patch Set 4: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":4},{"id":"ea2efb2f334563be168fe99908e1282f26d2b873","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-16 13:17:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c643ba29971bc4cf3ed51dbcec604ad7a6bd03d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-16 13:18:35.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n","accounts_in_message":[],"_revision_number":5},{"id":"f253b84ea285c703538e96aedf10caff8ba59b02","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-16 13:18:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1e610898bcdac2961c81b4d001793d83e08ec952","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-16 13:31:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"448bd3cd0392437c43f7078977869b4376290d50","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-16 13:31:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e650a8b4cb51eb443c837e2a6eca2301d2f0be7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:12:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e7e6129c63f4fb7b6746ff5e8305496\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2d9369119cd4785bc5ba112b6bf548b : SUCCESS in 1h 45m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13746d743eef4000a2d75592d1e40a75 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5d010405ccf4339866fa81579300d9c : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a41cb7d9aca346399c628f78b774d13f : SUCCESS in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e599ccdc8e0343a885b218ffbe2d3801 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e85ada2a881470cb193dbf7f44f4eaa : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfe5b2d27c6a43b6ac3d383643d59d90 : SUCCESS in 3m 05s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/08be4f68026a4ed08a89717a52c7666a : SUCCESS in 4m 25s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/0447b735e15c4ca186606ceeb2108298 : FAILURE in 6m 31s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/05fb19e3d986431181fc3be2f0600a60 : FAILURE in 3m 09s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/6c4e64943481414e9da238d06ba721a3 : FAILURE in 2m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9103e6c58f0f409ab98bcc2ae86e56ba : SUCCESS in 1h 24m 16s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/16788b181fd44f12ae265104f415eb51 : FAILURE in 18m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/01b2d1f7e5074362a3018bbd1183dbeb : SUCCESS in 1h 16m 01s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/e830381d37714c8ebc00549060570f72 : SUCCESS in 53m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/89cc401dcfd24aef8413f3fe3a12d3e7 : SUCCESS in 1h 22m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"93db6d5bf1cbd6007ba39c538ab687f2dd49a644","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 03:07:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9a40890b1d1eee8abc102970c4fb3d864a644313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 03:07:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d620daff8411aebd6ba8595c32e9a5296fc0862c","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-01-18 03:26:27.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"253ddbd7fe2b84b72ce766d4d24a42ffd900c1aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 04:50:22.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af2fb5ecb279434fb8cb202b8180e0a6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83daf39a2db340a78b5b802d738bc659 : SUCCESS in 1h 35m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45906d7fd9bc49e2bfb28ea0419f8925 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a598d768c4e455fab727bb08e26e6bd : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/091776bd80bb4069a671ecd5d9aa4ec4 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4be942844c492a9fc9a1b24c9fa19b : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f988253f9f174f738a1dad0d1d42ec8c : SUCCESS in 3m 49s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/9cc0082ed1a54d14a96ac76036554e29 : SUCCESS in 5m 14s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1a6c4de73c804ef8b842285c415313ec : FAILURE in 1h 27m 12s","accounts_in_message":[],"_revision_number":5},{"id":"a6686ad5c3acdf19959ef500946fd06e1dc6f36f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 08:43:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0cf010ad6c245c43e4ace2d4631ab662c5d30a47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:27:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea48279cd9344580a81127281d3f654d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ab54672aeff46ae821a9594024b469e : SUCCESS in 1h 36m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df040304a7cb4df8a9fbc226c839c260 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f788857a4f8e4105b3a5b7e7b8dbdd12 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f71e13ea3fb84b01a637789b7994947d : SUCCESS in 7m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2cf1429a49d451dad42c217422ce9f6 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50b44c480eb49aba51bcb864565e558 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa29ffecf6084ac1a65728532472e52a : SUCCESS in 3m 36s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/ed8349bb0ec340a58dea0efe544106c4 : SUCCESS in 5m 06s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/35a6d079e2c9475d9b9bf44ff569b46f : FAILURE in 6m 18s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/425605bd5ec841f89d7e3582d5d8db34 : FAILURE in 2m 49s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/49e6e3a0c28546b6a7da2410f78e1b3d : FAILURE in 2m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ba16731931b14a9fa790578be9ebac13 : SUCCESS in 1h 30m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/70948441c40b4229a4281620ed4d9612 : FAILURE in 18m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/4c7bdfbeb2244543a4c4b20d0334230d : SUCCESS in 1h 38m 34s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/90d777989d6c4244afa8c617497626be : SUCCESS in 52m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/95233a893130424698b734a3eb75d990 : SUCCESS in 1h 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"308a9248efb66ffff4f86b1d9d333616bd8ab043","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 10:28:01.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6f4819b08971dde0a4d9a776b519c06b82fdc8be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 12:15:33.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15a8487e7041423ead39c224c945a4b2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb041baf286c45b5a84e3f65290a07d6 : SUCCESS in 1h 41m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7fcb48ab589405081cd46478f707fb0 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26239edd4bc444d2941834ea3dd8c756 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cda2a64ebe8f4ec2b4519e90afaf3b53 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57224b0c569b4f869631e7db75deb2eb : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/509e5ba7cfd1478aa86e450a90aab467 : SUCCESS in 3m 11s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/46ba63be74a04fda9f907055cfaed566 : SUCCESS in 5m 09s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/faef3210711445d4a7cf932d197344bf : FAILURE in 1h 45m 51s","accounts_in_message":[],"_revision_number":5},{"id":"8ab0dada48259dcdb597acc99af890314cd1af02","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 12:45:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"41fbddc974341e4102a389449f64c95c5265cdb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 14:59:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2e1cc99ea904c5f862d346be0edba2c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f26e86609394e5f99943f38d02488a2 : SUCCESS in 1h 39m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ef63cbf7be34065be73a4e1711d165d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8327ce017da34851890a42866f4e9157 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef70a2529f2f4e5abeb3b3c8cc5ae9ba : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cfa1ace474ea4b7091aacb7e46e54ba3 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507b3a6261584bde9cbb80e33857bec0 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b0a04e1fb734193824d9d96d35af3e6 : SUCCESS in 4m 04s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/7d8fc9f0d65a407d86f23e927df906ce : SUCCESS in 4m 26s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/63750f328ee6427b8368745cd6167c29 : FAILURE in 5m 50s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/fe60d3b10b1f427585e9d2ef412f6b7e : FAILURE in 2m 37s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/0fd3eb86b02a4c7aab63c2950204c1b6 : FAILURE in 2m 30s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/083ff6837aa346468a623f064e82e9a8 : SUCCESS in 2h 07m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/09b8ffe5297e493fb68926906e287296 : FAILURE in 28m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/78e62164ff9149f9a64feb1fe479e4af : SUCCESS in 1h 35m 42s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4d474ba9a58b4786884f44dba15ea4c3 : SUCCESS in 1h 04m 38s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7bf7b315e7f414b90bc057122b3c4c0 : SUCCESS in 1h 17m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e60243f123d4d72f6d62d0553cc3bf9854b56579","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:00:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d4db2f403993d5d18ead4cd5ede8a8c25622a1b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 16:34:26.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e90984d7d38e4e4da6e67d9071776005\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1cbc38ea0f474ac8821036b2154715b9 : SUCCESS in 1h 29m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b891a22117c54728a82444537c3f1a73 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c953b8a8d19948d49766ad985a9098e2 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df3600b77d06411b8ebc6c87766f238f : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e66d7c272a475f8b83aebe04f7d090 : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23767098801d478b9a05a8e9a77f1975 : SUCCESS in 3m 20s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/9234ec752ca44babbb74c59beeb07aa8 : SUCCESS in 4m 48s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/886a882932fe43549834f299ae3001e2 : SUCCESS in 1h 29m 16s","accounts_in_message":[],"_revision_number":5},{"id":"7c12b9e8dee156f3341faedc870f2c36ced69961","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 16:34:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"2b306f44b3850eeeecfbb12a4a0dd7237a9743b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 16:35:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd1735d450234e0b95fee4043f400220\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b388546b65a44e6e91ef8505b5c9c5e2 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38bc8669cc1b42d9a456c3749fde073d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6e7a5725fa90e66c060a06a8ffe5e5454fd7a7b6","revisions":{"acf8faec20ee724a63f4d291ff2409e88fc23b86":{"kind":"REWORK","_number":1,"created":"2022-10-25 08:46:34.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/63/862563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/862563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/1"}}},"commit":{"parents":[{"commit":"0f63c227f5425995ae8c61f1d40ec85e7728528a","subject":"Deprecate the amqp1 driver and Remove qpid functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-25 08:46:19.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-25 08:46:19.000000000","tz":120},"subject":"cleanup amqp1 scenarios remanants setups","message":"cleanup amqp1 scenarios remanants setups\n\nChange-Id: I9e042732b1d512bb4bebff7a72ea2063a25d2926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/acf8faec20ee724a63f4d291ff2409e88fc23b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/acf8faec20ee724a63f4d291ff2409e88fc23b86"}]},"branch":"refs/heads/master"},"57d13fc134552603e54858afb7d021b14a738a86":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-25 08:47:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/63/862563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/862563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/2"}}},"commit":{"parents":[{"commit":"0f63c227f5425995ae8c61f1d40ec85e7728528a","subject":"Deprecate the amqp1 driver and Remove qpid functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0f63c227f5425995ae8c61f1d40ec85e7728528a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-25 08:46:19.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-25 08:46:58.000000000","tz":120},"subject":"cleanup amqp1 scenarios remnants setups","message":"cleanup amqp1 scenarios remnants setups\n\nChange-Id: I9e042732b1d512bb4bebff7a72ea2063a25d2926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/57d13fc134552603e54858afb7d021b14a738a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/57d13fc134552603e54858afb7d021b14a738a86"}]},"branch":"refs/heads/master"},"69a804330814c09515f580912b91dbef08f718da":{"kind":"REWORK","_number":3,"created":"2023-08-15 15:36:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/862563/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/862563/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/3"}}},"commit":{"parents":[{"commit":"fa1563004134e32b137605709bd7755986883025","subject":"Merge \"Deprecate the amqp1 driver and Remove qpid functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fa1563004134e32b137605709bd7755986883025"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-25 08:46:19.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-15 15:36:33.000000000","tz":540},"subject":"cleanup amqp1 scenarios remnants setups","message":"cleanup amqp1 scenarios remnants setups\n\nChange-Id: I9e042732b1d512bb4bebff7a72ea2063a25d2926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/69a804330814c09515f580912b91dbef08f718da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/69a804330814c09515f580912b91dbef08f718da"}]},"branch":"refs/heads/master"},"eb8e56ee334fa423b511b7d3dddd8cec38a97ac2":{"kind":"REWORK","_number":4,"created":"2024-01-15 09:46:25.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/63/862563/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/862563/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/4"}}},"commit":{"parents":[{"commit":"ee5b35174236708842945d08f7e5c29c621b7921","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee5b35174236708842945d08f7e5c29c621b7921"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-25 08:46:19.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-01-15 09:45:52.000000000","tz":60},"subject":"cleanup amqp1 scenarios remnants setups","message":"cleanup amqp1 scenarios remnants setups\n\nChange-Id: I9e042732b1d512bb4bebff7a72ea2063a25d2926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/eb8e56ee334fa423b511b7d3dddd8cec38a97ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/eb8e56ee334fa423b511b7d3dddd8cec38a97ac2"}]},"branch":"refs/heads/master"},"6e7a5725fa90e66c060a06a8ffe5e5454fd7a7b6":{"kind":"REWORK","_number":5,"created":"2024-01-16 13:18:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/862563/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/63/862563/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/63/862563/5"}}},"commit":{"parents":[{"commit":"ee5b35174236708842945d08f7e5c29c621b7921","subject":"Update python classifier in setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee5b35174236708842945d08f7e5c29c621b7921"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-10-25 08:46:19.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-16 13:17:57.000000000","tz":540},"subject":"cleanup amqp1 scenarios remnants setups","message":"cleanup amqp1 scenarios remnants setups\n\nChange-Id: I9e042732b1d512bb4bebff7a72ea2063a25d2926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6e7a5725fa90e66c060a06a8ffe5e5454fd7a7b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6e7a5725fa90e66c060a06a8ffe5e5454fd7a7b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
