)]}'
{"id":"openstack%2Foslo.messaging~866616","triplet_id":"openstack%2Foslo.messaging~master~I2aea9cbbf5136ebe1c6ef10ce94faf16e5d5c47c","project":"openstack/oslo.messaging","branch":"master","topic":"ripout","attention_set":{},"removed_from_attention_set":{"12932":{"account":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"last_update":"2023-05-22 13:00:15.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-01-13 11:31:50.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I2aea9cbbf5136ebe1c6ef10ce94faf16e5d5c47c","subject":"Deprecate kombu_reconnect_delay setting and remove associated code","status":"ABANDONED","created":"2022-12-05 18:10:15.000000000","updated":"2023-05-22 15:00:55.000000000","total_comment_count":16,"unresolved_comment_count":9,"has_review_started":true,"meta_rev_id":"36d81f26085aea08064630b28573a8e8bca8dcf3","_number":866616,"virtual_id_number":866616,"owner":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-01-10 22:02:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2023-01-13 11:31:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-05 20:10:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-06 14:38:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-10 14:40:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-04-11 11:55:44.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-05-22 10:57:14.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-05-22 15:00:55.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REMOVED"}],"messages":[{"id":"8ab876f5466bae7002cc81c88feb4240f470de51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2022-12-05 18:10:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58acc4959ef56c1b2e5f600625711abaf70e8fda","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2022-12-05 18:13:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a65756fb25b60607767e4395b77604b257ca32dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-05 20:10:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0c7e02c7d92447489328d7f5ee418a5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4ea3a7bac9b4648b8669a3b4f3ed308 : SUCCESS in 1h 54m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cea8d512e84d4343a8f84cffd3bac8f4 : RETRY_LIMIT in 2m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a498e12c16da46b9b4df54aff4ba8f36 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acf30553a19c4d16b8d5619e245a65bb : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d4fdfdf90f843f4b3738eeaaea66f2f : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d8c392b918e47aaa08c051f794da1d3 : RETRY_LIMIT in 2m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d5790a85be843109c80386e642bb5ec : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9982aa445e043f499b52d13e0e4d968 : SUCCESS in 5m 06s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/b7ed577cb69048958197dca487f25de4 : SUCCESS in 5m 05s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/8bf7299f39c24dfaab412a4bbb2a386c : FAILURE in 6m 22s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/5bcdbbc0841348ada7c06b3a7ea80e0d : FAILURE in 3m 55s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/d63cb60490e3483cafddaa86c513ef7e : FAILURE in 4m 42s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4bb41fd82f6d4b998c7cd616e4faff05 : SUCCESS in 1h 15m 16s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f72b2c8ff1a24f2da9f9121b502fe952 : FAILURE in 26m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/909e87fce8c8419989ca99a1b6420795 : SUCCESS in 1h 39m 02s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fb4aa946f67441039af7152bd8ee166c : SUCCESS in 51m 07s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b488a503fdbc4104ae95bbadbaa05eba : SUCCESS in 56m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28740bfa961b3e7109defdb3a42668db9f2577ec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-12-06 14:38:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8bda3f0830e708e7253feac6d8886f50ae06f86b","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2022-12-06 16:46:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"574772f3ccda90e5d054d122bfe68af9b8e96094","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-10 14:40:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c9e5b8a659570a6de3d9f1f4ad0fffbf0b104ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-01-10 20:18:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2b23c17ea69eb03d61feae9e4b17d284041ee58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 22:02:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae5a1cdbaa654427804d2ec175c138d5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf11989335b749069b5beb1e6e0e084b : POST_FAILURE in 1h 31m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fafb85530112451982b6d0cb2648f992 : RETRY_LIMIT in 2m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e39654b6c19403cbc0f8ef8a807e9a9 : FAILURE in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcbaa1f8dc2b4242b19771d752879502 : FAILURE in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc30be874f6441f29a322c99dfe28378 : FAILURE in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4111d27aef56472e8f2abf57d6b60ffe : RETRY_LIMIT in 3m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08a0879a7654aa4b00e23cf329e1bc1 : FAILURE in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0680104a14ad4e43b9020742ced140b9 : SUCCESS in 3m 15s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/cbffac4968144b79a621dc9122074f1d : FAILURE in 3m 03s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/cbb83c02364144818bcd3f0fad5d9e8a : FAILURE in 3m 54s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/c0ed01802fa8435a85d779bce8c4657a : FAILURE in 3m 37s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/fb7647dc259f43ab8952c66d5ad75de7 : FAILURE in 3m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/94745146424048eba94e90bf428c9a3e : SUCCESS in 1h 09m 38s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/99b51de12fdd41e0b0e48a90dbca0a86 : FAILURE in 14m 49s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/6a46485d7c7a4eb098282e8673ce7f58 : SUCCESS in 1h 38m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/9fe74d66cf7449e0b88579b477f688bd : SUCCESS in 57m 40s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c918c2744b494ab9be0033c2c1ccffa8 : FAILURE in 1h 02m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6a35f733e566d1f126ec4114f4514d73d8e1be98","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-01-11 14:24:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e6a20204b8eecfdf36ef5ba6bae86aa7308cd4c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 16:19:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43ace668c4234ce0bb3c627d2a7f7fb8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2521d2a3d5a449aac7626e4aed7560c : SUCCESS in 1h 40m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9004fc9c46cf432b8a3f572d9ebc3aa6 : RETRY_LIMIT in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e903416ae46c4ee0aae91cda610220ce : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d68c5b0984bd4027a9b548cc6a4623f4 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c48913ae838e41fbb92f84d12a9d464a : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c034d5d15194d3b8d78009fcfca3b80 : RETRY_LIMIT in 3m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b8ff2e9964042c3baf99702999eb7d7 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eefdd680a23c40e1bdefef903af478b2 : SUCCESS in 3m 25s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/4e190c6ba81746b7b35e1c72fd31b7c7 : FAILURE in 2m 46s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/d18aefe6a961475e98cbce779afaec10 : FAILURE in 3m 37s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/e0e3e27e10ff46bfb83dd33061b9dc3b : FAILURE in 3m 28s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/076efc1306b341e48d0cdaa04b7a6676 : FAILURE in 3m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/95535fdc9ca041be9d0ac44bed1face3 : SUCCESS in 1h 40m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/15631e567cf1472c9ba3c24369dbb7d9 : FAILURE in 32m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/33ca7f051dad42dfa8aff13342b6729f : SUCCESS in 1h 37m 04s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/f189e5a5a5324664850016d57aad970e : SUCCESS in 1h 01m 49s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a51398ba82146edb38d31febe061fdb : SUCCESS in 1h 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01e8b7cc71d7e6b945e223b886bd2f036c205d55","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-13 11:31:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dbc3fbf1f24f07cc44ec0e07c238bf5f4bcdc492","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-13 11:34:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b71f0772863c5c5543871a5735eee5f4d663be6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"date":"2023-05-22 13:00:15.000000000","message":"Abandoned\n\nfixed in https://review.opendev.org/c/openstack/oslo.messaging/+/866617","accounts_in_message":[],"_revision_number":2},{"id":"36d81f26085aea08064630b28573a8e8bca8dcf3","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2023-05-22 15:00:55.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_32553\u003e.","accounts_in_message":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"_revision_number":2}],"current_revision_number":2,"current_revision":"f6cb313bbe70f7045177c9afcdde22bd0ebd8640","revisions":{"09e76e1e8206a8b5247ca3b2934a196d2bb49901":{"kind":"REWORK","_number":1,"created":"2022-12-05 18:10:15.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/16/866616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/866616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/866616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/866616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/866616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/866616/1"}}},"commit":{"parents":[{"commit":"2e81fac97344d1eacf4c56a1022f25256d52e7ab","subject":"Merge \"Implement get_rpc_client function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2e81fac97344d1eacf4c56a1022f25256d52e7ab"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2022-12-05 15:25:00.000000000","tz":-360},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2022-12-05 18:10:14.000000000","tz":-360},"subject":"Deprecate kombu_reconnect_delay setting and remove associated code","message":"Deprecate kombu_reconnect_delay setting and remove associated code\n\nIt\u0027s unclear that this delay provides any benefit, but it\nIS clear that it does harm -- values of more than\n1.0 cause us to always exceed the 1.0 second timeout limit\nACK_REQUEUE_EVERY_SECONDS_MAX which results in various code\npaths never being traversed due to premature timeout exceptions.\n\nBug: 1993149\nChange-Id: I2aea9cbbf5136ebe1c6ef10ce94faf16e5d5c47c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09e76e1e8206a8b5247ca3b2934a196d2bb49901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/09e76e1e8206a8b5247ca3b2934a196d2bb49901"}]},"branch":"refs/heads/master"},"f6cb313bbe70f7045177c9afcdde22bd0ebd8640":{"kind":"REWORK","_number":2,"created":"2023-01-10 20:18:01.000000000","uploader":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"ref":"refs/changes/16/866616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/16/866616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/866616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/866616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/16/866616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/16/866616/2"}}},"commit":{"parents":[{"commit":"9f710ce6cd955eaf6d7f9b474efa21df3d531be0","subject":"Merge \"Remove logging from ProducerConnection._produce_message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9f710ce6cd955eaf6d7f9b474efa21df3d531be0"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2022-12-05 15:25:00.000000000","tz":-360},"committer":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2023-01-10 20:16:59.000000000","tz":-360},"subject":"Deprecate kombu_reconnect_delay setting and remove associated code","message":"Deprecate kombu_reconnect_delay setting and remove associated code\n\nIt\u0027s unclear that this delay provides any benefit, but it\nIS clear that it does harm -- values of more than\n1.0 cause us to always exceed the 1.0 second timeout limit\nACK_REQUEUE_EVERY_SECONDS_MAX which results in various code\npaths never being traversed due to premature timeout exceptions.\n\nBug: 1993149\nChange-Id: I2aea9cbbf5136ebe1c6ef10ce94faf16e5d5c47c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f6cb313bbe70f7045177c9afcdde22bd0ebd8640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f6cb313bbe70f7045177c9afcdde22bd0ebd8640"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
