)]}'
{"id":"openstack%2Foslo.messaging~870084","triplet_id":"openstack%2Foslo.messaging~master~Ic19c896af2ab0cf3570c43e8ceb8cba64fb45cdd","project":"openstack/oslo.messaging","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-01-31 15:43:20.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-01-31 17:04:16.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-01-31 17:04:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic19c896af2ab0cf3570c43e8ceb8cba64fb45cdd","subject":"tox cleanups","status":"MERGED","created":"2023-01-13 12:00:59.000000000","updated":"2023-01-31 17:05:36.000000000","submitted":"2023-01-31 17:04:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870084-tox4","meta_rev_id":"3aa49249a853f7cb3fae7d8d8f299780eace5081","_number":870084,"virtual_id_number":870084,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-31 17:04:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-01-31 15:43:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-01-31 15:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"CC":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-13 13:54:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-19 08:22:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2023-01-27 07:23:42.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"097d5ff29343ed039773d250ee781c39c930e94a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-13 12:00:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7efb162b901b5191976da977cf4aef88cff34b56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 13:54:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29947e74e25e4abf9958f5bf1b0933cf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24cdbc7fc6274e10ade6c0199b6aa435 : SUCCESS in 1h 25m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c12f1fadd07b4f179e619c477b87d709 : RETRY_LIMIT in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ce0c8ec5e247378257289f8ce93f4b : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be07818506564b1bac0a489a749fa60f : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6321bd35dde434292a26f8483556f7c : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6821dca4c4e84542b85ef3104a7fafc5 : RETRY_LIMIT in 2m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c656007bfc4f929d3238afbb9e682a : SUCCESS in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df8fb3ef966a421c9c87922ab366d0ec : SUCCESS in 3m 08s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/d3e5efeff584459ba7ce43679ba06b9b : FAILURE in 2m 27s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/07d1b4d2e5f542109b5e775067403441 : FAILURE in 2m 58s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/e703b9e357ed4bbebab3625395606741 : FAILURE in 2m 59s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/f29a250d87284139aaa1c1af8d65f151 : FAILURE in 2m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ddef54980d5942358d97933cd453b280 : SUCCESS in 1h 33m 52s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/006439c7d7c144af93739fa756eef052 : FAILURE in 25m 26s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9afc2acc14a948ed9e33216671ee8518 : SUCCESS in 1h 48m 03s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ea7173e15c4a409daf0be4f70fafc83a : SUCCESS in 54m 39s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/139d0032c82749d987ad83cd7f6e8e81 : FAILURE in 39m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"121658f539aeee8dbb4432b6ac47ad64e6ef37d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-18 10:40:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"206420fa46ead3416bb6aecc800e9a4145520551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 12:57:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ab16f9e2c454ea294aca9dbeebcf4d3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f60444131314d1eafebf760c02b348d : SUCCESS in 1h 43m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8caa94fefe6346688b32a2c3dbed5fc6 : RETRY_LIMIT in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3706125c86c44567a1d5fd5fe48f5b0a : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46f80124ae61435187b28d74b734ddc9 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/212a19be69104896aeddd42ff44f4bf7 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/774c41ae07bd49e9ad7a000398d47732 : RETRY_LIMIT in 3m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6fb2d876ec4bf4b60c012a3c4b4bcb : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a8362ac69d149e9af8c5acfd1e3b1ce : SUCCESS in 3m 32s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/d70b7be948fe4d29b65fd635be7ff4ab : FAILURE in 3m 36s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/91df8e5aad174428854827c2d6b835bf : FAILURE in 7m 40s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/b958d1dbbc1c4f2384ff4b22640f8f34 : FAILURE in 4m 11s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/546b4045100345ebb8eff87119df1859 : FAILURE in 3m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ba949b875baa4763b9e725306aae67e8 : SUCCESS in 1h 31m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e1b1c35eb48d47cd9d3828deb9f026a6 : FAILURE in 27m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/f15a48951db44664a5c060055cb13584 : SUCCESS in 1h 53m 22s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/acef49ce23034cdeb060886880ee655f : SUCCESS in 53m 34s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/cca553cfb4a84cea88f7c743cb2e27e7 : SUCCESS in 1h 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b21649537329ed3c7b216ed2ed7e26a93a48560a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-18 14:46:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a514c0768a660ffce46f11d081592fce4368eb24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 16:44:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6068714c04a41d4af72f0906a74e162\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0d3a61bab124f97beaf2d4c5b60d6c9 : SUCCESS in 1h 33m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70b5dc5565ec4ef493dddd5f2fb64ddc : RETRY_LIMIT in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9c154be44c44048ae301dbf28d4eef0 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a67c82e97ea74bc7a8fe024cdb9dca1c : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/535a6c80ca3f4c2f8a6ce8c102ff4249 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03374579d67446a1a9bee9a7f02250e9 : RETRY_LIMIT in 2m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9f002f5cff946668e4f5d7dd6adf6e4 : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e226b0ef8378453a9b72c027ad7b7bd3 : SUCCESS in 2m 52s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/cbfebc03e9b149bbbd8bfd62af8075e8 : SUCCESS in 4m 50s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/cbec8d2af34541d3bda93c6f3a5eed63 : FAILURE in 6m 17s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/8d78e5e139f746c6aeecb41ad8b74602 : FAILURE in 4m 05s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/ff3466ce6603407685270563f47c5e7b : FAILURE in 5m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b161f07166b14044848ec53023c982c1 : SUCCESS in 1h 52m 18s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a98aacab39284e8e9e5b0fc8a5fd366b : FAILURE in 19m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/a768f41b86a342c0854d162250772acd : FAILURE in 1h 48m 46s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3ddb524f39f9474eaa1e0337460f472d : SUCCESS in 56m 41s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/0456400f714e43289205498d61ac5bc6 : SUCCESS in 1h 07m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ffd7357c2ab347e1ebeb27c3a6101ffb4a9bbe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-18 17:00:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1e56ce6959121ce060ba10c36f91e853e484523f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:35:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a67ca73212ed40328ec7123717a0a9d7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d48109f08df24f99b5a8fa0b6de12719 : TIMED_OUT in 2h 06m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecdd5c9bcc10474090d113fc836c9880 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f09d6fd7a3ef4b29bd43c7d08fcc9498 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1fe8dc2f3234a4299fc9f338e6aa2b4 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cee69565903140b289a8119e76960ecf : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d27cb55c78b042ac908a44e3e9e92d5f : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dfc37b1a29744fda9c3c432f710eaa2 : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0cd04f6d40649ad9a513f6d3d4c3a63 : SUCCESS in 2m 53s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a1503f7bcc134173943dbcef9396ff7d : SUCCESS in 4m 21s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/ecfc419f85d0411a844312b49105c841 : FAILURE in 6m 43s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/8a3109b497e54a79ba3ab3ad23a68180 : FAILURE in 3m 44s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/6e908d4b6e28439da0fa74da49ae8771 : FAILURE in 4m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9d95f19c5eb9468c9bac548e308e268f : SUCCESS in 1h 45m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/23f57fb83e1f45b1a588f146e2aa8e25 : FAILURE in 30m 24s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d6e6711486424a8b8e48763af9dca27a : SUCCESS in 1h 31m 05s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/eccd251894ce460c99134ea7fc368343 : SUCCESS in 57m 14s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e4d3bb7bcc84a7eaa3a70ec37d6930e : SUCCESS in 1h 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"10895b7cd33e57f4571d914370e21615c5c46ab9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-19 08:22:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"96d7c2e8ce02d90170548050d08ad650a2179693","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-19 10:22:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1766de7b1254fd6adc99243beac314f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bef4b12ce7b34af0b0d01704e667fc10 : SUCCESS in 1h 30m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c21719e8b8549e48995ab88a2c9560a : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fdd80c88e847d68955023c1851ed8b : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c85c473f79046969b5dd7b7d27cb389 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a93391950b84e3ca7994764b7cff333 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/899916ad3f5a4c148ac11249f4b51586 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2815f0d846fe404cabc4fa7d3a2082fc : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1905a7945c2346a690516bf4ffe23ea5 : SUCCESS in 3m 44s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/ff911152c3df4bd2a61ab71cad40eb7c : SUCCESS in 4m 27s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/3108d8900a1d482fa004b586157be62f : FAILURE in 5m 22s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/d35f924049f3451d9ef017e172734b48 : FAILURE in 4m 03s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/575312d9673d49cf8591d8527ef63216 : FAILURE in 3m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/178c49e17f964d1d961934cbc61725df : SUCCESS in 1h 16m 58s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/62e79a04998a426aa5e3fce967d03b4a : FAILURE in 27m 01s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/9732ad1d8fef45d584ce5c3611c4122c : SUCCESS in 1h 48m 44s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a75b938107fa420a8c66defe6a0e6a44 : SUCCESS in 55m 55s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7cda9611ee04e0ea57aea121545d1c7 : SUCCESS in 1h 07m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0659b4aa08eb7ed78749b47d3a49744e78e8f5b1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-01-31 15:43:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"051524bf28ea1cbc3efd913ee7ba02bb2c963994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 15:43:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"46ac878ace7215982bac6dae094071adac646684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 17:04:15.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94ea209e80524c5c98bb7759a7140040\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d0ebf18678f4f34b73e01c62fe3d557 : SUCCESS in 1h 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da2665c36ae74c93bd719ac03bf879e2 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39dd00e58f464478ba947c8e430890c8 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/899f365809334948a34dc19c659a8c81 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60bb6f1e974f41cd8c39a8795cb85be6 : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77b080ce4f134716bdd0c68824d5218a : SUCCESS in 3m 30s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/2c4e550e574a41ae9ebbe7249d13d926 : SUCCESS in 6m 37s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7a1d25e8aab84d9b997103dba6483a4e : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":4},{"id":"4ad7f244acf0e078f74566b1ee8eeb404cefc361","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 17:04:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3aa49249a853f7cb3fae7d8d8f299780eace5081","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-31 17:05:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a03e426349b4c158c689a02d41434e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3425d21d386b4aaea0f45ab0bf7dcbba : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff35427a338849c19e95da7eccea44e7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7505316902d6469842b8c6f84d520ebd2314b7c7","revisions":{"bbd39ed733bb4d5aaa37fd4be968a09d06c5175e":{"kind":"REWORK","_number":1,"created":"2023-01-13 12:00:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/870084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/84/870084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/1"}}},"commit":{"parents":[{"commit":"9f710ce6cd955eaf6d7f9b474efa21df3d531be0","subject":"Merge \"Remove logging from ProducerConnection._produce_message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9f710ce6cd955eaf6d7f9b474efa21df3d531be0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-01-13 12:00:40.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-01-13 12:00:40.000000000","tz":0},"subject":"tox cleanups","message":"tox cleanups\n\n\u0027skip_basepython_conflicts\u0027 has been the cause of a couple of bugs in\ntox 4 and there is talk of it going away. Remove it and fix up a few\nother issues in the tox.ini file.\n\nChange-Id: Ic19c896af2ab0cf3570c43e8ceb8cba64fb45cdd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bbd39ed733bb4d5aaa37fd4be968a09d06c5175e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bbd39ed733bb4d5aaa37fd4be968a09d06c5175e"}]},"branch":"refs/heads/master"},"90effe3fc77d1cb4ceb49811d645a00cd07289e4":{"kind":"REWORK","_number":2,"created":"2023-01-18 10:40:07.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/84/870084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/84/870084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/2"}}},"commit":{"parents":[{"commit":"9f710ce6cd955eaf6d7f9b474efa21df3d531be0","subject":"Merge \"Remove logging from ProducerConnection._produce_message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9f710ce6cd955eaf6d7f9b474efa21df3d531be0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-01-13 12:00:40.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-01-18 10:40:07.000000000","tz":0},"subject":"tox cleanups","message":"tox cleanups\n\n\u0027skip_basepython_conflicts\u0027 has been the cause of a couple of bugs in\ntox 4 and there is talk of it going away. Remove it and fix up a few\nother issues in the tox.ini file.\n\nChange-Id: Ic19c896af2ab0cf3570c43e8ceb8cba64fb45cdd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/90effe3fc77d1cb4ceb49811d645a00cd07289e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/90effe3fc77d1cb4ceb49811d645a00cd07289e4"}]},"branch":"refs/heads/master"},"5b75ed5094a778d7b4b792fd9dbd5fb243767563":{"kind":"REWORK","_number":3,"created":"2023-01-18 14:46:50.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/84/870084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/84/870084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/3"}}},"commit":{"parents":[{"commit":"9f710ce6cd955eaf6d7f9b474efa21df3d531be0","subject":"Merge \"Remove logging from ProducerConnection._produce_message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9f710ce6cd955eaf6d7f9b474efa21df3d531be0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-01-13 12:00:40.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-18 14:46:45.000000000","tz":0},"subject":"tox cleanups","message":"tox cleanups\n\n\u0027skip_basepython_conflicts\u0027 has been the cause of a couple of bugs in\ntox 4 and there is talk of it going away. Remove it and fix up a few\nother issues in the tox.ini file.\n\nChange-Id: Ic19c896af2ab0cf3570c43e8ceb8cba64fb45cdd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5b75ed5094a778d7b4b792fd9dbd5fb243767563"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/5b75ed5094a778d7b4b792fd9dbd5fb243767563"}]},"branch":"refs/heads/master"},"7505316902d6469842b8c6f84d520ebd2314b7c7":{"kind":"REWORK","_number":4,"created":"2023-01-18 17:00:40.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/84/870084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/84/870084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/84/870084/4"}}},"commit":{"parents":[{"commit":"9f710ce6cd955eaf6d7f9b474efa21df3d531be0","subject":"Merge \"Remove logging from ProducerConnection._produce_message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9f710ce6cd955eaf6d7f9b474efa21df3d531be0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-01-13 12:00:40.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-18 17:00:10.000000000","tz":0},"subject":"tox cleanups","message":"tox cleanups\n\n\u0027skip_basepython_conflicts\u0027 has been the cause of a couple of bugs in\ntox 4 and there is talk of it going away. Remove it and fix up a few\nother issues in the tox.ini file.\n\nChange-Id: Ic19c896af2ab0cf3570c43e8ceb8cba64fb45cdd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7505316902d6469842b8c6f84d520ebd2314b7c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/7505316902d6469842b8c6f84d520ebd2314b7c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
