)]}'
{"id":"openstack%2Foslo.messaging~873630","triplet_id":"openstack%2Foslo.messaging~master~Icc438397c11521f3e5e9721f85aba9095e0831c2","project":"openstack/oslo.messaging","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-02-14 16:48:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-02-22 15:44:24.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Icc438397c11521f3e5e9721f85aba9095e0831c2","subject":"Fix typo in quorum-related variables for RabbitMQ","status":"MERGED","created":"2023-02-13 17:24:36.000000000","updated":"2023-02-22 19:18:43.000000000","submitted":"2023-02-22 19:17:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873630","meta_rev_id":"db4feeca090707c9bf137f8972a69ca1aa8235d7","_number":873630,"virtual_id_number":873630,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-22 19:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-02-14 16:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-02-14 15:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-02-14 16:48:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-13 19:23:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-02-13 19:28:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-14 15:21:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"53919cf16f4d3469cda9107237ed5f98c68dcc9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-13 17:24:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dff494c4f3307c780a99a52d00470c5b17b854bd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-13 17:32:04.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee0e7a2d6c9889146f631f2982822770da215c61","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-13 19:23:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b1cef871bad682e48bc7a0c73ea1349c3cbbef89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 19:28:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99d5da63b5f94a28aad41b9c9db0a075\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/999f78b6f5c34dc39c829c1339baea85 : SUCCESS in 1h 42m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/617491000cd842ab926a7d5d1bf1520f : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc6f50d59744abda94343d629cc0b77 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cadf68011c514589b8226610fcff702b : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0b4663a054643bf9f217b7e2ec81b7b : SUCCESS in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/861e00f126dc42cc98ee8287bbed6163 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12fb5189855245938cfd32413c8031eb : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f684c4c979e34eb898d29211a6a72923 : SUCCESS in 3m 26s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/9f31a1b292fa4a688f5c0adae40ec4dc : SUCCESS in 4m 35s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/dfb191ffb524496996eb9d4cfdf07b68 : FAILURE in 5m 53s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/0a8584e854a643758c3f0cbcea874f7b : FAILURE in 3m 49s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/bccdd9862e254d2f94390c3b8dc7a707 : FAILURE in 3m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/827b42a919374a3ba4e56e90812558ac : SUCCESS in 1h 28m 00s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/9cec5b8c7b074ae99aacba09a0adfd9e : FAILURE in 19m 19s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/7b2691e52d4448be98a86bb6c950bdc7 : SUCCESS in 1h 23m 42s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c8909d525d814ecf893df7c2a0073f6e : SUCCESS in 50m 19s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/7a1f5ca1500244249aa945ca8b362286 : SUCCESS in 1h 08m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da7dad5f3dd4491727291ba457e367ffd5dcaa49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-14 15:20:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e584c8413d2dac07c5083f467835b685b5ec7d3a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-14 15:20:20.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8ac9b5f2c1a9d143959c8d2d62b4e3345f9b601","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-14 15:21:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4331f65db12eda042ef2c4dca87f000ab9bdace3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-02-14 16:48:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa1df574f3f065e1ebd3f01b678255acb167ddc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 16:52:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc9e9dd017094fa1b488faea02e66ae4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68c54c3a85b644ce96246116a238d975 : SUCCESS in 1h 27m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97dcd073b6e84189971e078290b1c94a : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/390c04f1aa7d43cabc9397c3669e4f8a : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6af2b7126ad44c38a5f88ccd2df0d42a : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d307cb6fca6d4e3098b9b80d794c14fa : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f65ffd4e1e84d60bd0987c4fd7f9514 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e236f427cdfa4c31ab67baa4339b3e49 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a26561cded914a79aa6bdcf323616a39 : SUCCESS in 3m 21s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/d158ae61de284065b1ffdd89b681a9dc : SUCCESS in 4m 29s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/d42a033a6e34404a88c52934064e35d7 : FAILURE in 5m 59s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/6d5772bd30cc4775b1a4c5bcca9f32e3 : FAILURE in 4m 49s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/43c08e7c6cc34600804ee501c6df1591 : FAILURE in 5m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a7fa96d550a341708e7f38dfa07b9377 : FAILURE in 1h 29m 15s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/76de0c4d2e7d4f1eb13ff4c7a2e01d01 : FAILURE in 19m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/179f55573933458bae6f4e24682a6714 : SUCCESS in 1h 25m 38s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/0a783712e0284db39639db864ffff296 : SUCCESS in 56m 48s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/76b38da68045424fad12635e4e43b301 : SUCCESS in 59m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"26f2d40c49aa4851f73d1f3971f568ee2118a145","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-15 16:15:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd92624703c4ee7f3861e0152942b46e28921881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:38:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4896f0736cfb4bb890a5d323c3a88966\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/183bd62b8ac3413e9299273370bfd810 : SUCCESS in 1h 24m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45e4c12e3b234aceaffb9662a557f5c5 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d50a5639e44125a176dbb1180b5ea1 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8f9e76a3b4f42949ea6333cdbd769ac : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/173a3411e4b2491997a942dc7ce016f9 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83841c233fee4315b333f50276053109 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bfdad70718c474ea5ba14c1903b9e57 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55a519b92068462fa841b9a79db84ca6 : SUCCESS in 3m 06s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/1a00755182b340f28068a8821fe2da16 : SUCCESS in 5m 39s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/d230efe050844f8daad7759702387331 : FAILURE in 7m 06s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/9f15b5bbf5a64e9e854c36dace6fdd3d : FAILURE in 5m 35s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/f79b2985c21c4567a828bf77b2c4a4c0 : FAILURE in 3m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/57fbba7c6994404ead6afec363efe3d1 : FAILURE in 1h 46m 22s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/706cf82e44734a29b59cc6194439d1b0 : FAILURE in 27m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/28be3e3b3acc4ff194427910b8592de1 : SUCCESS in 2h 10m 44s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/ca4f0f9aa4c34c088eaf626ecfbcf45a : SUCCESS in 1h 11m 07s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/30e7ffa04c654fe1ae21f940041a7b03 : SUCCESS in 1h 13m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f61933881f9bf5391d03b8d34e9baf59b8876f75","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-02-22 15:44:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f850e73c3ab9dc8e0da4cb1800d955e7f2d7ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 17:45:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3073fd10c074d7890765b7cc4c9d76e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09d8c563c03e47e58a6c7c43b2e44a62 : SUCCESS in 1h 13m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b38121d06d34291a392202443a5bb7c : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e72b183d57473bbb0e70e254dca001 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea9958653dac42e99c9ef7de6a9a4a3c : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fc41ef16f3f44dab966194bff5e2aec : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/492d865b76694f2291c3b70959406354 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e1efcd233c04404bc22d8328a9c26c7 : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d039663dd4cd442eb741a037284388d1 : SUCCESS in 3m 39s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/d3c0ee9dee0f424da49bb42b6f5f6eda : SUCCESS in 5m 03s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/6723a6cf8cfc4589a77342293d4a84c6 : FAILURE in 5m 35s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/66984f10525f47d392ca83dbb315218c : FAILURE in 5m 27s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/a6a87edffbbd4da1a0e40a2f17811a55 : FAILURE in 3m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/1dd753796c1d475ba28721fb7b4dfd0e : SUCCESS in 1h 21m 19s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f1e5c2d3cd4640f6b47d3fd2f856fb4b : FAILURE in 26m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ba0e3a2fb2924143b116c9698e7d4c8d : SUCCESS in 1h 56m 09s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/eff739363ca3491382bb2bf466cfbad2 : SUCCESS in 55m 26s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c10de089bc8e4e5d87b0ec09b2aa6a4b : SUCCESS in 1h 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50777977347fb28f77254a82ab2a79c1ed7b41bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 17:45:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2de30b09940804b87da81437fe911b00b4612b14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 19:17:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c35ac51d7b504c5bb75e565e043e5d08\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87942ba0a3644d15bb0587ae25136406 : SUCCESS in 1h 23m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b2b7dfc3384b70b07b4ebff28be4b4 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8c192ea6a2b469b9a85bb2eec4fdd14 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46e08bb88816464aa698fe903856fdf1 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d3bfd4550a435389e3ff50c0365f27 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a0b3ea3cf54415ab547d031f23c6ccc : SUCCESS in 3m 22s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/458bfad0e879453fb93941325283be8b : SUCCESS in 4m 30s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d627f03a1e40481f9543e28b4ddc33e9 : SUCCESS in 1h 24m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3887e509e0dba0e5b09bc02ac2b43e086c70c9f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 19:17:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"db4feeca090707c9bf137f8972a69ca1aa8235d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 19:18:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3da2c8d21e4840219f896162a3fa344e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45eaa2a6e13c474d8f29938746cbe08e : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b895c14c5be448a593169251f244931b : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"115cfb5b7cff2a9803d64d02b98250b7b0fc09f0","revisions":{"9e4dfe95874f520e6c1da76fa3ef4e4a9e4b8b49":{"kind":"REWORK","_number":1,"created":"2023-02-13 17:24:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/873630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/30/873630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/30/873630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/30/873630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/30/873630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/30/873630/1"}}},"commit":{"parents":[{"commit":"687dea2e65814606bece8041fe94fb2674d2590b","subject":"Support overriding class for get_rpc_* helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/687dea2e65814606bece8041fe94fb2674d2590b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-13 17:24:20.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-13 17:24:34.000000000","tz":60},"subject":"Fix typo in quorum-related variables for RabbitMQ","message":"Fix typo in quorum-related variables for RabbitMQ\n\nIn [1] there was a typo made in variable names. To prevent even futher\nawkwardness regarding variable naming, we fix typo and publish a\nrelease note for ones that already using variables in their deployments.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/831058\n\nChange-Id: Icc438397c11521f3e5e9721f85aba9095e0831c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9e4dfe95874f520e6c1da76fa3ef4e4a9e4b8b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9e4dfe95874f520e6c1da76fa3ef4e4a9e4b8b49"}]},"branch":"refs/heads/master"},"115cfb5b7cff2a9803d64d02b98250b7b0fc09f0":{"kind":"REWORK","_number":2,"created":"2023-02-14 15:20:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/873630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/30/873630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/30/873630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/30/873630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/30/873630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/30/873630/2"}}},"commit":{"parents":[{"commit":"687dea2e65814606bece8041fe94fb2674d2590b","subject":"Support overriding class for get_rpc_* helper functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/687dea2e65814606bece8041fe94fb2674d2590b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-02-13 17:24:20.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-02-14 15:20:00.000000000","tz":0},"subject":"Fix typo in quorum-related variables for RabbitMQ","message":"Fix typo in quorum-related variables for RabbitMQ\n\nIn [1] there was a typo made in variable names. To prevent even futher\nawkwardness regarding variable naming, we fix typo and publish a\nrelease note for ones that already using variables in their deployments.\n\n[1] https://review.opendev.org/c/openstack/oslo.messaging/+/831058\n\nChange-Id: Icc438397c11521f3e5e9721f85aba9095e0831c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/115cfb5b7cff2a9803d64d02b98250b7b0fc09f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/115cfb5b7cff2a9803d64d02b98250b7b0fc09f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
