)]}'
{"id":"openstack%2Foslo.messaging~875614","triplet_id":"openstack%2Foslo.messaging~master~Icae57315a93a172754bd5fbf3b89f4dec894a7b2","project":"openstack/oslo.messaging","branch":"master","attention_set":{},"removed_from_attention_set":{"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-02-28 10:09:28.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"Icae57315a93a172754bd5fbf3b89f4dec894a7b2","subject":"Send heartbeat_threshold/rate frames before timeout","status":"ABANDONED","created":"2023-02-28 06:21:08.000000000","updated":"2023-02-28 10:55:14.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5c0663d4ec3a50b1f536d63ae97ba6cc2a5cf1d4","_number":875614,"virtual_id_number":875614,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-02-28 10:09:28.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 08:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"465a6bb03dc8890f13f29bd195b3e8f1357a6134","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-02-28 06:21:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccb85e91f2aebf6ff3c9051cfafcea0e4bf763c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 08:01:32.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ddfbe1067a740ed90770438d8576e4e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e8942648eec4c98ac9bb0cfcc8012e8 : SUCCESS in 1h 08m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/711e98f275774deeb9787a71dd77c17b : FAILURE in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a78960a7a64a578c977380858fd9fd : FAILURE in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b3c4ecd9f864eb19b6dd48732b00a64 : FAILURE in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/989e460ae3df4463b803eb278d686765 : FAILURE in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7802e14810eb4f6c8febcf78399ee1bf : FAILURE in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6c7c2b519e4420a16ab2ebdac36cf9 : SUCCESS in 5m 55s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/7523bc04b8574bba8f434a41c0490abe : SUCCESS in 5m 25s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/6cbbba438c734f9b8b2574e3f6d55397 : FAILURE in 8m 31s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/7ea298d15aff422a8ed4ab7f04a68ab4 : FAILURE in 5m 28s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/692369f3ae9f43ef8ada378c17398da9 : FAILURE in 3m 25s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9deae0dfea1a46878b49af4530b07205 : SUCCESS in 1h 15m 52s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7dc4f2ac90ed49d0828204cf5453e418 : FAILURE in 29m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2ca93d8a361b4b9999051df3255d05fc : SUCCESS in 1h 33m 37s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b671f0ee7c7d423aae5a26ed0198125c : SUCCESS in 1h 08m 36s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e2881dfc5b242eea4470afbe3edaf77 : FAILURE in 1h 21m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d855331f9707f677e22cb46465d2a5f2437fc372","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-02-28 09:46:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3f2007b9cf06b40048b2555b89a621b0256d49f3","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-02-28 10:09:28.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c0663d4ec3a50b1f536d63ae97ba6cc2a5cf1d4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-02-28 10:55:14.000000000","message":"Abandoned\n\nfix is done is amqp https://github.com/celery/py-amqp/pull/405","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2f363ec84b1eada8e961d90d5951a5306ec7433a","revisions":{"b58643f3d6242526af5af04f0a8578eacd76d306":{"kind":"REWORK","_number":1,"created":"2023-02-28 06:21:08.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/14/875614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/14/875614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/875614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/875614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/875614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/14/875614/1"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-02-27 20:17:08.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-02-27 21:26:05.000000000","tz":60},"subject":"Send heartbeat_threshold/rate frames before timeout","message":"Send heartbeat_threshold/rate frames before timeout\n\nThe heartbeat specs in AMQP is saying that we should send at least 2\nframes before closing the connection [1].\n\nThe current implementation was sending only one frame in the interval.\nIf for some reason that frame is lost, we have no other chance to send\nanother frame in the interval and the rabbit broker can decide to close\nthe connection, ending with \"Server unexpectedly closed connection\"\nerror in logs.\n\nThis is fixing the implementation by relying on the heartbeat_rate\nparameter. With a default rate of 2, it will send at least 2 frames\nbefore the heartbeat_threshold_timeout is reached.\n\nNote also that the current implementation is not fully correct as it\nwill wait an extra heartbeat_threshold_timeout/heartbeat_rate/2 seconds\nbefore sending the frame.\nAn easy workaround for this is to set the heartbeat_rate to 3 in the\nconfig.\n\n[1] https://www.amqp.org/specification/0-9-1/amqp-org-download\n\nCloses-bug: #2008734\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Icae57315a93a172754bd5fbf3b89f4dec894a7b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b58643f3d6242526af5af04f0a8578eacd76d306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b58643f3d6242526af5af04f0a8578eacd76d306"}]},"branch":"refs/heads/master"},"2f363ec84b1eada8e961d90d5951a5306ec7433a":{"kind":"REWORK","_number":2,"created":"2023-02-28 09:46:04.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/14/875614/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/14/875614/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/875614/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/875614/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/14/875614/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/14/875614/2"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-02-27 20:17:08.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-02-28 09:46:00.000000000","tz":60},"subject":"Send heartbeat_threshold/rate frames before timeout","message":"Send heartbeat_threshold/rate frames before timeout\n\nThe heartbeat specs in AMQP is saying that we should send at least 2\nframes before closing the connection [1].\n\nThe current implementation was sending only one frame in the interval.\nIf for some reason that frame is lost, we have no other chance to send\nanother frame in the interval and the rabbit broker can decide to close\nthe connection, ending with \"Server unexpectedly closed connection\"\nerror in logs.\n\nThis is fixing the implementation by relying on the heartbeat_rate\nparameter. With a default rate of 2, it will send at least 2 frames\nbefore the heartbeat_threshold_timeout is reached.\n\nNote also that the current implementation is not fully correct as it\nwill wait an extra heartbeat_threshold_timeout/heartbeat_rate/2 seconds\nbefore sending the frame.\nAn easy workaround for this is to set the heartbeat_rate to 3 in the\nconfig.\n\n[1] https://www.amqp.org/specification/0-9-1/amqp-org-download\n\nCloses-bug: #2008734\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: Icae57315a93a172754bd5fbf3b89f4dec894a7b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2f363ec84b1eada8e961d90d5951a5306ec7433a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/2f363ec84b1eada8e961d90d5951a5306ec7433a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
