)]}'
{"id":"openstack%2Foslo.messaging~877087","triplet_id":"openstack%2Foslo.messaging~master~Ic558ad392424a25b5fd9a10749163d8427159eda","project":"openstack/oslo.messaging","branch":"master","topic":"oslo-bump-bandit","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-07-12 10:30:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-10-20 13:47:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic558ad392424a25b5fd9a10749163d8427159eda","subject":"Bump bandit and make oslo.messaging compatible with latest rules","status":"MERGED","created":"2023-03-10 09:55:19.000000000","updated":"2023-10-20 13:48:35.000000000","submitted":"2023-10-20 13:47:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877087-oslo-bump-bandit","meta_rev_id":"7a58e780a2629257b3c55945b2cf67a4a6c6f415","_number":877087,"virtual_id_number":877087,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-20 13:47:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-06-27 12:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-07-12 10:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-07-12 10:30:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-10 11:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-10 02:06:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-06-27 12:05:14.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-10-11 10:43:54.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"}],"messages":[{"id":"d532df51aa6fff8cd3a6a2d972daec09573d646b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-03-10 09:55:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a2eea6600e5436b4c9fc19ab802b0e821785777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 11:52:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe369e00ac0e49c0941f423992b99b90\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9a8b667f18994b6ab1fef50ac28ebe65 : SUCCESS in 2m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b41e99ed9f734b95bcf6943a4e20dacb : SUCCESS in 1h 15m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1586c5db25704beab8ed8b9364fce499 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/629c24cf011d42e28c581d78aafd79f0 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2ebff2d11c24247bc533f18e2ba38e3 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/838021fa381d4799be7ef8edb84d680f : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f581f99ab1b74976b67f1765d7b6943a : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217c493bad494f35bb3649014e4523f8 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d01916bb556f445e86cf355d42094875 : SUCCESS in 3m 11s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/081ff350f2f64ec4beeec34a1435fee7 : SUCCESS in 4m 37s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/623fd6f1e8184226b903a4ea71bf27f8 : FAILURE in 5m 08s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/29ab643a73a84b518e5aa875cf97cdab : FAILURE in 4m 00s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/69f9861b2ef046029f26211a7a152ab6 : FAILURE in 3m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4fe9d222e0804556af81513ec49c6772 : SUCCESS in 1h 47m 50s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/5995126763c04f8f994af7607ab10b55 : FAILURE in 18m 05s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/3a5cedcd60f24077b41fa4046f13006f : SUCCESS in 1h 21m 35s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/89c96cea08ff4736965d384e93383566 : SUCCESS in 49m 11s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca679b6a8a9b458eafa5ea2c2d6c9638 : SUCCESS in 1h 02m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"395fbf37028c0946a5f80bc09e004b07c4b78d87","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 02:06:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d38924d8e747492e313143a1913f82ab32dbd2f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-17 09:05:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0f5c9d3f215a9abaa0dfc209fa335f7c1dfc2f06","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-17 09:06:08.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"763f47922fc3f61639f3a9057df215dab42f99d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-17 09:07:02.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"142af129371b29d9464afb4a1afd1f03417a0c2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 11:05:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/319609317e0942d88f39f55aab6eccc6\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/33fe66c149c545fab18239d4c541b181 : SUCCESS in 2m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39f2e90c16fd4d22994ae3d4777d88c0 : SUCCESS in 1h 51m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a466bfa3eb494f8098ed5a9b6f3cd80d : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dfd8db754144b7cbef507c6e821a47c : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a73267ea52934465af86fb9a1b052f00 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d3d2c765f20459abd75977f74cdc492 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a48a8ef61e4d450a8afc8210a90b5d15 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee197759ed64836a907c1f57a6a3f6d : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24e0028d5a324d6d998a71d498eef36b : SUCCESS in 3m 58s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/edef028e34954078ae1b55423bb0d87b : SUCCESS in 4m 46s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/e271e9c12a3b45b38bfc44958dd6194e : FAILURE in 13m 43s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/e7d1c6034e734277b3e1f97758d36384 : FAILURE in 4m 30s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/161932c52d454d72951b91fd82a2af5d : FAILURE in 12m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/be3865bbc4d5467d818610d86e15a581 : SUCCESS in 1h 41m 43s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/e4de86e52b2b43f89163292e122f6d35 : FAILURE in 29m 45s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2af71be1ec2b4d53bb3132b9ad6deaed : SUCCESS in 1h 48m 02s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/148976f8a2ed4697ba1decd63177f7da : SUCCESS in 1h 01m 33s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/784fc8aecf0e49f285e94d9fb36eef8c : SUCCESS in 53m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"94c74b7ef3b7c25d1a8b2efef8fbef84a0428283","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-27 12:05:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e5a71cdf787fe40dae063512daf5baf66d994723","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-27 12:05:21.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9bc64b84d75699de8855a8abde4a5c760b03cbd8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-12 10:30:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ab591e089cddda11555871dd25a0ff51ba4ab70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 10:30:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3def5cca07cb2df00e13fd45d69777c76a18d3bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-12 12:42:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c1bdbf78ccf4efa9da731a756cd9b99\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/db0492c69b2d49d1bdc884fcb6d02663 : SUCCESS in 3m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f70582e8dbb42628d9fbb894fb504cf : TIMED_OUT in 2h 05m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0896e1758a064ca29fca60d227c4587d : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe5da52966cc41498457e31f4beb4e69 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be1487a1934941aba2566474eef96858 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/967039e5c43f4267a13b996e71c5c717 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0cec9f8bea74a34abc276ee21828245 : SUCCESS in 2m 57s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/5f85f2dcc47b45d2926be340aa38d1dd : SUCCESS in 4m 45s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/28aeedd382ee4d50a6db736e22d4a7cf : SUCCESS in 1h 32m 35s","accounts_in_message":[],"_revision_number":3},{"id":"25040759bebca00dc032e8029d2cbc118c34c955","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-09-29 08:51:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0cb3c7fc6957d5bc514d7c7ecda9685024f61728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:22:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c17327db0df4ae0b0523d020e138588\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/007d72c7bf8f43a8bb308d087be60026 : SUCCESS in 2m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f9e29910b8a4c6a9d0d82ea08d4a446 : FAILURE in 1h 20m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e379f8cb4110412ca7a56c546cd8c068 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ee5446976c4fae9155b08f0e87b135 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/500417e30c73472498924403bf53baac : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a198f17456145228dc694e31eaa4caa : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ff0dde08ec44f008a428ccc57d45fbd : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22fc58226bf64750808dee09857ff0ba : RETRY_LIMIT in 2m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a6d0678566414d847c9d522864a8b8 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/952017c68c364a08852be5acd347b3b8 : SUCCESS in 6m 17s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/a92d35f74e9b4bfa87d2c1b3415b8b1e : SUCCESS in 4m 52s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/7558fc6dcbb34c54a6d2df0a6782080d : FAILURE in 5m 18s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/9ef0f5a617174ae9b8adc57c2f349fd2 : FAILURE in 4m 18s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/fc712567d17648bc8d17a7af77391499 : FAILURE in 3m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/028ead053bc2458b8a2149e804ecbe52 : SUCCESS in 1h 28m 40s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/72468dbd9d544adea13fb3a5f7d245a2 : FAILURE in 21m 14s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/bb2a36b958ab4d3b9364df2b233b0c51 : FAILURE in 1h 18m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/29ecbc13f3154b2f9b7ed83e4648a35a : SUCCESS in 1h 01m 14s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/28d9afcf620c41e8897e17c2df06414b : SUCCESS in 1h 06m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5372614bd04a5be15f62f88c980a3397906d380c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2023-10-11 10:43:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ed4141a72caeddaf5bc85ffc9978da93f8706b06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 12:20:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20945dd98b214769a6c0162697193812\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4959e730aa354d2e94ad0387efe0f4af : SUCCESS in 3m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9713f5d4cbac4bf3a6ae9fc64182db1d : SUCCESS in 1h 23m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/812fa9b75e934d8eb44376cab9413e7e : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ef37cd38ae74f6aa9448c193d772406 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcd7aa894f9847b293b8a300a47064a0 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/872c9c19a9b54bceba4f93b269a03aa6 : SUCCESS in 7m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8dfb5a4929a545c88a6e0d6dc5564d57 : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80a08fab7c7945a69f9b6b3e7aa4ce74 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d29949a3d34dfba1aa98eec031ed81 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34539e61be614cb0906b42c603a74dd5 : SUCCESS in 4m 14s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/914fc368b44749b2bd0b4582854c9339 : SUCCESS in 5m 17s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/081a2a7d6acc48df8220b3ef6ccd9992 : FAILURE in 11m 42s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/423198c492c44df8a40a5b79d71a7cc2 : FAILURE in 4m 13s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/70aedc1ad7b24dc2a05f71b428bcd476 : FAILURE in 4m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/de528b4903db428193e2b6b1869c7852 : FAILURE in 1h 29m 21s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4275d2deedd345f2a375d5fcd6f64810 : FAILURE in 33m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c32b4fb478ba4dc0936c90e37a9c8394 : SUCCESS in 1h 34m 21s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/fa85447e64264b49a5e99768e1c5efc1 : SUCCESS in 1h 00m 52s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b0870beb8674c00b96872f15f711fff : SUCCESS in 1h 24m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e8aaf305a3ff0e23dbe6c86d916c50a00340c7c6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-20 10:18:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"24fdabc559c63dfed46ecaa6809f3a6396efad9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 12:05:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ea119d66e9c46d6acf960380bb04fb7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/11b3d0488cce4420a1127e0740ac8289 : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b30e9b81e91461fafa7e239edd01dde : SUCCESS in 1h 25m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56534fab12524be4922d119da60c4977 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97faccdbc0e24b258fc153edf2756eb2 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9efbddb2c1e44fe9314c4083e7c9a6d : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c743751107847338084d225077f93cb : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b8fe1d3169c4b7c81d8ed842994c154 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fa6248c282b4f82be00a4ddd1b370d2 : SUCCESS in 3m 34s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/1a10e069ad6f46309300ed711947c711 : SUCCESS in 5m 03s\n- oslo.messaging-tox-py310-func-scenario02 https://zuul.opendev.org/t/openstack/build/f3fbed7e5caf498987ae8fd979e0490f : FAILURE in 6m 15s (non-voting)\n- oslo.messaging-tox-py310-func-scenario03 https://zuul.opendev.org/t/openstack/build/870c4c82573f46f18f3ac45a706e9ba7 : FAILURE in 3m 49s (non-voting)\n- oslo.messaging-tox-py310-func-scenario04 https://zuul.opendev.org/t/openstack/build/9f572d24cf0444cca802b1ecd33758a6 : FAILURE in 3m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2a88e61a26304f929467b97e6744405d : SUCCESS in 1h 36m 02s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/f976d6479beb4278b659c7bcfcad82d7 : FAILURE in 17m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d587407e87a44d2ba77d3180279652d8 : SUCCESS in 1h 38m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/77f28f34977741da9ac8c422bb4a7a2e : SUCCESS in 53m 08s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/93bb876594e444b7b0d077e13b5a3845 : SUCCESS in 1h 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"52859e94ebf6da63a0144dcb041f202138ba5d92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 12:05:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f9a90e1e37c2ac8aa1ce087de1727d635df84b6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 13:47:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce5430ae9d9d42b2a4e5fe6812318ac1\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/735ec6946cbe4b5a90d91fcb2a126f68 : SUCCESS in 3m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fcc449f6ebc479592016f89b4781075 : SUCCESS in 1h 36m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0060a5e5b8418b934e128291818ed3 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ac30ebd82444579a96b9306559122a5 : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f23a73338b284321b79d181962aee1fd : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/436a613cccf14e4d80968420b2ff7849 : SUCCESS in 3m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af1b812e269f414f8bb212df042ab537 : SUCCESS in 3m 29s\n- oslo.messaging-tox-py310-func-scenario01 https://zuul.opendev.org/t/openstack/build/2d3cf9e948f24cc79214f223207fd1fb : SUCCESS in 6m 09s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7403c864050b495383793dc61924b60e : SUCCESS in 1h 33m 40s","accounts_in_message":[],"_revision_number":3},{"id":"6ec7413ed924cdc837eb349f29f23068a72a645d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 13:47:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7a58e780a2629257b3c55945b2cf67a4a6c6f415","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 13:48:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ac3d03721254e3b89e25ebb5a09f5b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51cd69b60d9f465081230cf32c53b935 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d78f850002fb4155801ec41c73150749 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ee13e53614239572647e1ee0d4bba7dca1a860a3","revisions":{"4cdf01f8e2910776f56fc44bf7e9f454aaa3efcf":{"kind":"REWORK","_number":1,"created":"2023-03-10 09:55:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/877087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/87/877087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/1"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 09:52:52.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 09:52:52.000000000","tz":60},"subject":"Bump bandit and make oslo.messaging compatible with latest rules","message":"Bump bandit and make oslo.messaging compatible with latest rules\n\n- Apply a timeout to requests calls to avoid uncontrolled\n  resource consumption (CWE-400) [1].\n- Use the tempfile module to generate secure temporary file rather\n  than hardcoding it (CWE-377) [2].\n\n[1] https://cwe.mitre.org/data/definitions/400.html\n[2] https://cwe.mitre.org/data/definitions/377.html\n\nChange-Id: Ic558ad392424a25b5fd9a10749163d8427159eda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4cdf01f8e2910776f56fc44bf7e9f454aaa3efcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4cdf01f8e2910776f56fc44bf7e9f454aaa3efcf"}]},"branch":"refs/heads/master"},"d24afe2c94307f61c1075c519253abba584df124":{"kind":"REWORK","_number":2,"created":"2023-05-17 09:05:31.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/877087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/87/877087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/2"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 09:52:52.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-05-17 09:05:12.000000000","tz":120},"subject":"Bump bandit and make oslo.messaging compatible with latest rules","message":"Bump bandit and make oslo.messaging compatible with latest rules\n\n- Apply a timeout to requests calls to avoid uncontrolled\n  resource consumption (CWE-400) [1].\n- Use the tempfile module to generate secure temporary file rather\n  than hardcoding it (CWE-377) [2].\n\n[1] https://cwe.mitre.org/data/definitions/400.html\n[2] https://cwe.mitre.org/data/definitions/377.html\n\nChange-Id: Ic558ad392424a25b5fd9a10749163d8427159eda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d24afe2c94307f61c1075c519253abba584df124"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/d24afe2c94307f61c1075c519253abba584df124"}]},"branch":"refs/heads/master"},"ee13e53614239572647e1ee0d4bba7dca1a860a3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-05-17 09:07:02.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/87/877087/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/87/877087/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/87/877087/3"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-03-10 09:52:52.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2023-05-17 09:06:34.000000000","tz":120},"subject":"Bump bandit and make oslo.messaging compatible with latest rules","message":"Bump bandit and make oslo.messaging compatible with latest rules\n\n- Apply a timeout to requests calls to avoid uncontrolled\n  resource consumption (CWE-400) [1].\n- Ignore CWE 377\n\n[1] https://cwe.mitre.org/data/definitions/400.html\n[2] https://cwe.mitre.org/data/definitions/377.html\n\nChange-Id: Ic558ad392424a25b5fd9a10749163d8427159eda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee13e53614239572647e1ee0d4bba7dca1a860a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee13e53614239572647e1ee0d4bba7dca1a860a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
