)]}'
{"id":"openstack%2Foslo.messaging~878588","triplet_id":"openstack%2Foslo.messaging~master~I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b","project":"openstack/oslo.messaging","branch":"master","topic":"nats-py-driver-without-async","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-01-19 13:07:53.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b","subject":"[WIP] Add sync NATS driver","status":"ABANDONED","created":"2023-03-26 15:36:28.000000000","updated":"2024-01-19 13:08:19.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"484019573153dc4f881a1142d33260c538bb60be","_number":878588,"virtual_id_number":878588,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-12 12:54:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-26 17:07:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-20 09:55:04.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2024-01-15 16:12:03.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"}],"messages":[{"id":"9dfbd8892aed8ef305a1b0816eefea46ab72c7ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-26 15:36:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e352d5a6bc3155451be97b3c29583a3655e25cdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-26 17:07:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03a41125797a4273b5a6494afc0a0120\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b6923217b8e6492eaad52d6f65316e1d : SUCCESS in 2m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cea5f673bb374faf88b93de3615bd827 : FAILURE in 2m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60cda9d21d624f598f9cf0497f7256ad : SUCCESS in 1h 25m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3811017323154579b140c4b1c3712143 : FAILURE in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2559cdfc4a845e1b80e12002844fcc5 : FAILURE in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a3d98ff853b49afa4f34c6ec98a31d8 : FAILURE in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cd7ccadc75647d9acb444cf13793518 : FAILURE in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c3c5e3dda7c4a0f912ffc5aab8b46f0 : FAILURE in 2m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372927a867184070a852b04c9dbfa6ad : SUCCESS in 4m 39s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/6bdf7e23f8384f9ea3a0dd72f6affc3d : FAILURE in 2m 27s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/8123e3b0176847478f16a2b1325b7204 : FAILURE in 3m 12s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/27fd8ed023194a6d947807a91c3d2c9b : FAILURE in 2m 56s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/7d8740558bd541f1b5bedc650bdcc06c : FAILURE in 2m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ac98b64c10ce40e6903bb618b66248cb : SUCCESS in 1h 23m 55s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/7c0e1f904be340c6aa6ab9f6dcca5c65 : FAILURE in 27m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5201d142094746feb242dc1e2c4449b6 : SUCCESS in 1h 23m 05s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d9045f65ba8a4c8ea5f20f1ed7bd323c : SUCCESS in 1h 00m 06s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/598083f7e0884fd19547e23a21eaad04 : SUCCESS in 59m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a65e36a368a1e45aa1c407c844ba3e39196f7802","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-26 17:58:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2926141d87337187d6fbda680a75eb870ca12415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-26 19:43:45.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13e48ea9dd304f76a7b319dbccdd4c04\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1fd223aa2dee43f48830dc801c7df6a1 : SUCCESS in 2m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a33b0f331c0b4e8dbd79c7290aaa430e : FAILURE in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8126a713fea448c489767071a59a8635 : SUCCESS in 1h 29m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a90aece0b9ed473a8de6561aafd9832e : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2946bc3d63dd467993d8d2223773c8f4 : FAILURE in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a113a2e3fa0344388d44c373544c619e : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9632f43b963f41f29559c4a1056d37ce : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9521799373684ecf9d07507dcab94e97 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4da3fcb7e4cf43b6be2fcc8e5c78dd67 : SUCCESS in 6m 51s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/2eb451c1036a47588056f6c64eb39aeb : SUCCESS in 7m 13s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/a09a04533ae640b1849bc04d79806163 : FAILURE in 5m 46s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/0bf60a556f5340b483afab455b8ba30c : FAILURE in 6m 18s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/dca4b9b3d54e4123b5c48588e6812163 : FAILURE in 4m 06s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b92ffb1b258e4bb9bbdc55aa83c743fa : SUCCESS in 1h 15m 51s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a9c73005e4434a4b88218bd3a70ea7be : FAILURE in 20m 57s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/c02e7f91600a470da16ecd3d1d8a552c : SUCCESS in 1h 17m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/18137037779c4954b2224a306a4dfacb : SUCCESS in 40m 43s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/67c0a0227a564a1a97c0e8096425f003 : SUCCESS in 55m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"37cbcfbd985b8476890a485ce4880fc0b6b109f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-27 09:23:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"731b4c4b6f6c70d6f1bdaca4afbd9d6c463958b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 11:13:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2b5cce8e1fc4183934ba32d4464d5fe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8dfd5f28f5c147309b9d633bc6eed41a : SUCCESS in 2m 06s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/251cad0d1bd34be98639a2ba8008bd22 : SUCCESS in 1h 47m 54s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/2aeb08b7a29b4f2195b23542beed8622 : SUCCESS in 1h 33m 58s","accounts_in_message":[],"_revision_number":3},{"id":"5cca013778ccd9c958bba5c335003573e0244a8a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-27 12:10:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef55a7202974839a90c31f840d905344b9db6f84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 13:30:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2762bbc25ec409abd94ebcee63b68e3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/251e1c2462204b6990972ea11ba0c763 : SUCCESS in 2m 29s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/59e6d4c6562648888ba6e09febb81628 : SUCCESS in 1h 17m 41s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/86f5986b068746b5b19515670d88e9f6 : FAILURE in 19m 48s","accounts_in_message":[],"_revision_number":3},{"id":"102a22ecc45711aedb11c82b0ecbceb2702f0516","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-27 15:14:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"75f15987355ee3895d74c06ef7eb12627b0a7b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 16:27:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5009d3743f6e4f1d8c0bb96f3ec47423\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b89aeb12e1524a40898318cde8f5ea5f : SUCCESS in 1m 58s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/303b4972dd5c4a03a020824648e43583 : SUCCESS in 1h 11m 24s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/1e334614203646fabae7822914198cc2 : FAILURE in 28m 15s","accounts_in_message":[],"_revision_number":3},{"id":"0177159d2f4c165f07db1af91fb18ac8c8f6b8d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-31 08:45:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9d196ffac98e71e40979a0fcf0ee1316cbdccbad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 10:12:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71c539ba886442e4a658891103dc4be4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ba23bf3a3c9d40fa81ba9cd23c68f286 : SUCCESS in 2m 24s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4d522d2326cd43098ad11a73ba4a82c6 : SUCCESS in 1h 26m 17s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/52919b0029f549c4bd520822219d9d3f : FAILURE in 28m 13s","accounts_in_message":[],"_revision_number":4},{"id":"79d77127a1aea660a99676b2870d4e3370f1ed57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-31 12:41:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"aec33f2dab082882040b9078906043c52dcfb53d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-31 13:53:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6286aa010f7ce150b563c1d42b31eb9a28fe37ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-31 14:25:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4e51fd4fd3f736247ff3c82a32c953a6f8cf963e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 16:10:35.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f62dc6ebcc4448f280915f885303cb5f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/23c7734c90934cdaa0ecc43cdb794e40 : SUCCESS in 2m 33s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a94692134f5e4cdab331fcf113ee9ed3 : SUCCESS in 1h 39m 10s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/566b02d94cdb4825ae0f1a60ca383ae0 : FAILURE in 20m 50s","accounts_in_message":[],"_revision_number":7},{"id":"b153d1da48b6de7943fddd9eb2758ddc422fa0eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-03 06:30:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5f3360da1348840425afa047479eedadb68a72fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-03 06:38:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"035869ff8abf9e43ba0c176422c87649db6f8e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 08:18:17.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/338994391acf4ba4bff1107202a5e901\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/38c7d3143a6d4561bd2d17d7d479bd57 : SUCCESS in 2m 30s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d772fa4ed1a7489a9c7ac2004f0c4bb5 : SUCCESS in 1h 31m 56s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/dac147dc74454c9ba0c10e159e4e4d72 : FAILURE in 23m 44s","accounts_in_message":[],"_revision_number":9},{"id":"8d8dcc9ab75d19712182e9451b655a3090665534","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-03 08:53:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bb05a7b069c6f594ed73e815ca9fd25d72c6cc53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 10:11:27.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1e1233be3d84e499a707e6ff6a2ba97\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7705fd2bbef442d7a2590c532b7e4ebf : SUCCESS in 1m 47s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ba9523734bfc4733a8208ca1ae715c00 : SUCCESS in 1h 15m 30s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/285f2e549ba24bc8b2e6cb0d34e741ab : FAILURE in 33m 17s","accounts_in_message":[],"_revision_number":9},{"id":"a3d01d262b1565ec739b7492c35d535316cd60c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-11 20:54:13.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"d9ff8b201c3a2dbb47e62c4cb915570c598b59db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 22:36:30.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42dfdba7889f417c84d25925b93db653\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1b8036bbe44d4e848f635baf37f8a341 : SUCCESS in 2m 13s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9a2936a4ecf1435c9697136ac4f204e5 : SUCCESS in 1h 36m 54s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/4dfc6c7475cb4466873559b3f774e01a : FAILURE in 17m 05s","accounts_in_message":[],"_revision_number":10},{"id":"7dd768a6b70abf75e3f9a136d4f6ba983fb05d70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-04-12 11:13:09.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"9c1523d38f68528bb561de5341f94a59eca94704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 12:54:19.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8325aa9c35d48c286e80b03a591f53d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd4008163237442d94d7ade8b8959181 : SUCCESS in 1m 46s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/614576b74f184d2b93071474c069c318 : SUCCESS in 1h 32m 52s\n- oslo.messaging-src-dsvm-full-nats https://zuul.opendev.org/t/openstack/build/6520c66a290145b6b897be9115a60f5d : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":11},{"id":"f10b77088e237561a3624ec774336010a32333d9","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-15 16:12:03.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6c4dc9dc46b335d48f97f94762bc969f7b32aba0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-19 13:07:53.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"484019573153dc4f881a1142d33260c538bb60be","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-01-19 13:08:19.000000000","message":"Abandoned\n\nHold your thumbs for a eventlet alternative to make this possible","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"fa4d7b2a4782151f4a7abbac080fc3701bb56883","revisions":{"8e36e6f1087658c8957310c0d22c72481e848222":{"kind":"REWORK","_number":1,"created":"2023-03-26 15:36:28.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/1"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-26 15:36:17.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8e36e6f1087658c8957310c0d22c72481e848222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8e36e6f1087658c8957310c0d22c72481e848222"}]},"branch":"refs/heads/master"},"aeb12180b49a00f8bc6780cf2f1122cde86e128c":{"kind":"REWORK","_number":2,"created":"2023-03-26 17:58:39.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/2"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-26 17:58:33.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/aeb12180b49a00f8bc6780cf2f1122cde86e128c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/aeb12180b49a00f8bc6780cf2f1122cde86e128c"}]},"branch":"refs/heads/master"},"6e3331166bd8591cd08b45a05a0b95ff77e36e2f":{"kind":"REWORK","_number":3,"created":"2023-03-27 09:23:59.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/3"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-27 09:23:54.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6e3331166bd8591cd08b45a05a0b95ff77e36e2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/6e3331166bd8591cd08b45a05a0b95ff77e36e2f"}]},"branch":"refs/heads/master"},"8697a1987e2f5cf8f0e3766df94d9ce4b0c3f109":{"kind":"REWORK","_number":4,"created":"2023-03-31 08:45:19.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/4"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-31 08:45:13.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8697a1987e2f5cf8f0e3766df94d9ce4b0c3f109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/8697a1987e2f5cf8f0e3766df94d9ce4b0c3f109"}]},"branch":"refs/heads/master"},"21cb11cf6e2483fa3dc8973e7f26316ff6547a23":{"kind":"REWORK","_number":5,"created":"2023-03-31 12:41:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/5"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-31 12:41:03.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/21cb11cf6e2483fa3dc8973e7f26316ff6547a23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/21cb11cf6e2483fa3dc8973e7f26316ff6547a23"}]},"branch":"refs/heads/master"},"a3a60810a0fccfbb362267b87fa9f5c2168df2b8":{"kind":"REWORK","_number":6,"created":"2023-03-31 13:53:44.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/6"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-31 13:53:38.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a3a60810a0fccfbb362267b87fa9f5c2168df2b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a3a60810a0fccfbb362267b87fa9f5c2168df2b8"}]},"branch":"refs/heads/master"},"bbcd12e5cd32ced45ad3b511ebf52fa165d2981e":{"kind":"REWORK","_number":7,"created":"2023-03-31 14:25:41.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/7"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-03-31 14:25:35.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bbcd12e5cd32ced45ad3b511ebf52fa165d2981e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/bbcd12e5cd32ced45ad3b511ebf52fa165d2981e"}]},"branch":"refs/heads/master"},"31b38dcfb12587bf500cf3a2b67541cdad9d688a":{"kind":"REWORK","_number":8,"created":"2023-04-03 06:30:29.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/8"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-04-03 06:30:23.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/31b38dcfb12587bf500cf3a2b67541cdad9d688a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/31b38dcfb12587bf500cf3a2b67541cdad9d688a"}]},"branch":"refs/heads/master"},"1c74736935a8ebdd8fcd6ba826042b5fc0a4c6c5":{"kind":"REWORK","_number":9,"created":"2023-04-03 06:38:42.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/9"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-04-03 06:38:36.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1c74736935a8ebdd8fcd6ba826042b5fc0a4c6c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/1c74736935a8ebdd8fcd6ba826042b5fc0a4c6c5"}]},"branch":"refs/heads/master"},"af8b4b40b5c383b741a8293aad7fc3ffb2d72d94":{"kind":"REWORK","_number":10,"created":"2023-04-11 20:54:13.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/10"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-04-11 20:53:56.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/af8b4b40b5c383b741a8293aad7fc3ffb2d72d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/af8b4b40b5c383b741a8293aad7fc3ffb2d72d94"}]},"branch":"refs/heads/master"},"fa4d7b2a4782151f4a7abbac080fc3701bb56883":{"kind":"REWORK","_number":11,"created":"2023-04-12 11:13:09.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/88/878588/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/88/878588/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/88/878588/11"}}},"commit":{"parents":[{"commit":"9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/9b1e2dc48e83c225202955c8c7affe5e0bcf6d8b"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-06-29 15:43:33.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-04-12 11:12:57.000000000","tz":0},"subject":"[WIP] Add sync NATS driver","message":"[WIP] Add sync NATS driver\n\nChange-Id: I6256ef97c1a3ccc1d93612dad96bacbef4cbbb9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fa4d7b2a4782151f4a7abbac080fc3701bb56883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/fa4d7b2a4782151f4a7abbac080fc3701bb56883"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
