)]}'
{"id":"openstack%2Foslo.messaging~880187","triplet_id":"openstack%2Foslo.messaging~stable%2F2023.1~I34ea0d1381e934297df2f793e0d2594ef8254f00","project":"openstack/oslo.messaging","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-09-14 04:54:02.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-09-14 04:54:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34ea0d1381e934297df2f793e0d2594ef8254f00","subject":"Disable greenthreads for RabbitDriver \"listen\" connections","status":"MERGED","created":"2023-04-13 10:57:29.000000000","updated":"2023-09-14 04:55:04.000000000","submitted":"2023-09-14 04:54:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880187","cherry_pick_of_change":876318,"cherry_pick_of_patch_set":1,"meta_rev_id":"b7d12c9582eceec91139d964d65534a1a15a6251","_number":880187,"virtual_id_number":880187,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 04:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-05-15 09:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-04-13 14:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-14 07:13:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-05-15 09:55:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 10:57:29.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2023-04-13 12:52:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 13:11:35.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-04-13 14:12:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-04-14 07:13:39.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2023-05-15 09:55:57.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-05-22 19:01:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"5a949e8d3f286d3e4e40d765109859bfca88f4e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-13 10:57:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c5eca34405829c317e89fb6fa8b190288c2c1f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 12:52:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2eabb36527a441192c275ababb1d9a4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82b10f59a7b948dc8d72fbf41accc50c : SUCCESS in 1h 46m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/825bbbd6cd7b43bf99c7e0406d1a3d7b : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a62511a8f2c94a8a9be4adf62dd0ed7c : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7b2eb1006d44ae7868c42e50710216c : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62e3f9c49aa74d24becd210f25a7fa19 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a36b5b1eee441569e91847488047d9e : SUCCESS in 3m 52s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/ab297dfe6ea9467b9dc9c9f3ac761853 : SUCCESS in 4m 35s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/ed691c5a931241318206a319d1e25605 : FAILURE in 5m 01s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/e66acd6bc46c4e54bf0e360e6199999c : FAILURE in 3m 33s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/93f6c3851a83460a83e39c5f89896620 : FAILURE in 3m 47s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/21f8691e571d4ad5aa7bc8491c4fb06f : SUCCESS in 1h 29m 25s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/cf58a42aee884ab8b540bd8e0aacd4fe : FAILURE in 28m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/d783c3174da8485a8e898bcb3e10e284 : SUCCESS in 1h 37m 11s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/18fb430f347148ca94046b63b3e1a8c7 : SUCCESS in 55m 54s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/065b183e90814d30845b5325c1a80e4d : SUCCESS in 1h 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fd50eb2e32837e4f55edd66a9da6b408f9717b5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-13 14:12:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4c7cad3014b82a5f2e8e8e20c69ad7dc6f0fa14","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-04-14 07:13:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"21cc267b95d2a3987e2fd0156ce45d5c8dfb6796","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-17 15:09:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57b4207763e6cbbad7e118616384f9e9c84cb6aa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-15 09:55:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"de1ffe947f194a7c38bc9dfc1e0f19bc9c4c1362","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 09:56:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b817533e0f36cdfd173bf641ef6b317bbf23d2bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 11:25:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b78d0fef223d405a91f6d39acb6f713e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64ad32a38c82412aa697cbddb624db7a : SUCCESS in 1h 29m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d6054ff1eb4fad9708b8d4cd2f7636 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be79247b048d465a8c5edf951481aeea : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/999109b876e64c36b56fbd3fc895de5c : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ead71becf5d4761bb2a420e48633d91 : SUCCESS in 4m 09s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/fe5fff3591fa4ab0aa9a0c1794a7ed4d : SUCCESS in 4m 17s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/54c3b372a33b40e1b3e6555bbbf27ccf : FAILURE in 1h 28m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3cd07de496444bd07ff5046b9435630d412e9ba4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-05-22 19:01:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"542d7d005036611db9b73be6ea463e30e457b5b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 20:48:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55d546ad3f1d4f21964a68490992d7a6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d5969600a7149bba50fb9db522b4d85 : SUCCESS in 1h 11m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d28e6bff1c314a62b15910c688e6aa21 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/170c1711e92b40b4a4859f363b665d62 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4982d0be9f444a12b35fffd83a23d259 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e81361e0825438ba19bacc93eba697e : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8062abe831b1455a88348901f8e68176 : SUCCESS in 3m 43s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/477ca8e150fd408a82fe778cc35b0560 : SUCCESS in 4m 57s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/b3bb4a75976644719e6dd5f22242c9e3 : FAILURE in 6m 14s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/122a75fe79714a01a615f5060e5e0472 : FAILURE in 5m 18s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/a782c939704343bca0c3b92e955851a9 : FAILURE in 3m 48s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/9c051643bb9a432fb67cc2219c952ab6 : FAILURE in 1h 21m 20s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/deccf27a39ab43079e537279abbacfb4 : FAILURE in 35m 08s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b38ecfc816b24fc6a672f7a27189cc9a : SUCCESS in 1h 38m 51s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/cd241d82c60845529279f6ff79422da3 : SUCCESS in 55m 59s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/1393b765fb094455a7d93a156e6d713c : SUCCESS in 1h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b85b630729f10b4affadda2b9b6e39faea15bf19","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-12 19:46:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e520a7f7837017e6a86d6a839b18cc610056696d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 21:04:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8e6717167e243ccaa9c49c6c16cc633\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/484df3f69b4c48f19dac8b406ed854c0 : SUCCESS in 1h 04m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a7400d3849c4e2083075c6cc1e70872 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22bc446efca54cd1b89ae2ada06cb5d4 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20412cca036c4ad1a71c6aa718dad540 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6df90f64dca64028b94946cb871d0eb5 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb266974f71487392f1556e0278782d : SUCCESS in 3m 33s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/623200f4950849588c0c70ba6dafc2bd : SUCCESS in 4m 32s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/0ef17f3a16dd4129aaa0c931dae2000b : FAILURE in 5m 08s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/137df9d57b7f4e10a80c8681b3739d22 : FAILURE in 4m 21s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/767453e8970246dd80cfd8c8464867ca : FAILURE in 3m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b9165d2a63d243cfb1048dcef02aefb8 : FAILURE in 1h 07m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/dbfce4fbacd24fb18195b25b2d89ad77 : FAILURE in 14m 40s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/ca375246a0c6485c87aedcfde0e31247 : SUCCESS in 1h 16m 58s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/44516a2ca9974375bb4d18b2173ef5ea : SUCCESS in 39m 40s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/198aa6d4e31c4a79a09bf332e7f923ad : SUCCESS in 54m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ca9c3c2beca8409da6ffe6107b74d5f4d254c86","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-19 12:17:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31c7cc809a4e9cf2f0f416b0473f46ada6d1e774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 13:59:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65370c47c8944baa83efe355d0d9e0ba\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c85b336f2874b0d9d0fd5a5de21756d : SUCCESS in 1h 07m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d93b22a499384668a9fa78b1df9dee25 : SUCCESS in 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f4dff08e214dd38ca8204336960e87 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab3fe7ea91ff4f6f8a31420e1fe33192 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8283071c0f194b399960ddd9534ae624 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d103a9b077c0402198d3aa06e7a471ce : FAILURE in 4m 36s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/ab66e28c6b3d44fdba6ab5aee0e3804e : SUCCESS in 4m 53s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/15a57b9f9b714389949a2955780c716a : FAILURE in 6m 20s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/9e7887719090468b976529ca668be69b : FAILURE in 4m 49s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/aca6a962ed4242b3b7c95ef3baf18b72 : FAILURE in 4m 31s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/fc477069adfa4f97b8d89413aab1f75a : FAILURE in 1h 22m 30s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ff774a8480ce4344b51e37b368104b52 : FAILURE in 15m 56s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/2ab90d8061d2413a87617126e19fce86 : SUCCESS in 1h 40m 25s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/d10b40bc199c4ad18d9cbe5b0b556699 : FAILURE in 54m 25s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebd1a6370f2546abb14734c8e872a242 : SUCCESS in 1h 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd8102565cb277e76d8ce8939dc8982207fb3573","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-14 01:46:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"170a82fce96a20c1f79f060e9e18eb1b972d1bab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 03:24:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b21e287ce4c481aaf54a4467048c9d2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68254dcf3abb4af9847fdb9317c5de4c : SUCCESS in 1h 28m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75a75d071ac64591941e3d7f467867fc : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f6b21dd6d8449239c6bd9c5fa6771fd : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea680cfbac40449b9607a9cb1f87b28e : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f4d86eedc42450d9d1c2e301fe340b1 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e7d5a4b44ce4f57abf96330ac9c5c05 : SUCCESS in 3m 46s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f4437d904a0141a3a616dc82b7e6a51f : SUCCESS in 4m 32s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/daf511b28f9949adae0dd1fae9a52286 : FAILURE in 5m 58s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/a76c79f8b87146be8b163e2a7056e6f6 : FAILURE in 4m 47s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/2ee156b32515478a84d30a0bded84bc1 : FAILURE in 4m 18s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/923975f7434a425f9ccee7e90aa9713e : SUCCESS in 1h 18m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/33f7d58fd9d7457e9ab04ca0ee1f63d5 : FAILURE in 16m 54s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/62be623d3585469ebdcc56dccefabfc5 : FAILURE in 1h 20m 30s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/daddc1ab67ad4ddda46bf7ba3a04db73 : SUCCESS in 1h 05m 20s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f1ee85d2625432db257e5ec45dceb7e : SUCCESS in 1h 04m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"610dd059296bebe508e44e74f1c0911fb8f6268e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 03:24:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6733122bddd672497d6e8588bf68742aedfb0f71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 04:54:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbfd2e8d2220419f862527ffef8c7c10\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3ba96894bb74894be23e4515852d0f9 : SUCCESS in 1h 24m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/824d4ab997a04d368e6cdae7d3542b17 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c53778e79ad447bbd25a90c23d01c17 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/397f8b1c1af64c03bcc51b2c05981c75 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3fb2e5059024c6e889099f0f0fcc760 : SUCCESS in 4m 41s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/9fcff6235492441394d7bd6f924611c8 : SUCCESS in 4m 32s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/62ae719c192a42e7b8b7dc1fd95fde62 : SUCCESS in 1h 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d437e8e546f49c9ba498c165e612127aa6f3b3d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 04:54:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b7d12c9582eceec91139d964d65534a1a15a6251","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 04:55:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f793904a92941adb38cd6038af8850c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33dda61bff494d4fa0d29c55c88ce472 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3645839d162a989da46ba1c0fcbe7f32b48b9fb1","revisions":{"3645839d162a989da46ba1c0fcbe7f32b48b9fb1":{"kind":"REWORK","_number":1,"created":"2023-04-13 10:57:29.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/87/880187/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/87/880187/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/880187/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/880187/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/87/880187/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/87/880187/1"}}},"commit":{"parents":[{"commit":"50d87b95b6583bc6a4f02c5a04cb7151d2546289","subject":"move py3 jobs from zed to antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/50d87b95b6583bc6a4f02c5a04cb7151d2546289"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-03-03 10:16:56.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-04-13 10:57:29.000000000","tz":0},"subject":"Disable greenthreads for RabbitDriver \"listen\" connections","message":"Disable greenthreads for RabbitDriver \"listen\" connections\n\nWhen enabling heartbeat_in_pthread, we were restoring the \"threading\"\npython library from eventlet to original one in RabbitDriver but we\nforgot to do the same in AMQPDriverBase (RabbitDriver is subclass of\nAMQPDriverBase).\n\nWe also need to use the original \"queue\" so that queues are not going to\nuse greenthreads as well.\n\nRelated-bug: #1961402\nRelated-bug: #1934937\nCloses-bug: #2009138\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34ea0d1381e934297df2f793e0d2594ef8254f00\n(cherry picked from commit fd2381c723fe805b17aca1f80bfff4738fbe9628)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3645839d162a989da46ba1c0fcbe7f32b48b9fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/3645839d162a989da46ba1c0fcbe7f32b48b9fb1"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
