)]}'
{"id":"openstack%2Foslo.messaging~880189","triplet_id":"openstack%2Foslo.messaging~stable%2Fyoga~I34ea0d1381e934297df2f793e0d2594ef8254f00","project":"openstack/oslo.messaging","branch":"stable/yoga","topic":"disable-green-threads","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-01-29 18:56:08.000000000","reason":"Change was submitted"},"11221":{"account":{"_account_id":11221,"name":"Ian Kumlien","email":"ian.kumlien@gmail.com","username":"pomac"},"last_update":"2024-01-29 18:56:08.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-01-29 18:56:08.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-28 08:06:44.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-01-29 18:56:08.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-11-09 15:12:52.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-01-29 18:56:08.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-05-04 08:52:58.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I34ea0d1381e934297df2f793e0d2594ef8254f00","subject":"Disable greenthreads for RabbitDriver \"listen\" connections","status":"MERGED","created":"2023-04-13 10:58:04.000000000","updated":"2024-01-29 18:57:58.000000000","submitted":"2024-01-29 18:56:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880189-disable-green-threads","cherry_pick_of_change":880188,"cherry_pick_of_patch_set":3,"meta_rev_id":"104a84c21be1e6dcad110ac0d276490b5d4ae6dc","_number":880189,"virtual_id_number":880189,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-29 18:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-01-28 08:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-29 17:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2024-01-26 12:33:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-29 17:22:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"CC":[{"_account_id":11221,"name":"Ian Kumlien","email":"ian.kumlien@gmail.com","username":"pomac"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 10:58:04.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2023-04-13 13:11:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-04-13 13:45:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 14:14:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-04-13 15:14:33.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2023-04-15 13:29:01.000000000","updated_by":{"_account_id":11221,"name":"Ian Kumlien","email":"ian.kumlien@gmail.com","username":"pomac"},"reviewer":{"_account_id":11221,"name":"Ian Kumlien","email":"ian.kumlien@gmail.com","username":"pomac"},"state":"CC"},{"updated":"2023-05-03 08:03:18.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-05-15 09:14:01.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-05-15 09:14:01.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-05-15 09:14:01.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-05-15 12:45:02.000000000","updated_by":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"CC"},{"updated":"2023-09-14 13:19:17.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-10-25 20:28:02.000000000","updated_by":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"reviewer":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"state":"CC"},{"updated":"2024-01-23 09:30:02.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"},{"updated":"2024-01-26 12:33:46.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2024-01-29 17:21:55.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"1299a2632884bbf22770615d19cc661079da651d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-13 10:58:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d7f1113b616f7ee527f40c036480f6bcd125d4fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 13:45:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3db5bc419dfc447788928be0b9d70060\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3c0c3e2689949db8cbd7ec409358567 : SUCCESS in 1h 15m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f6469b88c364072991f83080b91d0d4 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943604054cd942c3b7dbf72f79f604cf : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72c6acea4dd944a7816a683fb44c965b : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdf5f6c2d37f42aea69d834639a4e9b9 : FAILURE in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ccd6b2f0fb439b93f9f69767601e86 : SUCCESS in 4m 04s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/cbf391af46e048b7ab0895f163d2a899 : SUCCESS in 4m 57s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/ea30a3a1744a4556b7b090f48ae7b918 : FAILURE in 6m 04s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/a8bf819a0a874f63ab7fdcdb2fa4786f : FAILURE in 3m 48s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/69797b4460d148fcb43952e9083287df : FAILURE in 3m 50s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/84536acc6b154205a3d9f69626dcb12f : FAILURE in 1h 20m 41s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/94bc31f13b6847149d1648e719bd52bb : SUCCESS in 1h 47m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/15e8fc0aeded4b899a468b24b4bc2857 : FAILURE in 2h 41m 15s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/752c577699624b7da991036f069dd25c : SUCCESS in 45m 03s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/04fd43fdb52d4532a4cd9210bb058169 : SUCCESS in 54m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"230ff5813b3c138ffbb641e80c0844ca7cb6ea26","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-04-13 14:14:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6696b0b32a232fca17ea7c4b1143c32bb4895c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-13 15:13:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3994b2fa4fe09a115cf021b7967c1a5928387d8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-13 15:14:33.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1462e04782b220ed733a3014ac74901a036c03c7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-04-13 15:15:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e96c17b2a9941c2b3b0e6c608ddf7959408b0296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 17:14:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3bc5a975dd040c1a3e44225c88a2d6c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60392a3c2c3243e98ded9fcb18d6f46e : FAILURE in 1h 16m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75d84c6a85c344a39d656333248f87a2 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba2a39b29c1e41f9b98191b259a6df0f : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60e1f576a170459baf2644955b8bdb42 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1c6565ca5464949a1dc2d9f3d2b39d1 : FAILURE in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe40acc335044c58ca8ed7b34de9ef9 : SUCCESS in 3m 30s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/aa73b97a087c4464bbb119879db6ca11 : SUCCESS in 4m 55s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/06c5b4b372ac4bafa80bcc8704492d98 : FAILURE in 5m 57s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/90164cd4dc574c559ce10ce56f409ac8 : FAILURE in 4m 30s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/e3706ebfc6b74414aab9ee3be1c0ef21 : FAILURE in 4m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7b6af492a43b496abe086d01d6319de3 : SUCCESS in 1h 42m 43s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/4d62f0b50b98424eba967a8ed2e524e9 : SUCCESS in 1h 51m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/57fa0d2e00e945d8b4009af3c4b7ff44 : SUCCESS in 1h 34m 47s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/48ecc05b1fc54bbab97e04db8a1bf515 : SUCCESS in 1h 09m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/15d9ddb1e48349e8bb6cacb080509af2 : SUCCESS in 1h 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"58e5b63893304265e56b5a2798a8cbf5ee7197d8","author":{"_account_id":11221,"name":"Ian Kumlien","email":"ian.kumlien@gmail.com","username":"pomac"},"date":"2023-04-15 13:29:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9416c0e7f9126e82d5d6248f4e32a5ee5ce852bd","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-05-04 08:52:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"93cda87641d397690f1e047b87c7070b9af02627","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-05-04 09:14:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d843ed6e3b632329828c277ed0d99e8f90112ee3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 11:01:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9149081735e4a9eb8201ce1170ff27a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6323503deeb04bc1b0c07ae89ac995b8 : TIMED_OUT in 2h 04m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/480b388aad6c41598145ef295b2db3a9 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53c69f72c95e4191b0eb64dca43c3631 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8298d8efeae84e1abcb1d2f7ff0f5621 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eea171e6c4ad4ed9aa4daadd4bdbe447 : FAILURE in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816b0ff3d8c94dc6a464f70c9974f4c6 : SUCCESS in 4m 46s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/98eabe3b2fd1423c8f9ee948a195b92d : SUCCESS in 4m 50s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/25d9259ced27473cb1dbe42a835a0e5a : FAILURE in 6m 24s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/c6291a8c0032442e917b658e54c5e376 : FAILURE in 7m 37s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/1fe947a0295f4e2e8c1a7cce272a0ba4 : FAILURE in 6m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/4e8f7e41b7254ee2b307e2954beee1d7 : FAILURE in 51m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/96152df5d83341b9b8626a8e61ffb094 : SUCCESS in 1h 28m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/74cd68eef1184646a4f3bbba04928fea : FAILURE in 1h 46m 27s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/135b5682ca684f5eb130fdb92b9449e6 : SUCCESS in 55m 15s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/8496385a8f6d4deeaaaf6da3de371154 : SUCCESS in 55m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d3b0a997d987dfae9da11eac1958dbcbbabbbcc9","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-05-15 09:11:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8937a687a2f1682b23b3d7469491b77f0df97dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 10:38:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dc8264f068f4b06a2d48bbc353b96db\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0f184b0a1d0421cb16e5bc86628e221 : SUCCESS in 1h 22m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fa4a28880ef44fab3fd507619b89a10 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80982c2a4e3744f3b138d14148f6ba88 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/047640c17df44b6b89b517112d53c8a4 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fb45c57edaf47409d03c6c250d9c755 : FAILURE in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84dd0171de004333ac27ca476fcdade0 : SUCCESS in 3m 17s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/e9f97f314ace4fa182f202bc31786098 : SUCCESS in 5m 36s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/6889e6e38f3444da97ef2944704556cc : FAILURE in 6m 07s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/76073d93b9064b59bb84c82da6ab8dfc : FAILURE in 4m 54s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/537127cec1684445943066af5f8de60b : FAILURE in 4m 43s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5a7a9bcd766d41f0985d78781ab148f4 : SUCCESS in 1h 25m 21s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/6a9f8179bec540d69269749fe30a63b3 : SUCCESS in 1h 16m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/e4d7c27596744d258909cc8cb0a71e0a : FAILURE in 57m 44s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/67eac1dd51e049f2a14261c040d2d3ad : SUCCESS in 57m 24s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3744303a8ee4679b693eaec458624b9 : SUCCESS in 48m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09b21eafdfbad06255ae92714bb3cc46cfb0634f","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-09-06 12:19:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ece40dc59935575d00de0d6cfd65a18bc95344e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 13:59:36.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc13086d77ab42469a262d7b9562d208\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76a2da22f6874172959fba349fd13357 : FAILURE in 56m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ff8a2873b02435ba295ebe5ea7d5ca2 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/563ae463aa2944ae9e3ac751f5de20c2 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b722a9c7db645279bc0f11e78a41ccf : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e92599e03e445af9ba3fdc41dbcbee2 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696fe9c6aaed423f9644e2ab0a239d29 : SUCCESS in 3m 51s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/bc443d683de846609071849a287281e8 : SUCCESS in 5m 20s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/f0ac23236fdc4f22b083a5b3af357b1e : FAILURE in 5m 56s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/99469f5aef3e4a28a67f305fca5637ad : FAILURE in 4m 36s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/fe5f99a5993f422db9e342409da71990 : FAILURE in 4m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e03eba3ce203492c9d7821aae0c4508d : FAILURE in 1h 07m 41s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3bc610f4a7c44717aabe936599a78105 : SUCCESS in 1h 34m 09s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/527cb2fbb6094f54b121304a8ab810d2 : SUCCESS in 1h 33m 25s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/54393ca252784b68b1eef69eb13c7bad : SUCCESS in 41m 29s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e11df2e9ff4e4371b6f54b847ba6d6ec : SUCCESS in 1h 05m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"87c85d01346107875fc368ffa837cf86a476d5ce","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-09-14 09:23:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee00e8218c4580fc5e2ff4cb74233073a0480862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 11:18:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8396f083056544caa3b43607ea1ddc93\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9c488ee12e04cf88603c100f8ac1a28 : FAILURE in 1h 04m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a01cb44767240109de1adebc7704529 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb19ac4348c46faab8641d68246e9f8 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d84adaf4da448cf9652dd81591e05a1 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a465c4754bf44b2d9bfdf18a1883ac2c : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7104b410e44d518929da76665b7c30 : SUCCESS in 4m 08s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/822ef093ad8343d3a684ab9f8da5913a : SUCCESS in 5m 15s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/b900d6abb150405787c60bba0cb29656 : FAILURE in 7m 59s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/ce1c988e9dd240b3837f06bc59510386 : FAILURE in 4m 55s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/1d5869a1e6f743c3ae25891624afdce4 : FAILURE in 6m 29s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/8415e493496a4af5a0e720ca42ad5458 : FAILURE in 1h 01m 10s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a1f07fd2391a4a2083842526bf6541aa : SUCCESS in 1h 48m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/96edd1e690c3413683d84aebf245d1b5 : FAILURE in 1h 07m 39s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/68b6b8b213e54df7b6889ad025d1db02 : SUCCESS in 47m 40s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/c943243223dd4495abe1a7e6dc921257 : SUCCESS in 51m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f7701551d4cf7f4a5d704d6fe2e80190dd5de5a1","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2023-10-25 20:28:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e6bc0910c0fb0149b0b129b3048fcbf948e9c5cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 22:02:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a62125b40be4a6987e7387db7e41cd4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cefe8cb0a6cf409fba246c1f00ec2533 : SUCCESS in 1h 24m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9237783454643bba6081149a4c0bdaa : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3626db226541bb9ef0c94459c60e7e : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7409e249703a492891d855a46181a58a : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af57ff1d48fe40d19ac6a915b44385dd : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61d0d8a5e2664af2981cb9f9fab911fa : SUCCESS in 3m 12s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/958870c123ed463db8291a22e078af0d : SUCCESS in 4m 29s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/153dffc429e7406d902d93d66de6e7c1 : FAILURE in 5m 23s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/6d2c431bfc7143e98adaec3b4ed06c15 : FAILURE in 5m 15s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/9455cae90b244abfa287d02eca917b45 : FAILURE in 3m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/b417e1f23326467ea4dcda0391019141 : FAILURE in 45m 54s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/8220f767b80a44f6941f2273874f664f : SUCCESS in 1h 26m 46s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/308344a7cbfa44d1ade05998130345cf : FAILURE in 44m 24s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/b2ceb74167414fa4a075709e0bf89855 : SUCCESS in 1h 06m 41s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e625acf008e9488dacc4536efe3c3bb1 : SUCCESS in 57m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f31ce2f50b21e300ff4fa7c0183cb0de5e2c368","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-02 04:33:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f18fdb6e03d7c3fb4865484717fe7796d04a2af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 06:23:40.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb511810e3a64db7a4f0e8f10ab4d753\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc1d28604074444e965d37ed3d26db26 : FAILURE in 50m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2ab88a233424a3eb3fcc11b606e48cf : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57741968ec5f46d7ac3c57245357cf15 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55b3b85b4e2a41b99ed7113eb2ea10ba : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e830f752fc424080957f3c674ede85d9 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f9d1343d124107af1b2cd9c5ea475f : SUCCESS in 3m 33s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/75633fe1d5d344ae84adcea381464171 : SUCCESS in 4m 38s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/d41ad216506042ee8acf483e6e9434b4 : FAILURE in 7m 02s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/a3edd1fd0b2b463c96800bc083875a05 : FAILURE in 3m 53s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/b836801f25524485b54268c360e2f02a : FAILURE in 4m 39s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/51b654de33d74117bdf4f1bdeff43ce3 : FAILURE in 56m 49s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2d5df60ff3ed40c28a385511a5320afb : SUCCESS in 1h 41m 55s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/01726d381a01460b8c1b0e5483785c10 : FAILURE in 51m 54s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/7a350fd4f3574a5a9054a5ea60e1539a : SUCCESS in 48m 09s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5e14be9e26b468b8025061883c57b24 : SUCCESS in 1h 18m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"271109036f4c7d6c751849ba7cd8bb1c53163c6d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-11-09 15:12:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c59653fdc0e6b72580cf1619b61c6f3f8c68ae15","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-23 09:30:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aa58fc3a1eb28df56d861c7bed494fa2a6b5deec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 11:07:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55f68a99db434289872c7cd257d0f8f1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd1c5ebcb1884d198c9ad49381939520 : SUCCESS in 1h 27m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9155be47ef1d4a5e929a00a038f21967 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cac356ff3394977944b190ccb1799f1 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b928e83578694e6aa9cdafd55146cca9 : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5c21d7b598243018af1e77876f04e27 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4951999bd6a49b2a9477d0b1278324a : SUCCESS in 4m 53s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/3c395809ec354646a71c10c72988a0ab : SUCCESS in 5m 05s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/4f6cfc2fe40c414fb2ca41fffa4450b5 : FAILURE in 7m 12s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/bb03c2370bfb4400aa4e343bcf6b6595 : FAILURE in 4m 28s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/af1a37557509400cba72d05db224c1d0 : FAILURE in 5m 44s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/660e61f7814741bd854fe662bf425163 : FAILURE in 42m 11s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/155e7d54f5354542b834defd6f324700 : SUCCESS in 1h 27m 28s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/bfc3cb9555084eb7a97431cd6efd6d83 : SUCCESS in 1h 14m 57s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/539c1482bd664cfc8e6f34be1a645e8e : SUCCESS in 47m 34s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/67b9916f453540eeb3914b06fcde4676 : SUCCESS in 1h 08m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6a56b04e3e1092de46317e1a3c70213e5a633c0e","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2024-01-24 12:34:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a9fb350cd0ec69576e325ebb6a3b4150758ebf4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 14:07:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62295a4703944e5e8ac255954f491ef5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f38057ce609a4036993550438d003951 : SUCCESS in 1h 27m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2d74db195c745609842bcdb88084cbf : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06872e25e1dd4de49b09e04fa56e57f4 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb3a9a5c8fde41059e17194b512d4bae : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d72004917e6456eb30ef317a925c40a : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3501a44f38640db9ca27e579dbf3622 : SUCCESS in 3m 39s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/4ef52be852da444897551697a77ede50 : SUCCESS in 4m 43s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/ddafd387c8864a8cb5c303921603e884 : FAILURE in 5m 32s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/93dc6683980b4ab0bd72f251ae6b91a0 : FAILURE in 5m 58s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/05d050164e2e445e82f3392e2d2d4f6f : FAILURE in 4m 11s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/3952ef3689b44ba097543f4ae5d91cef : FAILURE in 1h 04m 56s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/3207a84c1dc54a6fa8623ecc9d4480ea : SUCCESS in 1h 22m 00s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b0d25aebaf144e3a80cc0791dbf63f20 : FAILURE in 1h 01m 24s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/2638d8b8d97e4d878e2ff3ae7ecceb8d : SUCCESS in 51m 19s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/52e5b1bc6bed4777b8b28183312328d7 : SUCCESS in 56m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"df4a87d74140adc54e59cdbc2766c6102bbc99ba","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-26 08:47:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"73217244f3fc11c92fcbdc1cd5c67063ef92e1f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 10:34:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba8fe942724a41ae8b0a6988b57f75cf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0e492d9709d4adaae75a5ad1d9b6b72 : FAILURE in 1h 10m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a87ef9551a714128bb7d3871231e2257 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/224d27571090456a8f4fdfbe0f38919a : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7fd69b5ee984b9f96a8499f0e47bc92 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82693e4102e74addba952469f181d93c : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98c778ac4ca84c66ae97fbbb5e4576e1 : SUCCESS in 3m 34s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/adff9f4712e5460eb6e1dee60fe39796 : SUCCESS in 5m 23s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/9cad49d72e804e79983aa678d8d67b62 : FAILURE in 6m 26s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/9e4bbe1c4b6c4d8d86fa8f732f023358 : FAILURE in 4m 27s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/744c178e8e994888a3ab3c8808fd3c08 : FAILURE in 4m 27s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/cf2b138f7a3e4eb08d116bc07585e510 : FAILURE in 1h 07m 14s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/a928545f64394cec9e81c81f4b318239 : SUCCESS in 1h 30m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/0882c767498440be882d88d539ed19cd : FAILURE in 1h 00m 05s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/7dd0a5a5adb646618b36385595bdd43f : SUCCESS in 58m 13s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0ec07647b794a40bca12dc6d7c5e540 : SUCCESS in 1h 08m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"21f2dedf74e5e6813f7f86b4d7e14e73f6d545a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-26 12:33:46.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c4fc0b65d2155ec934920cfa36ff5a6e2761122d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-26 12:35:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"983b407dbfcb942495f1054ae7d71c0c81c86792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 14:22:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/562ab12db60b4db987d47ef5046faa75\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7dd2eba18d4f43eab24bf71c6b961ca4 : SUCCESS in 1h 08m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f794373f6bc44c28963db2ed223f5f38 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dba9ae6ee11b444c8a31b8a73631890b : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e42c6ea04fc4f66a6e1f4297029b54d : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8eedeff14c7d490387e4e0c89a35768a : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9efd6d29625642a6be67d6dd82be0f90 : SUCCESS in 2m 57s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f69d29d7e34543a7b9f378bee255e20f : SUCCESS in 4m 48s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/d85ed58707ca4ae8993e2420ffe182a4 : FAILURE in 8m 59s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/b3da71c54e074bc880d9b7509a317b1a : FAILURE in 4m 23s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/8ef7ae029ee04629bb7f4a720b71a64e : FAILURE in 7m 23s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/a80e95be656d4cbca3dd5600f493e429 : SUCCESS in 1h 32m 11s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/937755a8163745bebd6433417510d0c2 : SUCCESS in 1h 40m 03s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/5e836ccef2764c788bc5f58d343d8a13 : SUCCESS in 1h 11m 38s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3534f40a7a724396a18abe73896c96e3 : SUCCESS in 56m 53s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/0363d029801f42c1ac52b8ffc20185d2 : SUCCESS in 1h 00m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"df99fd41a5c7beca6e1781ad126404cf21951a36","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-26 14:39:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ea022798e677b57931775061307ebbbe0e95b522","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-28 08:06:44.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1cc63c81d9548fea399d32949ea2c0ffad7cc093","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-29 09:23:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"33d559609547211a1d4b66abcb084ae98d6b21be","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-29 17:21:55.000000000","message":"Patch Set 4: Code-Review+2\n\nLooks good to merge.","accounts_in_message":[],"_revision_number":4},{"id":"77768c08d218084b907c0c67df3cc4d4d4e1ff15","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-01-29 17:22:20.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"46bb523a9b2cb8a517679ac39236396022ea55a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 17:22:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a89c5fdae12b89d636ac0480996e2c42aad6fb51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 18:56:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/857f2dac3c1a495da1e0c7027721d6c7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66da6bdbb41c4aecabe4160b59a4b5e9 : SUCCESS in 1h 09m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb6fe148b1b41b0ba1f7c07a951b352 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1a5cbea22074d52bd3d4665b92ff361 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a05b39f804e4e738e47ac712c96f7d8 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7d3a69034a489d88cc1650def4a9c8 : SUCCESS in 3m 37s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/1cb50df97d3a428aa775ac7e0feffa4e : SUCCESS in 5m 01s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/7dc19f09e4824fa8b8918a7c06e06292 : SUCCESS in 1h 19m 05s","accounts_in_message":[],"_revision_number":4},{"id":"bf3a308a47fd14743e55b47eb69d901bf3f389c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 18:56:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"104a84c21be1e6dcad110ac0d276490b5d4ae6dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 18:57:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea4dccc347624c3a99f6a7e70a54de66\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d779bcef0674687ac97808fc2033248 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"15779aa0733f3c9bd1f85fa8aea25e3bd8915a1c","revisions":{"4db3e263531defae052fe3c4b5e1e87d9164bdb1":{"kind":"REWORK","_number":1,"created":"2023-04-13 10:58:04.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/89/880189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/89/880189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/1"}}},"commit":{"parents":[{"commit":"02f21fd32651b07271e4e88295247b8677a76539","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/02f21fd32651b07271e4e88295247b8677a76539"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-03-03 10:16:56.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-04-13 10:58:04.000000000","tz":0},"subject":"Disable greenthreads for RabbitDriver \"listen\" connections","message":"Disable greenthreads for RabbitDriver \"listen\" connections\n\nWhen enabling heartbeat_in_pthread, we were restoring the \"threading\"\npython library from eventlet to original one in RabbitDriver but we\nforgot to do the same in AMQPDriverBase (RabbitDriver is subclass of\nAMQPDriverBase).\n\nWe also need to use the original \"queue\" so that queues are not going to\nuse greenthreads as well.\n\nRelated-bug: #1961402\nRelated-bug: #1934937\nCloses-bug: #2009138\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34ea0d1381e934297df2f793e0d2594ef8254f00\n(cherry picked from commit fd2381c723fe805b17aca1f80bfff4738fbe9628)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4db3e263531defae052fe3c4b5e1e87d9164bdb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/4db3e263531defae052fe3c4b5e1e87d9164bdb1"}]},"branch":"refs/heads/stable/yoga"},"0fbe13991a03db23ca3a25af1526fc34c968d562":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-04-13 15:13:30.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/89/880189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/89/880189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/2"}}},"commit":{"parents":[{"commit":"02f21fd32651b07271e4e88295247b8677a76539","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/02f21fd32651b07271e4e88295247b8677a76539"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-03-03 10:16:56.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-04-13 15:13:30.000000000","tz":0},"subject":"Disable greenthreads for RabbitDriver \"listen\" connections","message":"Disable greenthreads for RabbitDriver \"listen\" connections\n\nWhen enabling heartbeat_in_pthread, we were restoring the \"threading\"\npython library from eventlet to original one in RabbitDriver but we\nforgot to do the same in AMQPDriverBase (RabbitDriver is subclass of\nAMQPDriverBase).\n\nWe also need to use the original \"queue\" so that queues are not going to\nuse greenthreads as well.\n\nRelated-bug: #1961402\nRelated-bug: #1934937\nCloses-bug: #2009138\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34ea0d1381e934297df2f793e0d2594ef8254f00\n(cherry picked from commit 3645839d162a989da46ba1c0fcbe7f32b48b9fb1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0fbe13991a03db23ca3a25af1526fc34c968d562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0fbe13991a03db23ca3a25af1526fc34c968d562"}]},"branch":"refs/heads/stable/yoga"},"0af402227a744de8c59a34db0038e488dc057345":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-04-13 15:14:33.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/89/880189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/89/880189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/3"}}},"commit":{"parents":[{"commit":"02f21fd32651b07271e4e88295247b8677a76539","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/02f21fd32651b07271e4e88295247b8677a76539"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-03-03 10:16:56.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-04-13 15:14:13.000000000","tz":120},"subject":"Disable greenthreads for RabbitDriver \"listen\" connections","message":"Disable greenthreads for RabbitDriver \"listen\" connections\n\nWhen enabling heartbeat_in_pthread, we were restoring the \"threading\"\npython library from eventlet to original one in RabbitDriver but we\nforgot to do the same in AMQPDriverBase (RabbitDriver is subclass of\nAMQPDriverBase).\n\nWe also need to use the original \"queue\" so that queues are not going to\nuse greenthreads as well.\n\nRelated-bug: #1961402\nRelated-bug: #1934937\nCloses-bug: #2009138\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34ea0d1381e934297df2f793e0d2594ef8254f00\n(cherry picked from commit 4d15b7c4fe0c14e285484d23c15fe5531e952679)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0af402227a744de8c59a34db0038e488dc057345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/0af402227a744de8c59a34db0038e488dc057345"}]},"branch":"refs/heads/stable/yoga"},"15779aa0733f3c9bd1f85fa8aea25e3bd8915a1c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-26 12:33:46.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/89/880189/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/89/880189/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/89/880189/4"}}},"commit":{"parents":[{"commit":"44b3427bc9efea9f341edfb8ea7aea38f25d1a5a","subject":"Change default value of \"heartbeat_in_pthread\" to False","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/44b3427bc9efea9f341edfb8ea7aea38f25d1a5a"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2023-03-03 10:16:56.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-26 12:33:31.000000000","tz":60},"subject":"Disable greenthreads for RabbitDriver \"listen\" connections","message":"Disable greenthreads for RabbitDriver \"listen\" connections\n\nWhen enabling heartbeat_in_pthread, we were restoring the \"threading\"\npython library from eventlet to original one in RabbitDriver but we\nforgot to do the same in AMQPDriverBase (RabbitDriver is subclass of\nAMQPDriverBase).\n\nWe also need to use the original \"queue\" so that queues are not going to\nuse greenthreads as well.\n\nRelated-bug: #1961402\nRelated-bug: #1934937\nCloses-bug: #2009138\n\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\nChange-Id: I34ea0d1381e934297df2f793e0d2594ef8254f00\n(cherry picked from commit 4d15b7c4fe0c14e285484d23c15fe5531e952679)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/15779aa0733f3c9bd1f85fa8aea25e3bd8915a1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/15779aa0733f3c9bd1f85fa8aea25e3bd8915a1c"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
