)]}'
{"id":"openstack%2Foslo.messaging~883533","triplet_id":"openstack%2Foslo.messaging~stable%2F2023.1~I103d2aa79b4bd2c331810583aeca53e22ee27a49","project":"openstack/oslo.messaging","branch":"stable/2023.1","topic":"bug/1993149-stable/2023.1","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-05-22 14:20:49.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"I103d2aa79b4bd2c331810583aeca53e22ee27a49","subject":"Increase ACK_REQUEUE_EVERY_SECONDS_MAX to exceed default kombu_reconnect_delay","status":"MERGED","created":"2023-05-18 14:29:03.000000000","updated":"2023-05-22 18:24:26.000000000","submitted":"2023-05-22 18:23:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883533-bug/1993149-stable/2023.1","cherry_pick_of_change":866617,"cherry_pick_of_patch_set":4,"meta_rev_id":"fc66d5fe48df5567f506ec0b8efb49f03d95923b","_number":883533,"virtual_id_number":883533,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-22 18:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-05-19 04:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-05-22 08:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-05-22 08:49:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-18 14:29:03.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"state":"CC"},{"updated":"2023-05-18 16:24:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-19 04:21:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-05-22 08:49:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"44793486455970ed5eda33d83aacec388ef955ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-05-18 14:29:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"638db058d48c41d69efad50327b72fbe9cca3535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 16:24:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b65a9a12efd4899a7ee25633aa608f2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7fbeba209f148b8977036c215487854 : SUCCESS in 1h 25m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d9aff741847423a85d516a3425d6b10 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed68158ec6c5418fa03d850e7f728cc1 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72fb6db9e6ac4ec694d5b904205f1665 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/deda159a76854e84b9821bfa62fecae0 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2ac9ac1e3345e38d7f6fc3e11038c9 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d30c6696a5a4b999e24c6151e79454d : SUCCESS in 3m 07s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/f1b0e14193744734a1e3b13f8c07b272 : SUCCESS in 5m 31s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/3854d25af1e644bbbbe3996bf94126b8 : FAILURE in 15m 25s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/b5f0fc700eb04bf88a47347271c9f861 : FAILURE in 4m 16s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/c67adcaf33b249c893a4ace3ca4bb4e8 : FAILURE in 12m 16s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/91d3b966e51e46ee8da34196ce569c69 : FAILURE in 1h 29m 34s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2a7f4572e12a46f2bb4a58166a0c14b2 : FAILURE in 31m 38s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/93aa4d7404e04c649b4eed1ce46fea3b : SUCCESS in 1h 52m 54s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/78b734e69b9b4c9f97a16a4a74f3af0a : SUCCESS in 1h 08m 36s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/2135db777077481c84dee446c5d2a442 : SUCCESS in 1h 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee635cf5924685305f08b9f2255b9b894bf48efd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-19 04:21:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee5a90e276e263e97e99189ab9802d485dfc695f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-19 04:21:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bb7d607ae4f91599f8b0c2e7e5290624cf15dcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 05:51:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1362f26cda2640d9aa38b80b24f455d3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b656293d6d34c3ea0f62f1030cdc49d : SUCCESS in 1h 09m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaef0d581e3045288178dcf10fe5c1bf : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e37bc6668b24428b93c76cb9eeee490c : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c25efc8d1d4a457898fa24d878e9a236 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1c7e7d583c6452fbcab5ddeeb6b8d8c : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a979c09b36ce4920810f9a952fd54ea9 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eee1c20d29bf4621ab8e1c562f7e013a : SUCCESS in 3m 16s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/e5c432e37c6d45ecaf8296322737f025 : SUCCESS in 5m 25s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/de20b150fe4a4bc392886787e34b7a8d : FAILURE in 8m 44s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/0870e836940347a6896aff1ad5536a3a : FAILURE in 3m 55s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/3bf2e2ebe2ad454491b9ca5812fbdfab : FAILURE in 6m 02s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/5ea5d0ba448c450f914f738a71cbf7d3 : SUCCESS in 1h 28m 11s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/2cfc98787d8a4c4689b5d21c36751317 : FAILURE in 16m 41s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/56a6ed0959514faa8970bb849e47b863 : SUCCESS in 1h 18m 35s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/3b58acb9108b4aacaa92633a89a6430a : SUCCESS in 52m 42s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/162152f43b5f4e77bb197f2bebb7e4b3 : SUCCESS in 1h 00m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"269e3525e979316e61630c8f7411c269a6ebbdd0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-22 08:49:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21c2e3c8cf27da321aaa719ead3fee0008542b1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 08:50:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"418887aa60d2ce9e5429e4482de100e0d41756d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 10:27:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6c14ec92dc14abba0f55a337a2a7da9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49c73833cab14556ae68ac254026072e : SUCCESS in 1h 26m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba0ef3583a94c34a3e93f8ed8e0ec8b : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a5f6a9a787f4780842a24224178c274 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2dcbc625dee34f52984d6e4cb6c581b8 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95364f9dc71a41cab271bfde0d628d84 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f6022128dd14522a268729492427963 : SUCCESS in 3m 08s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/306c7d30055a458badbb0489037b24fb : SUCCESS in 4m 39s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e0434e7e0b11478987d35bf04cc87aa5 : FAILURE in 1h 36m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5b612d894e5fde4382f5562279427f1920ea9d95","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-05-22 14:20:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de3cc9b90ef1522b32d94b6700b397392cf6d731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 16:39:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4457a0314cbf4320973e5717dd817d60\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/548369e4c4f84a7c84f77b099ff522f1 : SUCCESS in 1h 14m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/583463804f4d4ac18f89c8b611425f2d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa317b08e47457a9e7c51c87e566f9c : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09112d317e7949039f0dd1e76ddf27b2 : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9a383cb264942269e5d91dce995aca5 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175afe1bb5f047ddbb8d309c5a1275d2 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4434b7efc5a5499dbf58dc30c768ed73 : SUCCESS in 3m 21s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/83eb110f5f1c420dbb0a63048dc24599 : SUCCESS in 4m 32s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/c65ce62819084bea9a011289a8be9570 : FAILURE in 7m 14s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/8e4a7eaf798849938620484c06e5894e : FAILURE in 4m 11s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/1e0d315e0f794251b7415cefaca82a09 : FAILURE in 3m 59s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/2f3c6d8c160a43db9f634a36724caa07 : SUCCESS in 1h 41m 06s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/51189018f75f4a9494abc3b567db33b4 : FAILURE in 25m 10s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/31ec8f6581854cd3a06166d08acbd9a5 : SUCCESS in 2h 09m 31s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/c10bd158cfa043d78190659d4154d787 : SUCCESS in 46m 41s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/322e6012de664005aae8d6a589b44e80 : SUCCESS in 1h 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53bac92b016f759f12fa478f2a04edc0895cf294","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 16:39:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c4869db3628390a84a065fb32760b4f786e8e6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 18:23:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74d2589fdf4341a7b1115e6a1dd3474f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d32cafb4c3f4936a02dea5476455a3c : SUCCESS in 1h 27m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2161bbd66d704278ae81eaa5938c4bf8 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f57c6e44f1dd45a5aa5799c38c03937d : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09c072b2f2ad41c483dd414f0924c3b0 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5f3f9308e7c49bd8f92258efe4a37b2 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bb5d23a98a44df6ab8f747ff370beed : SUCCESS in 5m 13s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/47df5609daa149c49e8cc00ef322b910 : SUCCESS in 4m 57s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/ea12a96380aa45f291c74be955bacf5b : SUCCESS in 1h 38m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c57fed0bedfaf1455471ee7e2e1f5f3319019707","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 18:23:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fc66d5fe48df5567f506ec0b8efb49f03d95923b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 18:24:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b10ac6abe7484869a2595535ae28eab0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08b6136269a64219aa5d6323153cae2c : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32601e712a8549258992cda72f0debf3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4b49248bcfcb169f96ab2d47b5d207b1354ffa8","revisions":{"b4b49248bcfcb169f96ab2d47b5d207b1354ffa8":{"kind":"REWORK","_number":1,"created":"2023-05-18 14:29:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/33/883533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/33/883533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/33/883533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/33/883533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/33/883533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/33/883533/1"}}},"commit":{"parents":[{"commit":"50d87b95b6583bc6a4f02c5a04cb7151d2546289","subject":"move py3 jobs from zed to antelope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/50d87b95b6583bc6a4f02c5a04cb7151d2546289"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2022-12-05 15:25:00.000000000","tz":-360},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-05-18 14:29:03.000000000","tz":0},"subject":"Increase ACK_REQUEUE_EVERY_SECONDS_MAX to exceed default kombu_reconnect_delay","message":"Increase ACK_REQUEUE_EVERY_SECONDS_MAX to exceed default kombu_reconnect_delay\n\nPreviously the two values were the same; this caused us\nto always exceed the timeout limit ACK_REQUEUE_EVERY_SECONDS_MAX\nwhich results in various code paths never being traversed\ndue to premature timeout exceptions.\n\nAlso apply min/max values to kombu_reconnect_delay so it doesn\u0027t\nexceed ACK_REQUEUE_EVERY_SECONDS_MAX and break things again.\n\nCloses-Bug: #1993149\nChange-Id: I103d2aa79b4bd2c331810583aeca53e22ee27a49\n(cherry picked from commit 0602d1a10ac20c48fa35ad711355c79ee5b0ec77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b4b49248bcfcb169f96ab2d47b5d207b1354ffa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/b4b49248bcfcb169f96ab2d47b5d207b1354ffa8"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
