)]}'
{"id":"openstack%2Foslo.messaging~883538","triplet_id":"openstack%2Foslo.messaging~stable%2Fyoga~I103d2aa79b4bd2c331810583aeca53e22ee27a49","project":"openstack/oslo.messaging","branch":"stable/yoga","topic":"bug/1993149-stable/yoga","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-06-02 17:15:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-12 02:22:01.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-06-05 10:40:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-06-11 17:20:30.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"I103d2aa79b4bd2c331810583aeca53e22ee27a49","subject":"Increase ACK_REQUEUE_EVERY_SECONDS_MAX to exceed default kombu_reconnect_delay","status":"MERGED","created":"2023-05-18 15:10:54.000000000","updated":"2023-06-12 02:23:27.000000000","submitted":"2023-06-12 02:22:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883538-bug/1993149-stable/yoga","cherry_pick_of_change":883537,"cherry_pick_of_patch_set":2,"meta_rev_id":"76abe49391756ba2bebcde005cdf8c59d23f2dcb","_number":883538,"virtual_id_number":883538,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-12 02:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-06-02 17:15:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-06-05 10:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-02 16:51:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-06-05 10:40:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-18 15:10:54.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":12932,"name":"Andrew Bogott","email":"andrewbogott@gmail.com","username":"Andrew"},"state":"CC"},{"updated":"2023-05-18 15:55:58.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-05-18 15:55:58.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-05-18 16:59:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-22 08:54:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-05-22 11:03:41.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2023-05-24 16:51:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"89d23493e68f3c6d538e20966b5972c11ac0b4b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-05-18 15:10:54.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"98d35d78a7f269141e8936b782355ec5a80d80c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 16:59:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/03b4758c6807429c82307a0d40b355af\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09922d6769fc411080bb071394ef29bd : SUCCESS in 1h 28m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3591f45156d74ee99a116d6d2026ff43 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c78b4ceed154883bd6b569bf1d7dd8b : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6666a6654e0c49f3b72ef4ccad20d0e1 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6513b0c367e470399eb4d18a0c75815 : FAILURE in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf5b82567fd640fab37b0cfc059faa99 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4376075281084ca89d8169568a88e2fe : SUCCESS in 2m 59s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/c8d6a3e03e0b4228b549c358583d71aa : SUCCESS in 4m 57s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/1ad3eb2334294856bf833f54b238becd : FAILURE in 13m 40s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/3a21bb29ea1e433b9f56424cc2d102d0 : FAILURE in 4m 42s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/e6b51ff49414480284f2b35e22dd4dc7 : FAILURE in 12m 17s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/d77a4bca1bad4e54a9024bafd51d1038 : SUCCESS in 1h 37m 09s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/c356ca7fa85b4fa6b9ebcb2e45f41b6a : SUCCESS in 1h 39m 36s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b0a8db4f8be3448e9986a6e8d83c423c : SUCCESS in 1h 43m 00s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/68cc918b3e9640968852e7c7b15530d3 : SUCCESS in 1h 11m 52s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e4ee11a9315443da9a9f817d51315c3 : SUCCESS in 1h 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ede43b09e7edfe3a569fe825a616e02f2845749","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-22 08:54:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7acac0e0f09227883986af18f2be719f315c0148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 10:36:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f167b34386744a88b9b0b7c496f3e32\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2adb65e274d8422ea2c243b3c3fdd272 : SUCCESS in 1h 26m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9314816a64474cbb9adccfa8192db442 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91bf86f1b2964510934692759008c3dd : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c75d620c4d0f43a789334d1b37849689 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc98a3fbedb54be391c301113ac0a0b7 : FAILURE in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ad5dbafd1ad4986b495eb8986242759 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/344355945e7149d3bc5ceb87fd8b8a43 : SUCCESS in 2m 58s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/90928489b00c4e258089473614d691f7 : SUCCESS in 4m 23s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/4ec7c7e3226342039b2e6ae204febee0 : FAILURE in 5m 42s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/443493db56114084830ce81dabe9a385 : FAILURE in 4m 01s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/ee435b944e9746fd80976a7f54b4541c : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/e592487558654287bad6990f47d7a039 : SUCCESS in 1h 08m 08s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/48b6c588e43847a3a9524e755283c128 : SUCCESS in 1h 40m 15s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/562e18a9eaa543139edb9f14a42cf263 : SUCCESS in 1h 27m 11s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/01d9a7f21cb3439cba8e10b0eedd8036 : SUCCESS in 1h 06m 08s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d7805524ee64aad9ca3faf0008b2eac : SUCCESS in 1h 09m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17b83a215c8d432d8ef65844a69e4c51ea27b4d7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-24 16:51:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"acd2cebf11d0d3c2857b54a63785c030bcff41a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 18:59:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bdf1634020a4104afcc636cbf706d1e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c50740e13d914a8da39a235b59a25e34 : SUCCESS in 1h 08m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d34ec1f38d7d49c8a36ef5803d34cfd1 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5e25157f7414b81ad214747baeb93d7 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61ebf7602d134e80b213d896ba1ff59e : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80e8b224feb14366bf7e0467a7bba181 : FAILURE in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44465b1ab72f45059dddc844093a5fba : SUCCESS in 3m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2199025c21b343c3a05857e3b25cc1a9 : SUCCESS in 3m 20s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/37d4067f42c04e46b15c804af965af66 : SUCCESS in 4m 33s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/97285074f7934555864bfa009b24eb99 : FAILURE in 5m 36s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/926f42e20de74b048c2f0290637145ad : FAILURE in 4m 51s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/f1943e8149e14293a26da3708e7e78b3 : FAILURE in 4m 33s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/bc31041212f848c9a29ce8d921675c33 : SUCCESS in 1h 19m 05s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/ef2c40b09fd844d09158b663a16e5926 : SUCCESS in 1h 44m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/fe148290753b4e4e90052e8e95c82718 : SUCCESS in 1h 59m 06s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/52fe08d20c26449f818ba72b04f1d1de : SUCCESS in 46m 31s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/07c262b1b4d94fa99a5dd5a16c4a7f25 : SUCCESS in 51m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d14ee5a289face8e01c1672f4e79a6b47aa70bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-01 14:24:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e99ce0c607721c9da2938395f6378e501b185764","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-01 14:24:26.000000000","message":"Topic bug/1993149-stable/zed-stable/yoga removed","accounts_in_message":[],"_revision_number":2},{"id":"568da99235e6776b22264000473d1c56f884e55b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-01 14:24:33.000000000","message":"Topic set to bug/1993149-stable/yoga","accounts_in_message":[],"_revision_number":2},{"id":"755645082eddac4f648383e33e331c6332f9d48b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 16:30:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14886481c03e46d1a496398bca673b6d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a165633281d541b88d00f6904c51e43d : SUCCESS in 1h 17m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61d2145c5349422d971e9f9ffc6f3db0 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9f4007893ec474ebc1006a0adcd2bf8 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af3a401d7bfa40c4ae6b803a764a12da : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d54a19795e474ed7b2897690f4cbb031 : FAILURE in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b5a6ca5fc614311a70ef4f4a4aeb400 : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/466693137c0c450996e670e95c735c5f : SUCCESS in 3m 16s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/b3ae48e8740847088b93c68224d282d6 : SUCCESS in 5m 12s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/92ef7dbe364c4b67ab401950e29f42af : FAILURE in 6m 26s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/f3e399be4b0f46208360f7f36eaa4a77 : FAILURE in 4m 26s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/8843ee30b7b94ccfa29d489ace563f27 : FAILURE in 4m 34s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/29eda707a25345a89d3f2b5b0c32219a : SUCCESS in 1h 54m 47s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/af89b8d2663f45af974c028c89f1c3aa : FAILURE in 1h 24m 58s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/b7cbb89ad1714a959e9a72e92dc57de3 : SUCCESS in 1h 23m 19s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/4c17d9af5a244cbeaf2b457a037a84d9 : SUCCESS in 1h 15m 30s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/888f546720c54ce78fe285be0fa43517 : SUCCESS in 1h 06m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e08142182cd09b462fdb6dea33709e3d03937d60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-02 16:51:04.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bb9e7bbd5123807b0d6c63b33c53195db6b964e5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-02 16:51:48.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"43d5da988f1618ea6ad3b3a28eaeeac1affe684d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-02 17:15:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"39cd1ab8a2adab0c409b7cb6c9176655e6cda20d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 19:16:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deb396c8797f48ab976131627b644ad2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/498e83638fcb4de6943dda83c350d9ec : SUCCESS in 1h 09m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96839e32c5114b54ae3f214fa46100f7 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aca2eb614c844afa63d98108044a48b : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22ba8cead03146c6b1d7459c212b5118 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21f8b5b59820483894f97247458ec9ce : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1f3d0b5aac46d6b93937344999eb27 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8e226924e8e4c499e71ea1d5177138a : SUCCESS in 3m 32s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/d24c38ae55784a128dbdb0f4b4242208 : SUCCESS in 5m 03s\n- oslo.messaging-tox-py38-func-scenario02 https://zuul.opendev.org/t/openstack/build/52ce9221f55c46bf87cf73cf5a1bc139 : FAILURE in 6m 19s (non-voting)\n- oslo.messaging-tox-py38-func-scenario03 https://zuul.opendev.org/t/openstack/build/42cd991c7ce04690a5e106a5e313ae7b : FAILURE in 4m 07s (non-voting)\n- oslo.messaging-tox-py38-func-scenario04 https://zuul.opendev.org/t/openstack/build/79674dfa95534542bb1d5e2f6f802146 : FAILURE in 4m 35s (non-voting)\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/edca1143f7af4aabbc490cdd304f34f9 : SUCCESS in 1h 11m 48s\n- oslo.messaging-src-dsvm-full-amqp1-hybrid https://zuul.opendev.org/t/openstack/build/dea903481625436cad82a92f86ab6dc5 : SUCCESS in 1h 15m 52s (non-voting)\n- oslo.messaging-src-dsvm-full-kafka-hybrid https://zuul.opendev.org/t/openstack/build/dca0fac2bc5b481fa54ea1538631ce44 : SUCCESS in 1h 41m 52s (non-voting)\n- oslo.messaging-grenade https://zuul.opendev.org/t/openstack/build/a77b630ab93b43bea331458d442be4f7 : SUCCESS in 57m 38s (non-voting)\n- oslo.messaging-grenade-multinode https://zuul.opendev.org/t/openstack/build/3936208d9d4544a3ac351d59a7ca1666 : SUCCESS in 1h 20m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b1ffbffaa2c8b6273ff4dc4e5216475bae17fc5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-05 10:24:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"72dd668b73b79d3a15c243d839ca3a7a4a6a1ce7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-05 10:40:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9be9db93fe7e262b5dd23629431706a175d8326","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-06-11 17:20:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ece6ac8f8c92f864fa99ddd88080f28027252d0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 00:58:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5f3e0fa2a1820e982df58be56aaa1e6812de13ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 02:22:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cd188e430a9406181088b650b8ae786\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa2b2053aba74b61840551b110a56c2a : SUCCESS in 57m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0139caf753640ffa6ad378f8443a31d : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfcf59d1be094b4196abd9720167935e : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f36fbfb664d143c096e586bed6fd02fb : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d08fb627003d4e0d9a320a9d3732bffa : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7705e2c19d44251ad4f663aab63db74 : SUCCESS in 3m 38s\n- oslo.messaging-tox-py38-func-scenario01 https://zuul.opendev.org/t/openstack/build/b6902e29c61c45c9a9aad7bf78e909f4 : SUCCESS in 4m 51s\n- oslo.messaging-src-dsvm-full-rabbit https://zuul.opendev.org/t/openstack/build/17f059cc3c3f472f94180ea512de0837 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":3},{"id":"a1a607351824851f4a94a121e3a57744aa0d2dde","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 02:22:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"76abe49391756ba2bebcde005cdf8c59d23f2dcb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 02:23:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9a12ec9bd9d45268ff74a962b6489e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/690b12219510418697d694c74038d4f1 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/087675fcfa6940cfb769f558ccbe2a0e : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f20a905ea6f41399c1723f8f1cbd0bc1097b8672","revisions":{"a5ca24fc0a6c1dcc765b4a2b1466f13a5060ba23":{"kind":"REWORK","_number":1,"created":"2023-05-18 15:10:54.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/38/883538/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/883538/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/1"}}},"commit":{"parents":[{"commit":"02f21fd32651b07271e4e88295247b8677a76539","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/02f21fd32651b07271e4e88295247b8677a76539"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2022-12-05 15:25:00.000000000","tz":-360},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-05-18 15:10:54.000000000","tz":0},"subject":"Increase ACK_REQUEUE_EVERY_SECONDS_MAX to exceed default kombu_reconnect_delay","message":"Increase ACK_REQUEUE_EVERY_SECONDS_MAX to exceed default kombu_reconnect_delay\n\nPreviously the two values were the same; this caused us\nto always exceed the timeout limit ACK_REQUEUE_EVERY_SECONDS_MAX\nwhich results in various code paths never being traversed\ndue to premature timeout exceptions.\n\nAlso apply min/max values to kombu_reconnect_delay so it doesn\u0027t\nexceed ACK_REQUEUE_EVERY_SECONDS_MAX and break things again.\n\nCloses-Bug: #1993149\nChange-Id: I103d2aa79b4bd2c331810583aeca53e22ee27a49\n(cherry picked from commit 0602d1a10ac20c48fa35ad711355c79ee5b0ec77)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a5ca24fc0a6c1dcc765b4a2b1466f13a5060ba23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a5ca24fc0a6c1dcc765b4a2b1466f13a5060ba23"}]},"branch":"refs/heads/stable/yoga"},"ee6923c07c469183fc82cfda30ab78c295cc6a12":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-06-01 14:24:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/883538/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/883538/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/2"}}},"commit":{"parents":[{"commit":"02f21fd32651b07271e4e88295247b8677a76539","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/02f21fd32651b07271e4e88295247b8677a76539"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2022-12-05 15:25:00.000000000","tz":-360},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-01 14:24:17.000000000","tz":0},"subject":"Increase ACK_REQUEUE_EVERY_SECONDS_MAX to exceed default kombu_reconnect_delay","message":"Increase ACK_REQUEUE_EVERY_SECONDS_MAX to exceed default kombu_reconnect_delay\n\nPreviously the two values were the same; this caused us\nto always exceed the timeout limit ACK_REQUEUE_EVERY_SECONDS_MAX\nwhich results in various code paths never being traversed\ndue to premature timeout exceptions.\n\nAlso apply min/max values to kombu_reconnect_delay so it doesn\u0027t\nexceed ACK_REQUEUE_EVERY_SECONDS_MAX and break things again.\n\nCloses-Bug: #1993149\nChange-Id: I103d2aa79b4bd2c331810583aeca53e22ee27a49\n(cherry picked from commit 0602d1a10ac20c48fa35ad711355c79ee5b0ec77)\n(cherry picked from commit b4b49248bcfcb169f96ab2d47b5d207b1354ffa8)\n(cherry picked from commit fa3195a3459cae3f4e9be43f114ee2d5eb7a60f1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee6923c07c469183fc82cfda30ab78c295cc6a12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/ee6923c07c469183fc82cfda30ab78c295cc6a12"}]},"branch":"refs/heads/stable/yoga"},"f20a905ea6f41399c1723f8f1cbd0bc1097b8672":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-06-02 16:51:04.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/38/883538/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/oslo.messaging","ref":"refs/changes/38/883538/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/oslo.messaging refs/changes/38/883538/3"}}},"commit":{"parents":[{"commit":"a1b36e7823e2764db7914ad9fbcbc0a5dc300205","subject":"tests: Fix test failures with kombu \u003e\u003d 5.2.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/a1b36e7823e2764db7914ad9fbcbc0a5dc300205"}]}],"author":{"name":"Andrew Bogott","email":"abogott@wikimedia.org","date":"2022-12-05 15:25:00.000000000","tz":-360},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2023-06-02 16:50:39.000000000","tz":120},"subject":"Increase ACK_REQUEUE_EVERY_SECONDS_MAX to exceed default kombu_reconnect_delay","message":"Increase ACK_REQUEUE_EVERY_SECONDS_MAX to exceed default kombu_reconnect_delay\n\nPreviously the two values were the same; this caused us\nto always exceed the timeout limit ACK_REQUEUE_EVERY_SECONDS_MAX\nwhich results in various code paths never being traversed\ndue to premature timeout exceptions.\n\nAlso apply min/max values to kombu_reconnect_delay so it doesn\u0027t\nexceed ACK_REQUEUE_EVERY_SECONDS_MAX and break things again.\n\nCloses-Bug: #1993149\nChange-Id: I103d2aa79b4bd2c331810583aeca53e22ee27a49\n(cherry picked from commit 0602d1a10ac20c48fa35ad711355c79ee5b0ec77)\n(cherry picked from commit b4b49248bcfcb169f96ab2d47b5d207b1354ffa8)\n(cherry picked from commit fa3195a3459cae3f4e9be43f114ee2d5eb7a60f1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f20a905ea6f41399c1723f8f1cbd0bc1097b8672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/oslo.messaging/commit/f20a905ea6f41399c1723f8f1cbd0bc1097b8672"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
